)]}'
{"id":"openstack%2Foslo.messaging~889701","triplet_id":"openstack%2Foslo.messaging~master~I34bad0f6d8ace475c48839adc68a023dd0c380de","project":"openstack/oslo.messaging","branch":"master","topic":"bug-2031497","attention_set":{},"removed_from_attention_set":{"28627":{"account":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"last_update":"2023-11-14 16:21:32.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-11-14 14:37:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-14 16:21:32.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-14 16:21:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34bad0f6d8ace475c48839adc68a023dd0c380de","subject":"Allow creating transient queues with no expire","status":"MERGED","created":"2023-07-26 11:51:53.000000000","updated":"2023-11-14 16:22:50.000000000","submitted":"2023-11-14 16:21:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889701-bug-2031497","meta_rev_id":"ed058ee10e691abc6dd6fecb8ea4f232d3242e55","_number":889701,"virtual_id_number":889701,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-14 16:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":2,"date":"2023-11-11 23:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-11-13 05:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-14 14:37:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-11-14 14:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-26 11:53:14.000000000","updated_by":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"reviewer":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"state":"REVIEWER"},{"updated":"2023-07-26 13:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 10:34:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-08-24 13:40:55.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-08-29 09:18:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-08-31 09:43:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-11-02 04:46:30.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-11-13 05:34:45.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-14 14:37:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b16550f381857d6d9c334c19f3b0236a0ecec7cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-07-26 11:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f84c97c1c20248ae57a43b2960cae11a4480c0e","author":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"date":"2023-07-26 11:53:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9747a45c5cdbf65ccb269846ef7f5d2fa4d62cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 13:30:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14487c6ad91f48c69c67e6e6218eb9a3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90584384f2794000b1ade03c00314344 : SUCCESS in 1h 33m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b97f947e33e84e01a6b221bc8199191b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe82d38870a46a0b4074a4bbb65058e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38db0f56bfb34ca8a8d98710a24637e4 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1adb2ddcc8d4759acba0fd0599f3422 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09549a9574d7420eaecfe2e532853f7e : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1afcf6ede14e3b81452518ff0d07f7 : FAILURE in 3m 21s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/8a29263a17b74e98bf4ec7078926f259 : SUCCESS in 4m 49s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c12041b423bc415399103cee9ca953ba : FAILURE in 4m 31s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9eaf08d93206432bbde4044a495d9a23 : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/c9e4cf1a20da4a60aa99c284efc65b8c : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dd4ac6e7a8c24b8995b88e41d2df2fd3 : SUCCESS in 1h 29m 46s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3bc15b099ef740d3a709a146530d8e30 : FAILURE in 17m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0ba2b53b947a41cd85b157a8ab35c5d1 : SUCCESS in 1h 19m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d99fd46b29f043bd888e9e5ac722d350 : SUCCESS in 43m 54s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/617d004238614b31a4ea2ecaa4ab40ff : SUCCESS in 52m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ffcd46a7ca093732eb5798b70559b0610526eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-08 13:34:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"400ecfe74a3f199f1215a037eb1cc80e839d46c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 15:27:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a2b756d28da4370a670b0edd6a4042c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91a0c84221644577bf836217ddee5d96 : SUCCESS in 1h 45m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/278ff3f1908d4df690b309f9b991abf6 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c5fe0cc0064f61a356c27929f82a0e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5908f09e3f2640b38eb106b9b4e6feb5 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f32f02fd50f4365b14f2b82df73b839 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1ca1da011db404db9313108ddf6bd3a : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a675d58d02649b4a088819ba9700241 : SUCCESS in 3m 54s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/21689d66f7f542858fbfcc7ec4229046 : SUCCESS in 5m 16s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/dc48bc1937cf4a4abd11abe93859baf2 : FAILURE in 5m 39s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/84b48d475bef4d75bb8988c89c8c1fdd : FAILURE in 4m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/e34c8b93106c41689c1cc5b4bf07d7e0 : FAILURE in 4m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/907f77ae221f4c26981a8bdbd1e6606e : SUCCESS in 1h 41m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/eb29182268bf4d21925e7a202c8dd00c : FAILURE in 18m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/98c2bf280c8d415a8542770971d1bff5 : SUCCESS in 1h 38m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2fdb52e1706341d7b206195cde755f59 : SUCCESS in 1h 03m 25s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a57b7d60a2414214bc7b9a971e96f191 : SUCCESS in 1h 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e43b54b086b8d6eb947fde1982cd77ade540427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-16 08:45:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"97f6dbdd9593a40f6b75b6708b908b9b93a952e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 10:34:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb7bb356fa20768657b720b8667c8321c1007ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 10:52:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df76f335c3084d65978a3b147524e05a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb1ba01f530e4908b4e90fd7bfa1a508 : SUCCESS in 1h 18m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f45870db91a847abab15b6ac1eec8b14 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c4041633e24676adefe1575d80167a : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bcb131ff2504272900c41c87337a84a : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5932236c992b4aa09c6186dbeb64d621 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27d2c117867d4d8a9edd440d52295750 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34459842a6b461284a80c92c1e4442c : SUCCESS in 3m 59s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bdcd4668f6024027be493c41bc1175cf : SUCCESS in 5m 23s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/82c7ad400aa94577a232fa9749fa30fc : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/0e74a25468654da7a3fdf6740d0c7376 : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/82a8b005896b4e47a7aa4660c6adad9a : FAILURE in 4m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/07b587d3b24d47548b012da8c25b0c4b : SUCCESS in 1h 59m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1a0730ae498446248fef42666af7ff1b : FAILURE in 20m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4c19e066369e4e0192d045422ba8510b : SUCCESS in 1h 43m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cc76326e0179484188c61cdd85b4500a : SUCCESS in 52m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f3fdc140abe4a269397b4b64f1e22df : SUCCESS in 1h 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"747c3081816b9ff7aeda34952413dbc03f2348ce","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-29 09:18:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b232aee75a1269b2f073f0a4ecb130741cacaa42","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-31 09:43:58.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"87e0da9baa8b5a99d14882ca4ab1600e59fda26b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 13:53:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"aa18b753f6095c07d472e854f17e4224a5df5a1b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 13:53:57.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"03355125e4a0bc0a11da3c6d3db25ac689d8ea0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:41:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/234ff895373f4b7090eedf812d8a4066\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23396c58b1b44613b85a815831a324af : SUCCESS in 1h 36m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/677ac3c4f3b049bfa0b1c03411fbcf10 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d53265222b4eee9874522d2def3879 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b22bf4181254f31b76527e94f8d4388 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea456f379b4c4b5dbb3a5925a5a2286f : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ad3f19949354e8d89fcd976088a2dbc : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfaf9fb4ebd4dc3a832e64f0e723135 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37e2f0af60c4bff80bcf7a34a3f467b : SUCCESS in 3m 27s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3901da611db04eee84815f5616e21b80 : SUCCESS in 4m 49s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e55fe357e12341a197c861e07511cbca : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/add86f70b32b4bb88b40665d204ae04e : FAILURE in 4m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/848088ba1380435981806f32b7dfbfdc : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/99695bbc980745309e97fea33b52d1d5 : SUCCESS in 1h 41m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/aeb9b1cee8804eb8991f0a45fc4eeb0a : FAILURE in 30m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/aebfa7f4facf4635a542c9657cca1d42 : SUCCESS in 1h 37m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/405097d5adf0435cb7df96f6c1dfe1d6 : SUCCESS in 54m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/643b2036bc164e7aaadf5b8ab30c758f : SUCCESS in 1h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"00022e192dd97ff5fbbf23a3182396d63fe492eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-12 12:11:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2598e8d3d2784278ec2f28df1d56dfae15157541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:54:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b21c3ebe7fe4424a2b6232de582bcd2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fbb79f8197b4f7f8e8eded3c0bf89f9 : SUCCESS in 1h 28m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd4b5c97ba63470489915bb6f229ef42 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b623cf8db254994ab2c95ed6d13181c : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a1aa4520034b8c8457760c507706a0 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/915d5980400940028318bcf859969464 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a9bb542bfc540c8b5c22d0b4c9bf4a2 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39b4818331734f50af976f39c0ca0595 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9bfbd33f2cb4380a00d75d9728dae01 : SUCCESS in 3m 44s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9bd5ebf5b96b4ecb891a60ec167b95ad : SUCCESS in 4m 54s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/360f77f788474f8c8f77e8af719aa775 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/fcca429ca94540d990b3a07deee06bd6 : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/74a4854cb6264ba2abad7620e92d3487 : FAILURE in 3m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c16c3b77f00340b3aaf9940346a38b54 : SUCCESS in 1h 35m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2ac10d9fff124b4c9d6e80ad1f06a625 : FAILURE in 29m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d5ad845b2fd04601b23942b8a2d86789 : SUCCESS in 1h 33m 55s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/96b31b1b52de4199be10cb26b530f736 : SUCCESS in 43m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e0efaf323dd4ad8b3bf46499c31dfb8 : SUCCESS in 1h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"157baf4c603fff7953bd76d485937c52a3a42791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-19 15:59:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2710fef9c267dc1ea604c9e7cde13b61c462da22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 17:49:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217e5817e4b34416b709130ff96c6ef6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a960125cc5546ba87cc804686f19f5b : SUCCESS in 1h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ea616adf9f949afb0b66d7f4f6f2937 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405a31a1d2604432aa0a2661f02b9b82 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231d023c693643bd88baecf7511aaea5 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf5a3ab94f5b4ee2b2f3ffc0da1ef8ec : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f70036e0749492da4b6ea2907e2e418 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14118f4981a344d398a517109d56f891 : SUCCESS in 4m 28s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3514e658de644646950b8ff580d23a82 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/70cc41f74f5440ac8698bb4c3ff239b4 : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/dce71f3a5d094defa254bd2c6f3c5f29 : FAILURE in 4m 39s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/4e1bcf4cdf7540728530146443dc0d27 : FAILURE in 3m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0d5e58c6a2b64e9fa53491c2b4fac146 : SUCCESS in 1h 40m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/95619585c10448d7861fc54d55e02c43 : FAILURE in 30m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8f43d9005d3a47b58c0af8ffe7ad600e : SUCCESS in 1h 42m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6036218838264e3e91a709189bd49d9d : SUCCESS in 49m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ea1d31e62b246d1b0c391cddebab08d : SUCCESS in 53m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f291ca8699eab1ad74eea2fff253f4123a8972c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-02 04:46:30.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"379a95dd47d3a9a2c5755574e0319201f913766a","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 22:18:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"12b3d98204bbd145bcba3b7d9d2eb86887b63322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 23:08:39.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"82043543808a8e495aaba8c0a101c6bb02d9581a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 00:51:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56181086b69544c78927e6e2d5185c38\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e51d49fc477a4a1681fd5b3dbff34040 : SUCCESS in 1h 15m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4213a943288422bba0f7cef9e536437 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4acaece95e164693a80c4943ae07eb05 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc51990df67345f58ea168a45d029262 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95f29a4690a94b6484de17c6c80fb988 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c3c5a4f45441a48d308b0c92f4fbe0 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f697d28f875e4da19f8ccb3b5f306a67 : SUCCESS in 3m 21s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/85e081c1276b46a2b90866ad5e2a7fec : SUCCESS in 5m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/17367041c2e64b8ead43a671d038501f : FAILURE in 6m 00s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/37009b91c90f439385900257d9ad8952 : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/bb2e335c296645c3879326952125c632 : FAILURE in 3m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/49fa23f01e244bf89acba90d62c29be9 : SUCCESS in 1h 10m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7d8b038a99034c299caf975c2f45dfc5 : FAILURE in 15m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8a3a6096f1a14d3f851d7f575b561023 : SUCCESS in 1h 36m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2fada3b9b189416fa06f40c7f68e5eac : SUCCESS in 1h 03m 00s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/26b56459af134b69a105f84701dd0793 : SUCCESS in 50m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d27b01a9a901e1d58e6796e060df4a0d9faa2155","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-13 05:34:45.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"576f2819a4df2e9805d3e82efebc4259237a6cd1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-11-14 14:37:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"22d8bfb1659de4273630c88353371a25a6454010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 14:38:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0d8c86ae10be6818ebb51798d112f40a95b131e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:21:32.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2951d375f514deda6647fde81857818\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29556b60d70c41329c940f5b5f624166 : SUCCESS in 1h 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613c12b7c079429aa84a16cafe22a724 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1462f8a3ecd483786528ea13523d9e3 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10dc39cc957042eb8c9e2b08fd74a73d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4378c2135d8841ab84c3a0c637c1b722 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d6b327d2e004eb0a67cb87409402770 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/43aed4a93f924d1290da6f2af2417654 : SUCCESS in 5m 44s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/875de600af4b4dee95f7bdcf0374bd75 : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8175ad1aedfcb39c9672f15ef09e641adb765335","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:21:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ed058ee10e691abc6dd6fecb8ea4f232d3242e55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 16:22:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a3289370644b25bc27d11570fe38f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cf7ab8d68045e0b26c23b92942931c : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d7faf31c31444c0950e5f307702f735 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f23f3276c4f8c156c7a917ffd05db6b3b010b807","revisions":{"d05409b7bcb42afcb1d6c10ae63a537aabad98eb":{"kind":"REWORK","_number":1,"created":"2023-07-26 11:51:53.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/1"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 11:51:15.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possible to set the transient queue TTL to 0 will allow\nthe creation of the queue without the x-expire parameter and only the\npolicy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d05409b7bcb42afcb1d6c10ae63a537aabad98eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d05409b7bcb42afcb1d6c10ae63a537aabad98eb"}]},"branch":"refs/heads/master"},"960f51ea999fb9a952c943203d459a8a009fb6a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-08 13:34:41.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/2"}}},"commit":{"parents":[{"commit":"5cfd1f9937f02b245a3476e9ef648e4721455ac7","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cfd1f9937f02b245a3476e9ef648e4721455ac7"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-08 13:34:35.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possible to set the transient queue TTL to 0 will allow\nthe creation of the queue without the x-expire parameter and only the\npolicy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/960f51ea999fb9a952c943203d459a8a009fb6a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/960f51ea999fb9a952c943203d459a8a009fb6a8"}]},"branch":"refs/heads/master"},"03322a46e01f17a07030caa2b2c4af80e99b841e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2023-08-16 08:45:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/3"}}},"commit":{"parents":[{"commit":"b4b6f55de222716b0029d45adc33a60f8b0e68e0","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4b6f55de222716b0029d45adc33a60f8b0e68e0"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-16 08:43:09.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possible to set the transient queue TTL to 0 will allow\nthe creation of the queue without the x-expire parameter and only the\npolicy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03322a46e01f17a07030caa2b2c4af80e99b841e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03322a46e01f17a07030caa2b2c4af80e99b841e"}]},"branch":"refs/heads/master"},"f080db183344754bb2d547810d1da231785f0de1":{"kind":"REWORK","_number":4,"created":"2023-09-06 13:53:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/4"}}},"commit":{"parents":[{"commit":"5f3c31e10ea9cb0631049b06a3db300245ddb3d0","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f3c31e10ea9cb0631049b06a3db300245ddb3d0"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-06 13:53:16.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possibility to set the transient queue TTL to 0 will\nallow the creation of the queue without the x-expire parameter and only\nthe policy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f080db183344754bb2d547810d1da231785f0de1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f080db183344754bb2d547810d1da231785f0de1"}]},"branch":"refs/heads/master"},"7f707fd73431eb8c24090fb8b4e515a87c7794fa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-12 12:11:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/5"}}},"commit":{"parents":[{"commit":"7233839d5646b4712b3d7fbe88acdbef3c2a8dea","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7233839d5646b4712b3d7fbe88acdbef3c2a8dea"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 12:11:43.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possibility to set the transient queue TTL to 0 will\nallow the creation of the queue without the x-expire parameter and only\nthe policy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f707fd73431eb8c24090fb8b4e515a87c7794fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f707fd73431eb8c24090fb8b4e515a87c7794fa"}]},"branch":"refs/heads/master"},"86b67a7bb3145afa46b73d8c8d7e26f7d12de375":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-19 15:59:47.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/6"}}},"commit":{"parents":[{"commit":"33791d8662b0a44b976d8aa16b85ddd64e0b4123","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/33791d8662b0a44b976d8aa16b85ddd64e0b4123"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-19 15:59:33.000000000","tz":120},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possibility to set the transient queue TTL to 0 will\nallow the creation of the queue without the x-expire parameter and only\nthe policy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/86b67a7bb3145afa46b73d8c8d7e26f7d12de375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/86b67a7bb3145afa46b73d8c8d7e26f7d12de375"}]},"branch":"refs/heads/master"},"f23f3276c4f8c156c7a917ffd05db6b3b010b807":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-11 23:08:39.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/01/889701/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/01/889701/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/01/889701/7"}}},"commit":{"parents":[{"commit":"3438726dd08eef03a0962c72096f268a54fd7f4e","subject":"Add some logs when sending RPC messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3438726dd08eef03a0962c72096f268a54fd7f4e"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 10:12:16.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-11 23:08:20.000000000","tz":60},"subject":"Allow creating transient queues with no expire","message":"Allow creating transient queues with no expire\n\nWhen an operator rely on rabbitmq policies, there is no point to set the\nqueue TTL in config.\nMoreover, using policies is much more simpler as you dont need to\ndelete/recreate the queues to apply the new parameter (see [1]).\nSo, adding the possibility to set the transient queue TTL to 0 will\nallow the creation of the queue without the x-expire parameter and only\nthe policy will apply.\n\n[1] https://www.rabbitmq.com/parameters.html#policies\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34bad0f6d8ace475c48839adc68a023dd0c380de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f23f3276c4f8c156c7a917ffd05db6b3b010b807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f23f3276c4f8c156c7a917ffd05db6b3b010b807"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
