)]}'
{"id":"openstack%2Foslo.messaging~890824","triplet_id":"openstack%2Foslo.messaging~master~I7d0c318624d3d02182392ca3f06eed04d4133728","project":"openstack/oslo.messaging","branch":"master","topic":"bug-2031497","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-13 03:43:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d0c318624d3d02182392ca3f06eed04d4133728","subject":"Add some logs when sending RPC messages","status":"MERGED","created":"2023-08-08 18:39:08.000000000","updated":"2023-11-13 03:44:42.000000000","submitted":"2023-11-13 03:43:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890824-bug-2031497","meta_rev_id":"6be5034b2ed50a7e6f701f79ad5af45378793530","_number":890824,"virtual_id_number":890824,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 03:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"all":[{"value":1,"date":"2023-11-11 23:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":2,"date":"2023-11-11 23:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-11-13 01:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-11-11 23:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-11-11 23:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-11-13 01:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 20:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 10:51:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-08-18 09:23:28.000000000","updated_by":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"reviewer":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"state":"REVIEWER"},{"updated":"2023-08-29 09:18:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-08-31 09:44:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-08-31 11:44:38.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-11-13 01:43:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"7fdebb5f41eb496cb0503d70c710b664d7f78c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-08 18:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c9e44c2e5ba3c396a4931c740760742110e8cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 20:23:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/502671eb04d04c08aff33856d51843a8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35e53f12c2534c24a8747681a020389a : FAILURE in 1h 35m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c0edd2d74d04ae885a4c6e00336787d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd890eb65094933be16eda6857d2371 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc88648112654f539ef5d2fed9322b63 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2f0433d6236428f955231384eb99aee : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/962314d8ea5a48b1b6e02aabc2c4573b : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0f2934604d4de1a1f452f2d63812f9 : SUCCESS in 3m 47s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/0d39aeaa1fbb44458b7ea07c1cccad3e : SUCCESS in 6m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/cf118f21ce10474cbe83d4e87fd21844 : FAILURE in 5m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b0f7e66ea01d4e7d9dc8369fb67e2b7d : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/93f0258e32b541baaf6488cac162f454 : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2390455d2a2045fb9ead5e2ea7761851 : SUCCESS in 1h 14m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/946c4f376dea44ca9059f6e7c4f907a6 : FAILURE in 19m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/77778b86760840d0ab0804014c53b071 : SUCCESS in 1h 03m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f0e2e6faed6f4d7283f85ac37957057c : SUCCESS in 51m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fe80db9715c46c8a551e58eac266f67 : SUCCESS in 1h 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eedca363949e3e1fe5c3b0b5b2263d64ff2f24c0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-10 09:09:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9af01f19e00d444b5476937eba6f2e6a1f3af77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 11:16:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6004da345e6e475aba40363067eddba1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e15bcaac157e43848b334acb015ecb2f : SUCCESS in 1h 51m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e50b8476fe334621b4a4bf0a613f089c : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175068b1224a4fd592e7b5d8c330744a : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a7e5b9cfc94368a16936ec8662a0c8 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf8f3754a4a241c9b7a7aaa879a2daf6 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32b57dd9823b4e12a8b8cb1a984d87be : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cedbd47cc44759b92fbf8de0996f88 : SUCCESS in 4m 28s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/fdef91f87d304a64ad6bb16e8223b460 : SUCCESS in 6m 06s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/44ede74f7a17451ebf0f2f4721df89ff : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/f31884e72eb345538bc076d317af64f4 : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1bb3e56d3443456bb3635f6472cda781 : FAILURE in 4m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2a38f5560d044bcfae637e5d1d4f60e8 : SUCCESS in 1h 47m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3539db9c0e274852afda0eb85bd5a782 : FAILURE in 19m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/14ff792784ed4bfc9d56f1d1e7482169 : SUCCESS in 1h 54m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/70236115a40a43d4ba4a91817312e8ee : FAILURE in 54m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b00b52df423e4a86a3535d37798089af : SUCCESS in 1h 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"637dce3d06a9d4b66c3fc5d962cacec95235053c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-16 08:45:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a1bf134d7e85f143f21b22dfef25e2f6dd3140a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 10:46:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e1d372c54144386bc8470fa5dfffade\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/604db2c110f544d2958d028207960a96 : SUCCESS in 1h 52m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02388f5855be4cab883c0b5c59c0c363 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1decc41cceb24c92968a46a27bc8f25a : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baa4179136a248a1b5ba3bcb7adbf821 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12f7765dbb0a42deaa716f0c1b0c4d94 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8c49259429349779ed206e7150c5ee0 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09045b8817d64ce1be2110f0691d4b8e : SUCCESS in 5m 46s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/92b40226f91d41fcb238803c2efe5d2b : SUCCESS in 4m 41s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/7dbe0d6352e24900b3b6ff6102830b39 : FAILURE in 6m 29s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/0b9745f952f743bb986795480a9541f3 : FAILURE in 3m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/51920bc92d604567baa2ef5af9016a1c : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dea4579ffcc74ceb94f1edd509602e1b : SUCCESS in 1h 27m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/37fd60ffbcb3475c8c3175665520fade : FAILURE in 20m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/551dba1835174c37a88806f1f7cd692c : SUCCESS in 1h 23m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/880af9c8db714ff689c93dc60cf4cb5e : SUCCESS in 50m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7beff28005041f69eb2a90870365b68 : SUCCESS in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e5e544cff438e6c1d9558b5727b5df3a4c2c6ea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 10:51:08.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8f4a76b5add167f65a93c5a724b2c7073cb348b","author":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"date":"2023-08-18 09:23:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a807c707cde1f20ea1a93a35536014b200b83ceb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-29 09:18:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5e8f14f3b6acd5fcbd4605217ccd8a3fa33a1398","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-08-31 09:44:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"953c95b07d445f09fdc1cdc260c91e20b2946791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 13:53:23.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9112b34c17f8fdef6bd8c7cb6d2f9d1514cb26d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:07:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1206662aec9c4cde863a6c493a4e0108\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/334e38741927495f8001291ef3291c05 : SUCCESS in 1h 23m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d96b4528f744b9f8ebc9e8608602a3b : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64cffb3640c2444780693bcd061cd17d : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef1b93dba8f1469682608db5bd17a8c5 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdd694c798c43c5898939d8e6c2f153 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92a7708cf83c460fbcb962a889b9d51d : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805deebc9f0a4ed48152ecf9967e73b3 : SUCCESS in 4m 06s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/1abbff2bdd08477cb6868636ab4bb9f3 : SUCCESS in 4m 28s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e5bc97e3ee28488d959ab8c7ac539f18 : FAILURE in 7m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/f63074ae7aac460a982c125795dfa941 : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/4c744a2256304e4585cb56194169d56f : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/047f9890411346b2a40afa1a19f6950e : SUCCESS in 1h 40m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/43a9ab2a405f49f09cc66ef58571d51c : FAILURE in 33m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/155983649dbd4e8487975f3d433cd539 : SUCCESS in 2h 07m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bdb4837061c440a99ed83e42a373140d : SUCCESS in 55m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/09bfc065636c484e9f1438b62fdba940 : SUCCESS in 1h 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7af81de9add64bcb00676c23abaa6cdb51beee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-12 12:11:48.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d3d9f804fbff2fac949cc6baad23a455bfbb4dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:47:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/276ae86b50424af29dfdb76002d79222\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e102bc1bf1154621b439900f4388bbe5 : SUCCESS in 1h 28m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef241b47caa1472d93efd0333ea70bd5 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81930fa47c204117bb79cdda4a917d40 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27981ed697034ab7b009033e1255595a : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b0716b2b1bf4a94a67b9387d39af877 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e63a84a391384eafb7be9a35365744bd : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142bf10d95c7480ea645fdba4cc61bb6 : SUCCESS in 3m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/67e544ba5ad0416aae981fd0d9274dda : SUCCESS in 5m 26s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4a9283cf4ed842dbabedb9dd6a98f4ae : FAILURE in 8m 21s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/2107c2f4acab4954a067ac872ac62d28 : FAILURE in 4m 14s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/f45b4cef95d54bccb644db706e3c5d64 : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7109979f4964459599c443a39b35b359 : SUCCESS in 1h 23m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c9663bdc7fd8444399f0cd8fa8dd56c3 : FAILURE in 15m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/75eede602df04d28a380eadebaa91dfc : SUCCESS in 1h 30m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/70c9415a74e64492a09a6168fd5ea629 : SUCCESS in 1h 02m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/228f2e0903044122bd325ba6ed7fd2c8 : SUCCESS in 1h 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b321e2657e580c6003aae44dbc06775d22fedf74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-19 15:59:47.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c6252f7f7f29e4a159e57d80dd516de16dabdab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 18:07:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be2984eb913d42eca15b2fcbf9ce9453\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5357683012a4a5c915995a5c39364d2 : SUCCESS in 1h 22m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f68c650e19ca482990395e0592c1d847 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59ba13caf744f7caa72a90124db5933 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6293a0b44524f1ab8547b8b0ffa3847 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71cd795dae424ff092cb39e3b8e10171 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf00d34e54d4fcb9895eb8692d543de : SUCCESS in 3m 58s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b6b75f3e9b5e4636a71887beddd5a6f3 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f97176fba37b4c66863b59af12cc1f03 : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/1396a1a005194ebaaf282ab500d957b5 : FAILURE in 4m 48s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/c59344ad57954eaf97b751238d275998 : FAILURE in 3m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/182ba64c3a5f4903b8ea01a5a7f48430 : SUCCESS in 2h 00m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1bbd30acc99746aebd2a28c064e8923d : FAILURE in 31m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e5bb69e3abf1499184aa05690bea30db : FAILURE in 1h 23m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/21da954e7278420ca4faa8d9767246d7 : SUCCESS in 42m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b09cd05321547fd963165b9b110e67c : SUCCESS in 1h 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f0e6126e1eb6383497ae4f6a1fba6cc2a82b069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 23:08:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"21f2a30d830808388435dfea5530b9d8a3ffd22b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 00:49:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ca895bb84942d49b83dc93eebc302a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9827b67a1fc846f78ae4f69dbf49686b : SUCCESS in 1h 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89fecfaefc41430fb5aefeac08b550c1 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306c9879aec34ab89f552676eedb86b1 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18ba6c1d44ce4e049643132fa07103d8 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ab2954f5ffa482190bece6d30a38240 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de64ce0916e4b4c8bdc1cfc4e442045 : SUCCESS in 3m 54s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a7e8e6bdad4d438aa6b625918d948df4 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5925cbf907824f32a5f446a9a948b9d9 : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/3e32d7732ae048729efff42f60c5d2ff : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/01c328da1b574eeb80d8d987a2d8cf22 : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6cde57f0222342bca5dcf039e708010a : SUCCESS in 1h 08m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2b1af11c60b643faa5704fa6adc1fdcb : FAILURE in 29m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7c8db2b213bd43abaf4cdc661edeac34 : FAILURE in 54m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/93f9252ce6dd48558457e081596bbea4 : SUCCESS in 1h 04m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/10bc9a0658984e8ab1d9aecf676ecbac : SUCCESS in 52m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"665e5f73e99a53cf1288d0dbdfae958a60dbf78a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 01:43:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f1d8aec57864395167d1f2882271a1a1dd78a5d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 01:44:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0817c423a1336f8f25df99d08c30589d6f5f6a48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 03:43:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c051dd41c568455fb555e1d6b60a8df4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eadcac056b2845129fe2dc328325922e : SUCCESS in 1h 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570af36cb9044dedaf5f240fd1b92de2 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb2b4e47585b4b52af5e12118634ea02 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c850ccad87946dc8a2fee3f9bb1ac50 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b3755d68aa4faa9807496a731981d8 : SUCCESS in 3m 42s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/729c6d6d4abf46a0ab9aef340a02128e : SUCCESS in 4m 12s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/003345e74bbb4142a1340f1570c85ec4 : SUCCESS in 1h 51m 46s","accounts_in_message":[],"_revision_number":6},{"id":"275fafdb21164966400b0bffdd052ef4e224e589","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 03:43:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6be5034b2ed50a7e6f701f79ad5af45378793530","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 03:44:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f1ba8936954d18aeb7f28e61ea985b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1269b162de4a178636198f47ac568c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3438726dd08eef03a0962c72096f268a54fd7f4e","revisions":{"bc4cc0a98250cbe8a43aea06a1879c551253fcfe":{"kind":"REWORK","_number":1,"created":"2023-08-08 18:39:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/1"}}},"commit":{"parents":[{"commit":"960f51ea999fb9a952c943203d459a8a009fb6a8","subject":"Allow creating transient queues with no expire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/960f51ea999fb9a952c943203d459a8a009fb6a8"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-08 13:54:07.000000000","tz":120},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc4cc0a98250cbe8a43aea06a1879c551253fcfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc4cc0a98250cbe8a43aea06a1879c551253fcfe"}]},"branch":"refs/heads/master"},"14025bf34d0c0b7294f973086395665f32b4fb47":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-08-16 08:45:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/2"}}},"commit":{"parents":[{"commit":"03322a46e01f17a07030caa2b2c4af80e99b841e","subject":"Allow creating transient queues with no expire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03322a46e01f17a07030caa2b2c4af80e99b841e"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-16 08:43:37.000000000","tz":120},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14025bf34d0c0b7294f973086395665f32b4fb47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14025bf34d0c0b7294f973086395665f32b4fb47"}]},"branch":"refs/heads/master"},"ac3231a3764aede42aac014f269af3e5685cbf6b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-06 13:53:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/3"}}},"commit":{"parents":[{"commit":"f080db183344754bb2d547810d1da231785f0de1","subject":"Allow creating transient queues with no expire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f080db183344754bb2d547810d1da231785f0de1"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-06 13:53:16.000000000","tz":120},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ac3231a3764aede42aac014f269af3e5685cbf6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ac3231a3764aede42aac014f269af3e5685cbf6b"}]},"branch":"refs/heads/master"},"bd0cfc004072044e8c49f413cc7958b047fc9967":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-12 12:11:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/4"}}},"commit":{"parents":[{"commit":"7f707fd73431eb8c24090fb8b4e515a87c7794fa","subject":"Allow creating transient queues with no expire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f707fd73431eb8c24090fb8b4e515a87c7794fa"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 12:11:43.000000000","tz":120},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd0cfc004072044e8c49f413cc7958b047fc9967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd0cfc004072044e8c49f413cc7958b047fc9967"}]},"branch":"refs/heads/master"},"c882d80d373d37a0d97fed34c012b65838483553":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-10-19 15:59:47.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/5"}}},"commit":{"parents":[{"commit":"86b67a7bb3145afa46b73d8c8d7e26f7d12de375","subject":"Allow creating transient queues with no expire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/86b67a7bb3145afa46b73d8c8d7e26f7d12de375"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-19 15:59:33.000000000","tz":120},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c882d80d373d37a0d97fed34c012b65838483553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c882d80d373d37a0d97fed34c012b65838483553"}]},"branch":"refs/heads/master"},"3438726dd08eef03a0962c72096f268a54fd7f4e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-11 23:08:39.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/24/890824/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/890824/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/890824/6"}}},"commit":{"parents":[{"commit":"f455edd60153116397a4c240dd1194861cb0bbf1","subject":"Merge \"Bump bandit and make oslo.messaging compatible with latest rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f455edd60153116397a4c240dd1194861cb0bbf1"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-26 08:34:19.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-11 23:08:11.000000000","tz":60},"subject":"Add some logs when sending RPC messages","message":"Add some logs when sending RPC messages\n\nRelated-bug: #2031497\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I7d0c318624d3d02182392ca3f06eed04d4133728\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3438726dd08eef03a0962c72096f268a54fd7f4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3438726dd08eef03a0962c72096f268a54fd7f4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
