)]}'
{"id":"openstack%2Foslo.messaging~890825","triplet_id":"openstack%2Foslo.messaging~master~I5056a19aada9143bcd80aaf064ced8cad441e6eb","project":"openstack/oslo.messaging","branch":"master","topic":"bug-2031497","attention_set":{},"removed_from_attention_set":{"28627":{"account":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"last_update":"2024-01-19 15:24:44.000000000","reason":"Change was submitted"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2023-10-18 09:14:38.000000000","reason":"\u003cGERRIT_ACCOUNT_32755\u003e replied on the change","reason_account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-16 17:58:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-01-15 09:08:20.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-01-19 15:24:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5056a19aada9143bcd80aaf064ced8cad441e6eb","subject":"Add an option to use rabbitmq stream for fanout queues","status":"MERGED","created":"2023-08-08 18:39:08.000000000","updated":"2024-01-19 15:25:45.000000000","submitted":"2024-01-19 15:24:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890825-bug-2031497","meta_rev_id":"c69595354f2a18aabfd6b07d60a4c378fc1627c6","_number":890825,"virtual_id_number":890825,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-19 15:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":2,"date":"2024-01-16 17:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-15 09:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":1,"date":"2024-01-16 17:58:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 19:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-18 09:36:53.000000000","updated_by":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"reviewer":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"state":"REVIEWER"},{"updated":"2023-08-24 13:42:46.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-08-29 09:24:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-12-04 06:17:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-01-12 16:51:55.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-01-15 09:08:20.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-01-16 17:58:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"902e2406bda34211415d2ba600f1463789b54de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-08 18:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aba66d08c03a1ee398e8a4f447cac24b011c2ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 19:25:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06c9eac2283d4e8995a9c4a3631041ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/290ee5a5410741cbae2a13c6fb399cf8 : FAILURE in 16m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc36c2252bc24a4a9bd9e2fae91f16fa : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2522d27d9a4405ba88725a3b646a39 : FAILURE in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c9ef4b7c14046ecad535918c74cdd4d : FAILURE in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29e17f4c14704e8da88496dd1648e489 : FAILURE in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fee646ac760b49bb90afde4166f2c40d : FAILURE in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da023495828148a9a7cef285d4b270e8 : FAILURE in 3m 22s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/84efdd7007bf40bab44e050a87d7b264 : FAILURE in 4m 25s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/2de772b06ba04b4bb955c860f2095e54 : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/40c81050ef3c4bf8bd623d3abee34118 : FAILURE in 4m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1a358624d9f24fd99e63bd8455d25565 : FAILURE in 4m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d9181b57f48a488ca80c02c80d9672ea : FAILURE in 25m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3c85b7e330a14626b29403cf0d59db0c : FAILURE in 16m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7a118ad86a0c42acaa06bb96e35eb73b : FAILURE in 28m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e764e590ef1c4d86ac53ba74c372f906 : FAILURE in 41m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/79f05274b91a49a5a01243f6d7acdfb1 : FAILURE in 43m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/_drivers/amqpdriver.py","accounts_in_message":[],"_revision_number":1},{"id":"34e8294bd260abc0db1f9912488b641bb3f5ff75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-10 09:07:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"08e3a47082d07052ab712802c4345caf0517dd3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 11:00:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05bd212f4dbb4e3b8f07a7935a11eb5d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ee1422b88334cf1a455bde2c25470cf : SUCCESS in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9280178c29454beaabd5b4c3afed4918 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc1aed4e0914391a507f430965329ca : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cf22a0ab3bd450699dbae845363fbc0 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d73b2fc152f4668b764ed733066a911 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3efa8a51bebb4bd0a895b50448421f4b : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77635d92a5f64aa58f567dab9e458437 : SUCCESS in 3m 43s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/23a642d9bf7e4c75baa8af68dc09b930 : SUCCESS in 6m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/13e061d4819e4378b67ccebb60b3c7ce : FAILURE in 5m 48s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/47bef8b20b01405787bf190f09150910 : FAILURE in 3m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/4fa5892d9705465fa641f0355345347f : FAILURE in 4m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8bc0b8dfa1f743a7a14096c61bd74d25 : FAILURE in 1h 27m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/080f37e3e21843e9b11f51d4183553c7 : FAILURE in 27m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/61f43fcc95bb481d9892f9090b3fae0f : SUCCESS in 1h 41m 23s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/68564f25229f4f5e856cbec8a0eeafb8 : SUCCESS in 58m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea10dd1fc0a34a408fb29bb4eff72edf : SUCCESS in 1h 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1518a28245326ecc848b3a9c760d53c4c726e8cf","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-11 17:31:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d055ec36f180498893e1128563e55b4f17ed0c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 19:18:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f12162bba2b8404ab335a4067abc3260\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/616acbc258614bf69150d76cd0458ff9 : FAILURE in 1h 10m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c2d1d1b39f249b481e636411fdca7f4 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00227c4c4109439b9bbbc53b27b93e49 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7965c54cc9044914b6ca5770db81b94a : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce47fb18da24da98de8569d152e4d0f : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba209cb196174bea958c94a94525b760 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a7ae4a0a4b405284d144c538015d7f : SUCCESS in 4m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b397f54fa4d84853b8d8169841cb6e1f : SUCCESS in 5m 00s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/58ae1fdd90b346c69e66213f7e3804e1 : FAILURE in 4m 48s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7312ec7072d64fab93807f40e1e59cd7 : FAILURE in 5m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/6e0f6537fcfb43c1b3fde031a609715a : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c13f65ea2130494a8a30c136e691e931 : SUCCESS in 1h 39m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2cf27406729a49ecbcfd0387a32fe884 : FAILURE in 28m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/926ec1c200ec48fd84eb9ed1f4c0a2a9 : SUCCESS in 1h 20m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cfe7b7ee180e44658df21ebce7cd7dd3 : SUCCESS in 46m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/362241a7696b440ebd7006f3b26436f8 : SUCCESS in 56m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a8b7bf060ecfa570ae0b4e24eeea0311316de52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-08-16 08:45:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a123a5fa03840fef0108758d29d98d771a474784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 10:38:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fcf392042d9488ba063530635e2d5f7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1688deee277469588735d469fdba6d8 : SUCCESS in 1h 41m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30acafb3f1e84a908eb2824c10559675 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a27272a5c243a4950fb40b51e8570e : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc6cfbdc58094734a73986fe08e6df2f : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99e5e7ff3bba4923ac3066c18b968330 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9642050041fc457e866ec6a7d0c1f8a1 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8b86d3a919412b893dc4131628b447 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcb31e6b6fc741da97b2462579a3af7a : SUCCESS in 6m 16s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/7eb9cee3260a41339f9111e0849072a0 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e566dc4fcbe646d69bd8f317b631fefd : FAILURE in 4m 20s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b0284ad7f98b4e20b534cc4a0da1bb66 : FAILURE in 3m 58s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/498e46d6b3a24b6390cb330a32cf4a55 : FAILURE in 3m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fb60fc6c73a044debfcfd6aac1f9f3d4 : SUCCESS in 1h 44m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1cb1f02898234e2699855e5b98b5e337 : FAILURE in 30m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/eefc72e946084d9ebe8555515a4c102b : FAILURE in 1h 31m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/df7e71d0dcd340359321f8f819034f6a : FAILURE in 46m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5eeeb74c32a642588b0034e7e1479ea6 : SUCCESS in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43b22d9d12d32d0d76e7043ed8b50e18258af37a","author":{"_account_id":28627,"name":"Julien","email":"julien.cosmao@ovhcloud.com","username":"jcosmao"},"date":"2023-08-18 09:36:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"394191949dc2db388409eca9ad4bec976ce6f14b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-29 09:24:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf56da44387802d6da52cc48cb6205c70a657f0a","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-08-31 11:44:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32aa2f99d2fbb0c80640543d05be73f0966dc254","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-06 13:53:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"72bd499206b9389b7d6fbac2945a49a7b39504d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:29:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdb720a875974175baff4df16a383e0b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a12680cf6d7c4013914184cbc24a9132 : SUCCESS in 1h 27m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea8558ba34541cdb6dc21249fb59f3b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/612fc65bb0554b54be97ed25536a9efe : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/985231738cbe43dab005b3db1cd3262c : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2823316422e4bedb886ac78f6cebc5b : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28111960d48246138bfa08494f7d4289 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac05a9650f4944aba1488fef20652ec6 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79cc79024d4c41569e94db8a88f9749d : SUCCESS in 3m 44s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/381df3892e674da98cf3673a2d15a4eb : SUCCESS in 5m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d9333a7f78994439a57def155572adea : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/29018fa68c9f4dc0962bd2615636085c : FAILURE in 4m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/729879a763a64cc6a8eedabc47639186 : FAILURE in 3m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3f2db5c205fc439aa75374cce5baa614 : SUCCESS in 1h 25m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4f8143fbf1be4b259cf799d444864cad : FAILURE in 27m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5457533a46694a2d990cd15ed266f346 : SUCCESS in 1h 30m 23s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3e2c0e622b6e4fd29e066d022467e358 : SUCCESS in 1h 13m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/584ee2a264ab4059a031c15a4eb0351d : SUCCESS in 59m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2fa6eaf6ac876e199dca1da157266bc857d73644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-12 12:11:48.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fda9307560c084e87eeeb3b60154930f6991bbdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:50:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63271497ac147eea8c8cd0d685822d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95764fc2b15148128b43eda8c6b906f3 : SUCCESS in 1h 28m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f71554dd2d9045c7817dc4398a703e6f : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78339730c2bf4dfabb513b1030a4e1cf : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c8a258ceed3454ab1ecc3cd234feb94 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecc0e582d5bc43b28f1cb64301114f33 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c27feb0dcc60459290623c1cd082ba64 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77e767055824deea2e2e5a6d4eb20bc : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7f6402c7f642599db38991e5d3a53c : SUCCESS in 4m 14s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/75e8b048d1cc4efa97bd7d3704baf3c9 : SUCCESS in 5m 20s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c4f9a6ee56fc4a98bd5b2e55d550f762 : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/336d1f60aa584cfc8d20d9aa2ff7865d : FAILURE in 3m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/df6c4349f52d4768b513f55756f169e5 : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/167ff00f95e84348af971ee6e86dd4fb : SUCCESS in 1h 33m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ef85b25617504758a97e1909e4589a65 : FAILURE in 28m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d38adb2a0ab747e8830ac8cb55d20d51 : SUCCESS in 1h 33m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3b32277048ee4ac4a2e098f14ea4c2da : SUCCESS in 1h 01m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/12a13ee4f69f4de98c393c10a6377e35 : SUCCESS in 58m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"36c4e5fbe483e3e41953a46c8b291e82887b168e","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-26 15:07:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2048cbf6afc1c29312fe6f31959aac8a106b4285","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-10-18 09:14:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95fb0383cb28d5746abad041184a4fcf857c5cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-19 15:59:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"cdb80d84d9faf23659dcc4261fb1619da9cf9333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 17:43:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/799d1bb7dd6a4a73b539d8401db815c4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c257829b7d67424dbd406c1a1641acc7 : SUCCESS in 1h 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dce59fe798b348c6971a4952cb968a4e : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e946ebaee819473b9bdd9fd9a79c38ba : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3647ca3af98429690f8da48680bffa9 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ca2539554784fbc970665763d3fbaa8 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf3662cdf2d416ba90fd7675f23b7f4 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86c5e111b7e7433baf948ed4608d7bdf : SUCCESS in 3m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c94748467a06446cbf0de5d5757494f1 : SUCCESS in 5m 04s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4c969eca48244736ba8f448724639d06 : FAILURE in 6m 42s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/c51f4cdacef442aeac4b86efbbb116c0 : FAILURE in 3m 48s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/550d39ae15d241f7995a339de3b24e46 : FAILURE in 3m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4674f2a0fb9e4230abd18555fb916cf6 : FAILURE in 1h 16m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5b2ebb0017b64c809cfae4f6e5ebc62b : FAILURE in 29m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6048d195d3f546209936087b7977c525 : SUCCESS in 1h 22m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a2ac7fb7d57245cd80e68111911b1b71 : SUCCESS in 45m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c7173046f5541b48430dcaf00506a54 : SUCCESS in 1h 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d081d12eb3107aba4aca1fbc44f12d3cac63f118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 23:08:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a4fe558fae2125710daf88c139bf278fe092330e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-12 00:53:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a546e08365d4fa9a3c3db32d13db9fc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/628446d1f0814204aeb901830f6ee81a : SUCCESS in 1h 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d269b25ca654af2b36a7ea65d5c204c : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f486d8dc388448e1baac7cddaf344730 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ceaae3a69c47e394da28c259988c48 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b26d6e3f0ef94548bebf27e5c0796027 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d89ead9c2a24b0da84f494e05edd0d9 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb801eca1f894213b3bdc189a4652fb8 : SUCCESS in 3m 13s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ff825832b27a467891ed7167061c3d91 : SUCCESS in 5m 08s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3dde1e5b0dee4224b64ef5650a2f72e0 : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/d163d0e3d2b34e85bdbccdeabaa58a8a : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/f914586c1b48476bb73820ffdf1ff4e5 : FAILURE in 3m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fc48f9a7000d4ba5a9a8edf318de32e4 : SUCCESS in 1h 35m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9652c4215730495fbc452c5ea006fd83 : FAILURE in 28m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8050d3e11d964b0fbfb6f0339714a31d : FAILURE in 1h 34m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/501d2850937c47f2a73b61fca103d784 : SUCCESS in 1h 07m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fad07ba7c914188b7d02aabfb07d413 : SUCCESS in 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5d3bc947a856977b878ec22187f432a204f0859d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-05 15:01:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9e9e4c5798e5b5fef5d71dd012770e9f05720ca4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-05 15:02:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9cebb7cf16fa92a78ed900a69e074216b1cddae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 17:36:58.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4783daf27bef43a69adee1ec9d18d7bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e45b91b8ff6479db8df1288b315f35c : SUCCESS in 1h 30m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6648591f332740c38a164507de46d696 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5781ce3daf2a42eb9f5cba99a6a01cf7 : FAILURE in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d266a6f4dc34d99842d561520b8e51a : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f83a164f44624336b8d40919d6988813 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22abb6f9f0944c2e94ef75a7debddd85 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/901e1b8aba7e4874820b876f45fd933a : SUCCESS in 3m 56s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/24646a8e28204fc98a0f058db3e33809 : SUCCESS in 4m 49s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c0e7b2034ba640d6bc4b4f0310ea6c0b : FAILURE in 8m 03s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/2f3b117680af46a1a07a3739bb683988 : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/4a8dd3cb303147d390bcbfdf60e746af : FAILURE in 5m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/010115ff2e88477786fbff83359655ee : SUCCESS in 2h 27m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/40d86c280f06492a9e65c5f92d314981 : FAILURE in 26m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d886c651ef9c48de8242193477142f3f : SUCCESS in 1h 50m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3f4e8690bab64d2dbfb98f6260f81ae3 : SUCCESS in 58m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a57f62963454916b0f777bae2dc9f60 : SUCCESS in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ca3db2b9d76421a611fc9e527c022211da9ce5c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-12-06 16:40:28.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"713039ce86ba06d5addeb0dde9660f4e184f3bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 18:29:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743da214d9f2416cb0b00356f4a34377\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afb4885420574e3ebf31e1a1aa4106f0 : SUCCESS in 1h 42m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d0e8d290fdf4ac3be2e56e49a4dc1ee : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5316a2d335b74f5782b1ad38436b3ffc : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db52cc44cab5445c85ec2201096d3f4c : SUCCESS in 9m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c76936057b44ccaa976322eb1b83eef : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9628c1edbcb49fcaa68fe0efedd6230 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bfca46ca2c64b839d6f9b69bef274bc : SUCCESS in 3m 48s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/19f43eeb20384d70bcd670d8beb536ff : SUCCESS in 6m 02s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d526322884de4a2f8616534e15485bcd : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/9555bafbefe145e5ab266fee9797336d : FAILURE in 3m 42s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/d8d8e91341024e099904426983492877 : FAILURE in 3m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c81e58e43852483b95f43c68578af1ed : SUCCESS in 1h 26m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e3363a6659f74b92a9c1afb077b75ded : FAILURE in 21m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8e3d9765d77a4b65869ce1388bade36b : SUCCESS in 1h 26m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e3e1b605480d4eafbf7c7ef3e64dc684 : SUCCESS in 1h 05m 27s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7e5f63a13ab40d2b004d91514eff54e : SUCCESS in 1h 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a0aaf6647102161f7c95b10f382f80f62bacbccb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-12 16:51:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a05a950972934aa5d3537af752c36077eb05d7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-15 08:23:45.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"086d3b7abd67a2967fc2da067e53054c71ac43e0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-15 08:24:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"56bcfba71b76f4c1d7d910e0c90312c6ae6ecf9c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-15 09:08:20.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"788512a68b732c603a47c606bd03931b6efcaf00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 10:18:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b7b77e449414bd6973626a92ff0d4ab\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e63f3fd8b904f9c85ed3035713fe159 : SUCCESS in 1h 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16306a6f984c49b58a64776785c8d9a2 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98cab4e5641496c890002d9eaeb4fd2 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d06b3102fa86418eb1e6342e5204c00a : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e572b45167524e55b70e12d56cda97ce : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/611bce070cfa490c85b42f17f833e90d : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ad86fdf2fc4733a095fe66c78deb45 : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a64fc7fd86144be9ad9f56e6f28be5c7 : SUCCESS in 5m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/af04204e031b434f8bd35c035187d970 : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/31d43d6ef5844085a4325997313d0b37 : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/b5cbd73fd40b410681df080cdcb5d490 : FAILURE in 4m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/eb34c0548c37435da22b7c37d0ec4306 : FAILURE in 1h 03m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ad806c066694176931d82f297421b1a : FAILURE in 29m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fd4ff661f6a64910bcdbb1a747cccbfd : SUCCESS in 1h 49m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a17e3b91fa1045b9aa02034da403a11a : SUCCESS in 54m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/12c06d2a1bb442b4b482e3dfb6848c2f : SUCCESS in 1h 16m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cc158d651404f66b040c0288f7df69f538fe092b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-15 10:34:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c17d67e40a8ed927f34782e75519eae7f1330c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 12:44:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f77165330034a71b53482d86bb2aac1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/741c1ccbb7b345e785a05a0313ff1ea7 : SUCCESS in 1h 19m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebb5884693a840d2b4655fc52c054c85 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883699db7cce455f8936e276384adecd : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6aaee0318c748f682dbfa9681d9a04a : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18245500a96e4d95bbea3cc2db3d93ce : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69e5cb2b85b467483417313f82e2e92 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3fca826edc44101ba70cb5dec2f34a9 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/09fd4f44cf7741abbb0c4907b4744fad : SUCCESS in 4m 48s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6b46ee8f444f46249f9f60c287e08dcd : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/5562553c31f0437791103b7800e7c3bc : FAILURE in 3m 20s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/95e91057e0674c959ec94299e48f9c85 : FAILURE in 3m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c4b377bd17cb4a4484235b9045144285 : SUCCESS in 1h 38m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3e415c2838bb4c73a0552d0ec751e3f8 : FAILURE in 29m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0b3d4ceda75b4c7bbe7925bb8ef2ddb9 : SUCCESS in 1h 46m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bca255b601544f8c9141c2094fe681ad : SUCCESS in 1h 04m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8f2a38cde8449c78fe36ea5ed163310 : SUCCESS in 1h 12m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9fea7a8943fb58bbcf137e69a65e3543e5dab782","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-16 17:58:12.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"19604909a39f02bbb276fa69c5715fce594c9171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:58:46.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"6a6c987e3afd6cb3427eaf62094b7bdbe1caef70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 19:54:43.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b5932d47e748c5972f26ea7e854230\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e839a35e200e4006b686d56f1c07a0ca : SUCCESS in 1h 49m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c191eac7186044aab03c4bd5ab422bb0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/289915595b49457696840469bff0f7b2 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8da762fc6b34fc0aa7f622db6165e0f : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990851a85a77453398c8481a52667694 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8441671f16cc4d7db07afea485e33862 : SUCCESS in 3m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/5529efda3e8347dd9f793a12eac18682 : SUCCESS in 5m 48s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6fe37ab038b445649b262d45898776f9 : FAILURE in 1h 37m 50s","accounts_in_message":[],"_revision_number":10},{"id":"c3c828f303ac2b6d3e7a99b1729a50baa141aab7","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-17 10:46:06.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"500ce4577d65a02f0dd77d6125ca2dc89340c984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 12:34:59.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cb6bb3423de47f2a0a45dd5b6f1c6ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e0fd48775e14c0b98a59e60d5ac1fe7 : SUCCESS in 1h 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/107f1bb9f99e4ba48a66b90af4f90bb1 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930792bdfcdc41ee844c716f953704d0 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c9a3b3d15f241bfb9f4ac10f364dbc9 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2efd634d1159423282561847cb4cd236 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf7969bab9a4f7cb6d20f148f18a5d3 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a7a0bb26944f25bc026058393e7b7f : SUCCESS in 4m 14s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a9d0445c4ed34804a1f5160770c11539 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8318d75153d048569449dbacdd1c4708 : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/a6ce7ada77a64cd49beb5577f21f9dbc : FAILURE in 4m 06s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/ac15fc72d6204d8892b00530472f1066 : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/12e900b8fe1c42aa8c257e1f32cb921e : FAILURE in 1h 33m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7e4783ed80f04b3ca3e764d491a44856 : FAILURE in 20m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/46f5a707cf5140deb05f4898c9cc51b1 : SUCCESS in 1h 34m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cfc053bfff46442c9560f3ac42206a34 : SUCCESS in 56m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/977f9dfde7ea4d39aadbf69a6ab2d975 : SUCCESS in 1h 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"752bd45d5b2466b4869323ef2c72dc6e68ae1095","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-19 11:51:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a642c516e4a70517261a6a85ff80bfcf5fd53e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 13:38:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/798b274d688549a29a9879795ea0f555\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88706d62cb3845b3a5c3b7ecf2a93bbb : SUCCESS in 1h 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef46c7777bd14b8ca3b916d20ce094d1 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753eb064d4ce4994b84918342cdd7a1d : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83ae6afbde943b2add667d407792929 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/047757abd7c94c538f680ba2fbf621d8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b018353e2b08429b86be5dc61fd13848 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d2cabb89f344e7e84d05673b001b700 : SUCCESS in 3m 25s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3fb41f18fc164bc79d5ef78ae20ee79d : SUCCESS in 4m 48s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/501d73b7df4047aeb9da330597d03f38 : FAILURE in 6m 20s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/ec29d1fdd4274032b34edcb2b23aa426 : FAILURE in 2m 43s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/574a8ecaa54848169b44121e17724209 : FAILURE in 2m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4fedeba2e55440f4a5a25b9ad1225678 : SUCCESS in 1h 15m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1662ba25b45c4f8ea42dc6f0575eb1a8 : FAILURE in 18m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8b169caa46bc49a3821490236e22b912 : FAILURE in 1h 09m 28s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/422bfcdcda424e0a9e947ee5d44e115c : SUCCESS in 1h 01m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/165b33e695ef4108b7e7842b85792315 : SUCCESS in 1h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6c6fcc66b18179d4d96312ac9c2a80c16dc8ce95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 13:39:17.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dcaa6cef0b0c140236b96d8b89aac1acde8f9ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:24:44.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c025ee496144bc96943ae73b9627ce\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8a69a0fdc36418bbe1eb3b91120c909 : SUCCESS in 1h 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3984d378e0b4ecc989d40c2912e0a93 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9948ecd760704c2f8c5879b055cc9536 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7011b962526c4b19b7646f5ce2205557 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4f2bbea65d4d61a0c839a5bea38591 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca8c16cb1a184476a81887d74d37a7a3 : SUCCESS in 3m 03s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9e2c9ace5aab42b9af3041ad4fa71c43 : SUCCESS in 4m 48s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bdbd4683c3434eb189deca93eff391e5 : SUCCESS in 1h 40m 12s","accounts_in_message":[],"_revision_number":10},{"id":"baa99e0f800dae1858d3455d3aa4deaafd75954f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:24:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c69595354f2a18aabfd6b07d60a4c378fc1627c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 15:25:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a277c138f9ba4ce4a30d3401b4013679\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b7e01893104bdba1ef9d61cd4c8260 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8852883e5174af88f835f9eb80c2992 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e95f334459d4dfd3778ec9e84d716f69f2c08ad5","revisions":{"b2398c056d80fd6a4e4a9289cc2c85e128b1e74b":{"kind":"REWORK","_number":1,"created":"2023-08-08 18:39:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/1"}}},"commit":{"parents":[{"commit":"9692220a90d2822b0dd065da5795e9ac40e31138","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9692220a90d2822b0dd065da5795e9ac40e31138"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-08 18:38:52.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b2398c056d80fd6a4e4a9289cc2c85e128b1e74b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b2398c056d80fd6a4e4a9289cc2c85e128b1e74b"}]},"branch":"refs/heads/master"},"2a69394890e51b3978b5bf36bb3eb8cf3b31547e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-10 09:07:17.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/2"}}},"commit":{"parents":[{"commit":"e6817d44ee206466a1187f7f6272c0631986d396","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e6817d44ee206466a1187f7f6272c0631986d396"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-10 09:07:09.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a69394890e51b3978b5bf36bb3eb8cf3b31547e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a69394890e51b3978b5bf36bb3eb8cf3b31547e"}]},"branch":"refs/heads/master"},"adeeadb16ba774ab16caf1ed13cddac89d76bc27":{"kind":"REWORK","_number":3,"created":"2023-08-16 08:45:09.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/3"}}},"commit":{"parents":[{"commit":"0d9155465db227af15cb5c456da0cb8c88b6c526","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d9155465db227af15cb5c456da0cb8c88b6c526"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-08-16 08:44:20.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/adeeadb16ba774ab16caf1ed13cddac89d76bc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/adeeadb16ba774ab16caf1ed13cddac89d76bc27"}]},"branch":"refs/heads/master"},"0b18ccb09d0371a277a2d20d7726865baedb3926":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-06 13:53:23.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/4"}}},"commit":{"parents":[{"commit":"565ea670acc5a24825b1550604ef18990fdd35fa","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/565ea670acc5a24825b1550604ef18990fdd35fa"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-06 13:53:16.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0b18ccb09d0371a277a2d20d7726865baedb3926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0b18ccb09d0371a277a2d20d7726865baedb3926"}]},"branch":"refs/heads/master"},"bc311ad9f7fb135754a43591ccebee69097e0b96":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-09-12 12:11:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/5"}}},"commit":{"parents":[{"commit":"07b08183f21aa905fe87bb7df0b8e2849a78cd0a","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/07b08183f21aa905fe87bb7df0b8e2849a78cd0a"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 12:11:43.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc311ad9f7fb135754a43591ccebee69097e0b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bc311ad9f7fb135754a43591ccebee69097e0b96"}]},"branch":"refs/heads/master"},"b01e208c2df6e4af134be9d07a59d0f13778c5a9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-19 15:59:47.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/6"}}},"commit":{"parents":[{"commit":"1f97c4d4832bf07508ed66902bdd2d65fd5d4ae8","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f97c4d4832bf07508ed66902bdd2d65fd5d4ae8"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-10-19 15:59:33.000000000","tz":120},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b01e208c2df6e4af134be9d07a59d0f13778c5a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b01e208c2df6e4af134be9d07a59d0f13778c5a9"}]},"branch":"refs/heads/master"},"c9585df8eb52eb490540f3a553d1b8c502861905":{"kind":"REWORK","_number":7,"created":"2023-11-11 23:08:39.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/7"}}},"commit":{"parents":[{"commit":"4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-11-11 23:08:20.000000000","tz":60},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9585df8eb52eb490540f3a553d1b8c502861905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9585df8eb52eb490540f3a553d1b8c502861905"}]},"branch":"refs/heads/master"},"40b9b1dc396e36c245450c5ad5ae815a5b3f9731":{"kind":"REWORK","_number":8,"created":"2023-12-05 15:01:30.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/8"}}},"commit":{"parents":[{"commit":"4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-12-05 14:56:46.000000000","tz":60},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40b9b1dc396e36c245450c5ad5ae815a5b3f9731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40b9b1dc396e36c245450c5ad5ae815a5b3f9731"}]},"branch":"refs/heads/master"},"97b0ed799b0ec34c13689f729b353c8855423f9b":{"kind":"REWORK","_number":9,"created":"2023-12-06 16:40:28.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/9"}}},"commit":{"parents":[{"commit":"4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-12-06 16:40:19.000000000","tz":60},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/97b0ed799b0ec34c13689f729b353c8855423f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/97b0ed799b0ec34c13689f729b353c8855423f9b"}]},"branch":"refs/heads/master"},"e95f334459d4dfd3778ec9e84d716f69f2c08ad5":{"kind":"REWORK","_number":10,"created":"2024-01-15 08:23:45.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/25/890825/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/25/890825/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/25/890825/10"}}},"commit":{"parents":[{"commit":"4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e","subject":"Add QManager to amqp driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4614132ad01d7e4559c4ff43c1e7dba1c1bdb81e"}]}],"author":{"name":"julien.cosmao","email":"julien.cosmao@ovhcloud.net","date":"2023-07-25 13:42:22.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-01-15 08:23:36.000000000","tz":60},"subject":"Add an option to use rabbitmq stream for fanout queues","message":"Add an option to use rabbitmq stream for fanout queues\n\nThis is introducing the \"stream\" queues for fanout so all components\nrelying on fanout can use the same stream, lowering the number of queues\nneeded and leveraging the new \"stream\" type of queues from rabbitmq.\n\nCloses-Bug: #2031497\n\nChange-Id: I5056a19aada9143bcd80aaf064ced8cad441e6eb\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e95f334459d4dfd3778ec9e84d716f69f2c08ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e95f334459d4dfd3778ec9e84d716f69f2c08ad5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
