)]}'
{"id":"openstack%2Foslo.messaging~891096","triplet_id":"openstack%2Foslo.messaging~master~Ic94323658c89df1c1ff32f511ca23502317d0f00","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-08-17 12:43:26.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-15 07:58:56.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-08-17 12:43:26.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-08-17 12:43:26.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2023-08-11 13:45:49.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-08-17 12:43:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic94323658c89df1c1ff32f511ca23502317d0f00","subject":"Only allow safe context fields in notifications","status":"MERGED","created":"2023-08-10 18:33:18.000000000","updated":"2023-08-17 12:44:42.000000000","submitted":"2023-08-17 12:43:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"891096","meta_rev_id":"caf1d12fc63a6c54518146bd272439b80d58c3e8","_number":891096,"virtual_id_number":891096,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 12:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-08-14 17:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2023-08-16 15:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2023-08-17 11:13:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-08-11 21:23:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 20:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 21:29:45.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-08-10 21:29:45.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-08-10 21:29:45.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-08-11 08:24:38.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-08-14 17:27:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-08-16 15:07:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f9ae60dc946d750f2db4a938f3fe44189b634cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 18:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e344bfb59af917e613d6891b20063341e3695df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 20:27:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a483f58c23904824b2a22a72f17a82ce\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/659f4299b25d4700bb19f2c55c96ae74 : SUCCESS in 1h 15m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e522803aa154b17bc0b4fbbe1ef416e : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded95eeaf58a4d7bbd534134d07ed655 : FAILURE in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53b6b514206b403fa81eac64952213ed : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50055baa82564600b301774a05ce7c87 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ae8b00f506a40d6a274b45e7f0dbfff : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0301dfcd3f9a4f2f907907bad99994d8 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a7947e2e0d0f43d0a10f81ae9b751ba4 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/663e1df921a444cd8d066e73704ee36e : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/24e73c793235471cacaca497149a3e72 : FAILURE in 5m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/076ac260ba2f4386949bb37d8c6b31de : FAILURE in 5m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5e500a199b324c0594511eef648070ab : SUCCESS in 1h 34m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e2ad43f59ea6426398bc039736448ad0 : FAILURE in 18m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/28ed4e19a53143d0a3b72bd9f94381e7 : SUCCESS in 1h 48m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/235d719cd9b84986a3a3f63b612fbdd7 : SUCCESS in 55m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/7927e79a662a45c8879e0faf01d98b85 : SUCCESS in 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b561b61476f232507789dec662e0c4f39e9605c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 21:26:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"30440a18ca78fb45f2849867d63601a322730b71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 21:28:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3423f155da94428a22df1c023836ca217ba2d28c","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 21:29:45.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab6fd2a297c1e02910714b65b60c34bd085cd2dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 21:31:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9c7d33874b60425b5384033cd2bc903330e3e3ae","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 21:36:21.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ea4875751e22a92118c96f06720a3e84fde7764","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-10 21:40:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c274d448d0aad26a1f89a8d23fd10e485efb548a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-08-10 21:49:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"289d91427e1ea7b1107df351b39affc1563b7a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 22:17:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73ef665e07be427f88871589b01939da\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1646e83ec7f45338155bb283a3d5b16 : FAILURE in 29m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf22051d67e74d26a5664a0a3b27a0bf : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27f6fcefc68446380069428254f0d90 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762fbbdcdeeb4f9aa07afc664fdbffe6 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07f00a42eb4e47e2b679ef0fdfe7c135 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41f72dc6682642a49d418cc527874d63 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47ba3355b7541c78dca51d5e024d1c2 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e7b3cb92960c44369597693d2985dc57 : SUCCESS in 6m 12s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/02c2f00d183a45afa126be45e4bc4d7c : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/5ee0f490af0646e2a96980b5ab6f5e61 : FAILURE in 4m 38s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/0e529cf07be54e6ea7b6a2babc58748b : FAILURE in 3m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f9b194b7ba924426906d9acb5ab32fcf : FAILURE in 20m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a29b4f15e9a748519aedd72479e7ed55 : FAILURE in 15m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/917acca7f777488eb75ea43a252b869b : FAILURE in 18m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e70ddf9c47fe43419a33ae0c80db403d : FAILURE in 36m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c4e8e1fcb6042afb520b3746a7ad87c : FAILURE in 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a2999b746f5e96fbbfd21de4b62b36e0a6dd009e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 22:17:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4f36a846a6a17d78fabc4a0b3787b6c27ec37e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 22:22:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5716a37951acbd5e2de10271aff92a4d2f12cd53","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-10 22:28:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"27c40d41583a6f8b7b31cf8920e51fd2c839b40a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 22:34:07.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"aace6acf5406a1259605bc7a6ef9f8d39bffb723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 23:19:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/153f1a9335974230b355046668857828\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a38ea1638ba458c9ec6ed1554566db4 : FAILURE in 30m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c84626e80948248f357cfc8fe26f34 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf97a0f9ae1941288dc10e7c4dd61f98 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ded7f6db3054e538c2ffda4747d6490 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a206356b8a554d3fb6946c53ac81c6b1 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c1c0bfb44cb4381ac35922e99d7a932 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d2fa291afe4cf48dd5acad4b09bed4 : SUCCESS in 3m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/af9ad218fa5c4de281ddce54edcfd694 : SUCCESS in 4m 44s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/1fdfd023e54e440da7e98d12f56787f3 : FAILURE in 4m 50s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/2a89eaf348374d65a312280947ae9c20 : FAILURE in 3m 42s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/0a1ef06158114e4880611d573a418d1f : FAILURE in 3m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/019362cec47f40008b9cbe12e5f03c4f : FAILURE in 16m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7ca40d1b5aa14dfbbf8b9a280bd7f2db : FAILURE in 16m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e5273015291341efb0c9af58a7d595c4 : FAILURE in 17m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/507e18928a0c42559c8b815cd79a6721 : FAILURE in 35m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ff3442b62894770bde2bdad7ab64b12 : FAILURE in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e0909b6c14de51ee4f06e8d5da7d5c7e05c142c8","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-10 23:48:44.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c21e1aa973eb3254bf9049de03f0a144c1dd202f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-11 08:24:38.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"27fee1055867f351d41f9c1b7b25cea262d195fc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-11 13:29:39.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4aac0529f7a908d03c582ff61db3c25439a5a470","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-08-11 13:45:49.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e632df00e373753807c814b8384ffcd5ac1c7db2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 18:51:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"51c8a972c0fa4211bf38821fcceefa1d9a934264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 18:52:25.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"5f97d7a6e99fb88b0b5f03b645aa735842c3e633","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 18:53:47.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6819797a87ed3762e64a451b8cc99a4610475dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 19:36:44.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c35e1e501a5c45431be331240d2a37946de2e40d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-11 19:46:59.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"70428f0f1599636b957f09b90f9aeaae3e11b977","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 19:54:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3c0e726595e9c6e38640a7ad115338c63bd1e548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 20:07:59.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"d495f707ce50cce3992bba9548da05ea7597f392","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 20:19:29.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"59d4f664eedc70b273858c5b7cfff6b87c787f6d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-11 21:23:32.000000000","message":"Patch Set 11: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a81db9f0b6e4a896b78a294325141521b707cff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:05:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ad22e842934d3ebce3c80879b83fd1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae6b7c2535174fc5942b74d1206bb62c : SUCCESS in 1h 57m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/471afbb9adf3489cbd17a93e99986c83 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6185366eb3d84cfdbbb0bd11ff6eaab6 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fa5b8915ac94a27a1299f2ef06b7862 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09ff12d575ee468abd323ea33ffd1b7c : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a520846842842ca9f7170916f0f3fd5 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8991ec0727594ffeada363168b0def10 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/08680387a06645cf918d64dd3cfafacf : SUCCESS in 4m 49s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/aaced989cc414e32a7245f97fe184d09 : FAILURE in 4m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/3ce2e65ef80e4b91bbcfcf2d88f463cc : FAILURE in 4m 17s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/fb39797f2b1a48c08137d0ca8d1f432b : FAILURE in 4m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/56095ae4475b421f8880f871a83a0cae : SUCCESS in 1h 41m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2411116365db461d8ed640bee99f2cb5 : FAILURE in 16m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4ccafc72cda04e7f96016f7903d14083 : SUCCESS in 1h 26m 55s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/87a47bbe2ef44f16a2d237e5df8a327b : SUCCESS in 47m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d11a4c3a92de41aa880088e36a64c6ab : SUCCESS in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d18c50a274a4aff63045a15517230b4509633296","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-12 00:44:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"81c3e05bee83cca73a88761561e8adae5cd32e90","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-14 17:27:23.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"03e0c30be9ea825d5cb6d283251b4dd3bf464513","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-14 18:37:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7351c44e13a73f3acb40ce06293a7b27615f081b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-15 07:58:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8413754aa98f4537b1b1e6067c7aa17c41073d3a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-16 15:07:22.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f43c498bd29c1e6a1e993c2901db9f6d375ae16a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-17 11:13:35.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"264594b0b9e64047ca96a2f96d96eadc1e9172da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 11:14:03.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"424e30fb264539914dd54c3c3f342a393c883060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:43:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aadc88fa32904a1e9dcce392d733c3cb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b1e645f83324548b19ab6596b6c63dc : SUCCESS in 1h 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46196465fae34e94b8fca1dc3ebe5c64 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47bca8db7be045af8cb0d8e795f7d769 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4f27598b8b64a008ed39df585edeb20 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d1d034d07b4ffd9e4258bafdfcdef3 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9414edef7d7748249c6a69713b683bd9 : SUCCESS in 4m 54s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c07871fb87a44f809fe918ce5cb2da14 : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":11},{"id":"25903e0b5bb279c63fefcdbf9e1deb9e02e7e527","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:43:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"caf1d12fc63a6c54518146bd272439b80d58c3e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:44:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09baf6d144bd491985ebc55a12353277\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776303f49dc1495490fab3ff86cc424e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"1b315615e7dc61dbf845bd663560fc8d5a18fa09","revisions":{"25c7f8e31f146064ef74bb4456cd82e1591948a2":{"kind":"REWORK","_number":1,"created":"2023-08-10 18:33:18.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/1"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:33:17.000000000","tz":-420},"subject":"Only allow safe cxt fields allowed in notification","message":"Only allow safe cxt fields allowed in notification\n\nThese comments need to be implemented.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/25c7f8e31f146064ef74bb4456cd82e1591948a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/25c7f8e31f146064ef74bb4456cd82e1591948a2"}]},"branch":"refs/heads/master"},"8ccbc5930e74a7c3d20be9a44e9969d4e80157f0":{"kind":"REWORK","_number":2,"created":"2023-08-10 21:26:32.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/2"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 21:24:46.000000000","tz":-420},"subject":"Only allow safe cxt fields allowed in notification","message":"Only allow safe cxt fields allowed in notification\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ccbc5930e74a7c3d20be9a44e9969d4e80157f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ccbc5930e74a7c3d20be9a44e9969d4e80157f0"}]},"branch":"refs/heads/master"},"fe3e6c84ea5a26ff31f0fc8ddcf26396162253fc":{"kind":"REWORK","_number":3,"created":"2023-08-10 21:28:28.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/3"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 21:28:24.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe3e6c84ea5a26ff31f0fc8ddcf26396162253fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe3e6c84ea5a26ff31f0fc8ddcf26396162253fc"}]},"branch":"refs/heads/master"},"2153f9a2da707af54118fb782d62d006b7f4defd":{"kind":"REWORK","_number":4,"created":"2023-08-10 21:31:33.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/4"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 21:31:28.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2153f9a2da707af54118fb782d62d006b7f4defd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2153f9a2da707af54118fb782d62d006b7f4defd"}]},"branch":"refs/heads/master"},"385ad14214342219d4a236173b209d8c599ea50b":{"kind":"REWORK","_number":5,"created":"2023-08-10 22:17:18.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/5"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 22:17:13.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/385ad14214342219d4a236173b209d8c599ea50b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/385ad14214342219d4a236173b209d8c599ea50b"}]},"branch":"refs/heads/master"},"a1614d967879a40d504c9b89314bc65ad1514a28":{"kind":"REWORK","_number":6,"created":"2023-08-10 22:22:52.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/6"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 22:22:48.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a1614d967879a40d504c9b89314bc65ad1514a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a1614d967879a40d504c9b89314bc65ad1514a28"}]},"branch":"refs/heads/master"},"cb9d5211f0ea53d4738b770ae870d3155fb8b77f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-08-10 22:34:07.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/7"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 22:34:07.000000000","tz":0},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb9d5211f0ea53d4738b770ae870d3155fb8b77f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb9d5211f0ea53d4738b770ae870d3155fb8b77f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cb67dbeba8f7f54e1a1e9ba8457671682569fd50":{"kind":"REWORK","_number":8,"created":"2023-08-11 18:51:02.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/8"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-11 18:50:54.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: https://bugs.launchpad.net/oslo.messaging/+bug/2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb67dbeba8f7f54e1a1e9ba8457671682569fd50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb67dbeba8f7f54e1a1e9ba8457671682569fd50"}]},"branch":"refs/heads/master"},"b87f4110386e6e664cc0a7fffa23316e6664ff07":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-08-11 18:52:25.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/9"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-11 18:52:13.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b87f4110386e6e664cc0a7fffa23316e6664ff07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b87f4110386e6e664cc0a7fffa23316e6664ff07"}]},"branch":"refs/heads/master"},"576cfce7e04f01cd5001aff43c9ba7f22b3c471c":{"kind":"REWORK","_number":10,"created":"2023-08-11 19:36:44.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/10"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-11 19:36:36.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/576cfce7e04f01cd5001aff43c9ba7f22b3c471c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/576cfce7e04f01cd5001aff43c9ba7f22b3c471c"}]},"branch":"refs/heads/master"},"1b315615e7dc61dbf845bd663560fc8d5a18fa09":{"kind":"REWORK","_number":11,"created":"2023-08-11 20:07:59.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/96/891096/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/96/891096/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/96/891096/11"}}},"commit":{"parents":[{"commit":"ec1c99d562659f1ca246ecdd75e810c9ece0140d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ec1c99d562659f1ca246ecdd75e810c9ece0140d"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-11 20:07:54.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1b315615e7dc61dbf845bd663560fc8d5a18fa09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1b315615e7dc61dbf845bd663560fc8d5a18fa09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
