)]}'
{"id":"openstack%2Foslo.messaging~892372","triplet_id":"openstack%2Foslo.messaging~stable%2Fvictoria~Ic94323658c89df1c1ff32f511ca23502317d0f00","project":"openstack/oslo.messaging","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-08-31 14:34:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic94323658c89df1c1ff32f511ca23502317d0f00","subject":"Only allow safe context fields in notifications","status":"ABANDONED","created":"2023-08-22 15:51:51.000000000","updated":"2023-08-31 14:34:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f57aceee66f81f44a27c4e4682e51858a2e9e89","_number":892372,"virtual_id_number":892372,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-29 21:29:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":-1,"date":"2023-08-30 15:39:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"REVIEWER":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 17:47:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 14:52:53.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-08-30 15:39:09.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"}],"messages":[{"id":"cfad505bbd8d5202b4e83cfa5727e39caf0dd44a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-22 15:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f0e5b2fd9cc3c729476259e03bddb31e0c440a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 17:47:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c84499dc6cf04ea4a626aba187e24762\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db616e32b7cd480daa8745c26990b35a : SUCCESS in 1h 16m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2652683f8e194fb492e1278ea9b76db9 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bac5e9d70214a9f8fcbf18c12a9f03a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a01b4907f8647878c18c8269d5554a8 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2629fd31d8b04b3587541f2e8ce9c423 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28867d92146a477eaa64cb7e79531fb3 : FAILURE in 4m 13s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/208696f2f42e44c3bbfc56454beb4d3b : SUCCESS in 5m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/253df374b3304511b2a1c6f9eb46dc02 : FAILURE in 8m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/7579e4100a444b23bc1f7a4fc19fa5d5 : SUCCESS in 6m 34s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/cb6d55cb3dee497ab5d0b212a16ee492 : FAILURE in 7m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3575f3cde9954a64904de38f331e14e9 : SUCCESS in 1h 16m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ae0c931c08274d539d5f4ce69d4327e2 : SUCCESS in 1h 49m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8d04cdf9520746efab6cd462a14e7ca5 : SUCCESS in 1h 17m 17s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/145f2dd9743d409ea8ee6187755f99fa : SUCCESS in 59m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aa5cd34c39041fab96ac783e639831c : SUCCESS in 1h 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58ac47c5fa47a77258a771f81e787b2cf23b96d6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-28 14:52:53.000000000","message":"Patch Set 1:\n\nThe commit message should use Closes-Bug instead of Related-Bug so that associated bugtasks will get marked as fixed and we\u0027ll get notifications within the bug when the commit is included in a new release.","accounts_in_message":[],"_revision_number":1},{"id":"c0d7ecaa9bd98c322ab619205299a4bc381ac28e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-28 14:58:08.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"27235e0a92ea8d4b74cf051cb72b779fbe1766d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:35:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/638721e4d92c41b6b230d03fc5f09da9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a057301858947c4b0b42d8fa92c989e : SUCCESS in 1h 28m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7af3927db1a74e7a9d84185c7896bcdc : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b76cd742bc4734ac66d226b1b028d2 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5671b8d793f94fac89c48034f36c059c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98a6590bd9a4725901a4368010c7f31 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410933533f6a4fc5a65a1596e810659d : FAILURE in 3m 50s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/3d105484e22f455f8d5dee86d466ab0b : SUCCESS in 4m 47s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/4617b1f06abd46fa88c9d8e7272736d5 : FAILURE in 5m 16s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/724c99f1645e472f8be57d3826b5a4dd : SUCCESS in 6m 33s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/8e3c75b71f81436a8b4e762f9a6af677 : FAILURE in 5m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a45146e2a1254320b926601fb9cc832f : SUCCESS in 1h 29m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/48f95685f09f4e4997c8dc796d67d9a7 : SUCCESS in 1h 10m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f916263aab7f442c8d0a9b3f4d9bb73a : SUCCESS in 1h 05m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8498fff580bc412687f7bc4a0d51a6af : SUCCESS in 51m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e0b9ca66f4845adb4e299f546e557ab : SUCCESS in 1h 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4b378aa965e669098d4e72b5e78d314455f627d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-29 19:30:01.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"da7c577037b2599cd5b0b6df149cbbc3c0ce4513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 21:29:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72fef29e8d134896915a4be0c0507f08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b802f06b638943609aed69cb56dd65ff : SUCCESS in 1h 24m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9c31e3d74bf4c8784cde9a7f44eb3e5 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e9821ade294e4a87b33e426f1fcd21 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5df0c56dcdef4d889822f9f7f553507f : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbfd61838ce0453fb5fc119f642ad02f : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a98bcc8d994400982bb19459ce5f2cd : FAILURE in 4m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/0f06d12b91af4f688a4892f165863d01 : SUCCESS in 6m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/d3ec543cf66a4e3bb6fd329d16e3a279 : FAILURE in 6m 14s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c0b421f22e6b4e93a0a7c19c71f67f6e : SUCCESS in 5m 05s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/816078e2c55a4bf38ef81e9b918489b7 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0ba5cb75b9e949de9a2882f88e157f86 : SUCCESS in 1h 08m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1ca0c7f6b29f4c00a970e50ba9027e35 : SUCCESS in 1h 17m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ca07aacd66d243c68b6c49757ff79431 : SUCCESS in 1h 23m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7d4f8effb87e4721bb9d2d3b036c2d4e : SUCCESS in 1h 13m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e982651e6fec41cdab78aaf24179f9dc : SUCCESS in 1h 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6418e7e5d149be2a8d3a0f5120c17ed219f4d753","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-08-30 15:39:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f57aceee66f81f44a27c4e4682e51858a2e9e89","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-31 14:34:10.000000000","message":"Abandoned\n\nGoing to stop backporting at Xena. This patch should still be OK for people running Victoria with no Cinder/Manila.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"abd3ae4ccf601fc29daf9b671b7604caf38a6d0c","revisions":{"2470886ec42e779e87234f5dcb0ff5e3057cc71a":{"kind":"REWORK","_number":1,"created":"2023-08-22 15:51:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/72/892372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/72/892372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/1"}}},"commit":{"parents":[{"commit":"391ce7fc69adc8d713d8ca64e76d901eb3e65df1","subject":"Correctly handle missing RabbitMQ queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/391ce7fc69adc8d713d8ca64e76d901eb3e65df1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-22 15:50:06.000000000","tz":-420},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nRelated-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n(cherry picked from commit 1b315615e7dc61dbf845bd663560fc8d5a18fa09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2470886ec42e779e87234f5dcb0ff5e3057cc71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2470886ec42e779e87234f5dcb0ff5e3057cc71a"}]},"branch":"refs/heads/stable/victoria"},"07b280a6f55383d196692b383ebcbe19c4483fd4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-28 14:58:08.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/72/892372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/72/892372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/2"}}},"commit":{"parents":[{"commit":"391ce7fc69adc8d713d8ca64e76d901eb3e65df1","subject":"Correctly handle missing RabbitMQ queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/391ce7fc69adc8d713d8ca64e76d901eb3e65df1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-28 14:58:08.000000000","tz":0},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nCloses-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n(cherry picked from commit 1b315615e7dc61dbf845bd663560fc8d5a18fa09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/07b280a6f55383d196692b383ebcbe19c4483fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/07b280a6f55383d196692b383ebcbe19c4483fd4"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"abd3ae4ccf601fc29daf9b671b7604caf38a6d0c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-29 19:30:01.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/72/892372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/72/892372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/72/892372/3"}}},"commit":{"parents":[{"commit":"391ce7fc69adc8d713d8ca64e76d901eb3e65df1","subject":"Correctly handle missing RabbitMQ queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/391ce7fc69adc8d713d8ca64e76d901eb3e65df1"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-10 18:28:32.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-29 19:30:01.000000000","tz":0},"subject":"Only allow safe context fields in notifications","message":"Only allow safe context fields in notifications\n\nPublishing a fully hydrated context object in a notification would give\nsomeone with access to that notification the ability to impersonate the\noriginal actor through inclusion of sensitive fields.\n\nNow, instead, we pare down the context object to the bare minimum before\npassing it for serialization in notification workflows.\n\nDepends-on: https://review.opendev.org/c/openstack/oslo.messaging/+/892873\nCloses-bug: 2030976\nChange-Id: Ic94323658c89df1c1ff32f511ca23502317d0f00\n(cherry picked from commit 1b315615e7dc61dbf845bd663560fc8d5a18fa09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abd3ae4ccf601fc29daf9b671b7604caf38a6d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abd3ae4ccf601fc29daf9b671b7604caf38a6d0c"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
