)]}'
{"id":"openstack%2Foslo.messaging~894068","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.2~I673ffbc67b161cf9b2993044bf9e6b0073d2b889","project":"openstack/oslo.messaging","branch":"stable/2023.2","topic":"create-2023.2","hashtags":[],"change_id":"I673ffbc67b161cf9b2993044bf9e6b0073d2b889","subject":"Update .gitreview for stable/2023.2","status":"MERGED","created":"2023-09-07 09:37:12.000000000","updated":"2023-09-11 10:52:15.000000000","submitted":"2023-09-11 10:50:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894068-create-2023.2","meta_rev_id":"8746b2a3d96344bde830780829d67a18bcfb4fd1","_number":894068,"virtual_id_number":894068,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-11 10:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-09-07 11:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-09-07 13:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-09-07 13:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 10:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 11:37:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-09-07 13:09:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"bb2b7d6421cc84e7c46b76b6c32a0b3e42814208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-09-07 09:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac30a1113985821bce29956453a60e7cde01d25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d24956dd855246edb81b2f5808cf1181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f133d55dbd00484eaa94fd9a2bfbd691 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fed50cfe33f429eb62482cbb35e8829 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6ed15c1047b4880917e16b7fdec3420 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/479222b41cb44d74b723711dd655fbe0 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f566cb8b11374994b11aa31fa748ed81 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1203e5f4735f4e6fac693c5bed0e7a49 : SUCCESS in 3m 31s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/240b44f4572d4daaaca9e4282b4ae7c9 : SUCCESS in 4m 55s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/45cf7f4195b44d8b998258023290a837 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/a79c3a467ec847cdb6cfb3a3621198d3 : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b9da40b79d584e5eae1321989c5a7f9b : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff7cd9c8c5e86a6630200ca7c81a37f60724f914","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-07 11:37:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"777e758c40a528424581d088f8a689b6690d1f51","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-07 13:09:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acdcbec9576b64dfaa18c122b75d77543b113eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:10:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c7aaa15b1b69e9640bc2e59613e26c96dab0f62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 13:20:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/707d72f6f77942ea82577fc0c4baa5e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dafa3ab39d74614ae3a2bba403d70be : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa90c4222e1a4e3bbc259ddbb502efc1 : FAILURE in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa4a9a0bcd5040cebcea0ce9118abb6c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874871126010443590c3e1b1880f4886 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/de1a28c0d1ca4398b647f94ec33f21e4 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"3bb127cdcea1e4aa36497def74f4d8d9133f54b8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-11 10:30:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7522d87a7eb8e5362abffe754eb68e9feb7c838b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:40:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72b423c63924f04b1d3f683317acbee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b31043de06b149baaaf685f807b63687 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e110470e635c41328511fc8a2eae1b90 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d7465e9efa148caa98aefcd1563153a : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03989768ea8a4d0c8577047277f78d8f : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cef7e45a74746679a5575fdb31dcd64 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488319c7e691495baede90ee6f18bb83 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/8728ce438c7e428ab6049a32debf78c9 : SUCCESS in 4m 49s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/6078cf43790c42289dd51c9aba97e72d : FAILURE in 6m 58s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/bb53214cee254456a848c40b10a5ad32 : FAILURE in 4m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/bf2880ed287a4f418681b9e4899b8078 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31fba46fdd6e767487b7a03dcc1fce5936f3ad81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:40:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5e1da7b810c52cc7ecdb7b9b400164437da46d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:50:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a745aa75d004297906392fd6fa30d94\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e636f676c541f4b55e7a2ab2fb4659 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb01bc5b82d4ec0a791f4f77f35ad76 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e15cb55ebbfc44039d110f5fbcaf2c96 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc15ddde95934d9690d8b2c0397b0b4c : SUCCESS in 3m 24s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bdeb040cbb9540efb94b1489337d88fc : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ee59df30a2b3c32c086b318bc93d3d6bde172f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:50:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8746b2a3d96344bde830780829d67a18bcfb4fd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:52:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47529ae6b31945efb38e09e95759b2a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f5befe7c4e47a88bbe8a2a64551884 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a3bd4add571e0c78534376712573e8b14f19e6c","revisions":{"6a3bd4add571e0c78534376712573e8b14f19e6c":{"kind":"REWORK","_number":1,"created":"2023-09-07 09:37:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/68/894068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/68/894068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/894068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/894068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/894068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/68/894068/1"}}},"commit":{"parents":[{"commit":"3485301b1838353c16c6dd1c1cfbcab866df2e5d","subject":"Merge \"Only allow safe context fields in notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3485301b1838353c16c6dd1c1cfbcab866df2e5d"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-07 09:37:11.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-09-07 09:37:11.000000000","tz":0},"subject":"Update .gitreview for stable/2023.2","message":"Update .gitreview for stable/2023.2\n\nChange-Id: I673ffbc67b161cf9b2993044bf9e6b0073d2b889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6a3bd4add571e0c78534376712573e8b14f19e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6a3bd4add571e0c78534376712573e8b14f19e6c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
