)]}'
{"id":"openstack%2Foslo.messaging~894731","triplet_id":"openstack%2Foslo.messaging~master~Ie5cf5d2bd281508bcd2db1409f18ad96b0822639","project":"openstack/oslo.messaging","branch":"master","topic":"heartbeat","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-02-19 09:02:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-02-28 11:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-04-09 13:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5cf5d2bd281508bcd2db1409f18ad96b0822639","subject":"Use StopWatch timer when waiting for message","status":"MERGED","created":"2023-09-12 11:46:25.000000000","updated":"2024-04-09 13:48:43.000000000","submitted":"2024-04-09 13:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894731-heartbeat","meta_rev_id":"d3a48e0762dae737c05f0f0abf5c1833092801f6","_number":894731,"virtual_id_number":894731,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-09 13:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-20 09:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-02-19 09:02:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-28 11:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2024-02-16 13:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-03 11:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 14:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 16:55:09.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-01-29 12:43:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-01-29 17:18:39.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-02-20 09:53:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"27028a665816c5603f24610962d03ab07e85afaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-09-12 11:46:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9d655a6360feca2bd89336c9093ea8eec158fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:02:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/847d682b7600439c9deaad4f2683440a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d42de107999419f91b7bb0b92577015 : TIMED_OUT in 2h 05m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcff620bff1d4564838456be29ebfc65 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5070f29b92314ecfb5d4fb1b5006bfe6 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39672a90a1ef46928d140978fcf9b2ef : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274291fecb0542fbaef0fa1f08b086fe : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5940a6cc4b084ff5b8d20c90685caaa6 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c83d3d823043a296794c8705104103 : SUCCESS in 3m 47s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9821aa73aae941f7af5e5a7a9fb8124b : FAILURE in 5m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8d0ef7a609db4d49a5cfd758d37455bc : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/324390993f2c486098b1214457c56b2a : FAILURE in 4m 09s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/2690c18f2e894a3690892901d5da9f82 : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3f1221f0d5fd470fb148e13ca8fa64f8 : FAILURE in 1h 27m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/75ca98f2a31e4e01b56c66ddf62560f2 : FAILURE in 30m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d8eb7826cd9b41d19ba92391f736e1d7 : SUCCESS in 2h 10m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f235c33f3cfa4b87b99d1aa29ead562e : SUCCESS in 45m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dd8682f115a428ca0915a8247b78a99 : SUCCESS in 1h 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"281046838a534ee74e6bca7dc3f938d424f41f60","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-12 16:55:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95d45398f75892f19be928838f22202ed03c91e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 19:26:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce1fdf413044a73bd490130efb5b1e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20a59c37f2bf4e48b20d72bd6f203272 : SUCCESS in 1h 58m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dceeb3421ee4e088a63aebfe6caba7f : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635525e2b73c4772b0429d23d95a9442 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5050ff21a38494da7dcb1e6965369dd : SUCCESS in 8m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/673eec654757461e866b436d0326fe96 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1815f70a4f4e278b2737536abdbfff : SUCCESS in 3m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/219ee5a6d4f34fc69c6327336d96de22 : FAILURE in 5m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/450e07a77bec4cc3828d71c2ff800ebc : FAILURE in 6m 32s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/2209c93d88a94721b73b673abd088075 : FAILURE in 4m 17s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/a665a3ceb97d4da1aed0dbc3b7a5fc6e : FAILURE in 4m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84a9f81b491f4eeeb78ab2b73d29b6cc : SUCCESS in 1h 58m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9f25075a84e44bfc8f8ff39ab4dd5aeb : FAILURE in 29m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/23a75c2595e240faa5e28a47db507b26 : SUCCESS in 2h 07m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/053d37af8b324123ad041c9ea3a7e6c7 : SUCCESS in 1h 02m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3007f27380e43c2bc1db943afa99c55 : SUCCESS in 1h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a1b1d82dd7dee14bae4d2bc9d47c7a70590d304","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-23 09:31:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fe9895c89f772a6b694d33ffd981470d4a50637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:43:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f95f928b779845ea97399ea5caff0384\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f7b7cff66e3490ea24afdb88cc99702 : SUCCESS in 1h 58m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4522152184f14fdcaacf2dc53e804a68 : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639fefd4e3e248158fb5507b3e62b1e9 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd59e408780f4fa8b3b8c27cb4da5853 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad35b86e672a420b8b6102d3ef0e6c9c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9dd59ae92714499acc7dd002b6d55c8 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/83bc0ea1861f48038f0c02e3eccfc1c8 : TIMED_OUT in 41m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7a8d64d366f0451da6f022ce1f23a8c3 : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/d42bb8e8d1564328bdcda120ff44003f : FAILURE in 3m 21s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/4da67ab83f194c3e9de8f8263c647a56 : FAILURE in 2m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0bc853360c85445e9193616e9206d147 : SUCCESS in 2h 06m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d40874041bf740768acba1070a1efde1 : FAILURE in 18m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/454674537b4e45b8b3957817cc01cf34 : SUCCESS in 2h 02m 59s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e93e6f4a36d0477a9395ac72a9bd30dd : SUCCESS in 59m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a11da79bbc9450ea05950abfafa4b09 : SUCCESS in 1h 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5760a5ff5cc360d2001319bfcb37cfecf9f2ddc","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 11:19:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59647a54eb694110c0cfddfe6160a324e9b1d06e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 13:26:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b1d29dd97bd43c48782e9067a2ca50d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89d7d7bb5af046f79479a9da6ce5c5e7 : TIMED_OUT in 2h 04m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a1cbe40e3d4f8987be8f46f8edcbcf : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb5971fb76e433a9c06dfa0ff63e2e0 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecbed7f1a7ce46d49f92fcf2d9187511 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/308395f1f0094c7e97d999e58ebf57e7 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ba692d279444cc9acf7afc04f31163 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/5e1b5cc0d0cf41a8b0d5796338b0aa08 : FAILURE in 5m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/076477f2772c4c8ba45acfd54517870c : FAILURE in 6m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/55a6fed430584394abff0422af3bd025 : SUCCESS in 1h 37m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b80a837f635942e092f96181f7b868b6 : FAILURE in 16m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b843aed1b59a4f6bac90fd6abff61067 : SUCCESS in 1h 32m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f98ad4aa9e384f75822608087e6b6bd6 : SUCCESS in 53m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0393aec6afde45f18a222ca2ec716587 : SUCCESS in 58m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39fa5926259137cc6c5c9dc6f7102310466e2e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 17:16:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9fea9d603801a9b949e8381be5a006d4d7b87e99","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 17:18:39.000000000","message":"Patch Set 1: Code-Review+2\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review+2 has been copied to patch set 2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").\n\nLooks good to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d16a2436a49731a137deee142fd8960549204ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 19:48:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4893895117b849b7b6963423adb22993\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c406db0a1431442994de2892f9b14181 : SUCCESS in 1h 50m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/591cd7630cf94bb58a0872d0e5499e1b : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836febd899264a17a71d3a6e5a39acc4 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98cc571e0a904cbc9f0f049dac7f73d1 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50c730e70a374fb199d5d24418aa103e : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93317b6133e74fa89ed9d1823f629b18 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/02edb949826f494394d2087a3c31e2bb : FAILURE in 4m 53s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c39defa9e77c49b9966ff373f26f907d : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aa88b378168e45b3b8410b32a9ffe832 : SUCCESS in 2h 25m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9d2c67a10b984a44aced1bfb91bcae09 : FAILURE in 30m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f9db4b21449a442dbd67a669895f4605 : FAILURE in 1h 14m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b608e06de2bb41e7b2e264ac242cf610 : SUCCESS in 1h 01m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1972ffb97c64e23982418e0a6e0bcde : SUCCESS in 1h 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c374310e64d6a0db780d63c7d267229248b4fce","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-30 07:07:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93a15a969cbdd0776e82c30b37d7533210c800f0","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-01-30 07:09:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a210f851efd84222405c3ea873c852cbc649379","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-01 15:00:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c8bccbd89767d7189b96079684533096a2196e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 13:51:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"074af716954d12d09cfdb8b1729f00661f6451e7","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 13:59:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2bb3b1bceb6892b9563dc7fe51c3aa4322b3b48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 15:34:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77fd728fa9446d481e2dd2d7dd5e139\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cde37b0d6ab64e2faed4077cc6d648ec : SUCCESS in 1h 42m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c1eeaad569e47d0817aaad4473668ab : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa82442a6624a3a8b57f89370f031f9 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5dbbff176024875ad8ee83475c619e7 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c51359822ab457181d9b197954541da : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210cf18462e54cfdb97ff78298e1406f : SUCCESS in 2m 58s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ce6da9babf37497594967110b4b5b3b5 : SUCCESS in 5m 47s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/30052ee7e6054f298b9f000daa78cfe4 : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/114e69abea4e47da96346b7bde8a492b : SUCCESS in 1h 40m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1b23643580404bb5bdfa595912c5e036 : FAILURE in 28m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c3b37c54215949e891b50cc405e89bee : FAILURE in 1h 20m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/66f9b7d3b8cf4204b467a3b7096b9cb0 : SUCCESS in 1h 00m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/be1fbf4c8df548869285f0087c746080 : SUCCESS in 1h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"07bbe78c877e4494b622432269b1212537d2f090","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-19 09:02:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1a4da6241b8ddb216597336fa7b2d6b80bbb33bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 09:53:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6607aa614ca3b60c58775a7b3375f09ebe598a1b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-28 11:53:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6c468b4f6c75cb66be0b23a7f4d42f01b2fa8b63","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-03-20 15:57:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85a76997d79bf8a7a7270bfd836adcce674b20f9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-03 11:57:32.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"80de9e72d11a378ae3f96dce59ddfa171423b47f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 11:58:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60829cd21fd0b49a895cc7148b715b97526d7142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 13:50:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b87b9885f2e441dabe4fb1e8f4563e61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c822fef16b8d4e2f8c07e10e8c3c2240 : SUCCESS in 1h 43m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a3c111827f427d92e1e2a46079144c : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df3655f63f5a4daa95474f040c0d3be7 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43ba869acd884d06ab46d2dafe898e4e : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2991de61524039bdf7c3707a91332d : SUCCESS in 3m 32s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/09a3a84763784d2380f723e138f62d57 : SUCCESS in 4m 52s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3bfcba3c0a8c4e4b987ec30898bb5b12 : FAILURE in 1h 23m 11s","accounts_in_message":[],"_revision_number":3},{"id":"9a2e22f8447a73db68d9f8df2540e90c6151deaa","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-03 15:17:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"001b61aa3025c1b90d6b614ab3bfdb6170b5f260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 20:31:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aedda359cf74f6bbc70fa72ce73b5d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e06a800deda4269a6b4a2dacef01f01 : SUCCESS in 1h 37m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e453f59ae0684805b7a8d4ef39d57425 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5f8acd6ee04f119b0140ac868b1202 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36fdd736fb79433c97c646964f58475a : SUCCESS in 11m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e48402969c82419d8fb10103d6e1b30c : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f274e80d1c346d6a94bc88f660af9ea : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3a6b87c02740e290936998f1ec4ebc : SUCCESS in 3m 42s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c217e9e174814c54b1679bbc2234eab5 : SUCCESS in 4m 59s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/2479dcd37cb74eaf9c387944f5f53298 : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e5eeb748b2e84754be97b402c4795d53 : SUCCESS in 1h 24m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2a82be39258649eab358e3939041fdcb : FAILURE in 17m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4f90bbe6452640d5b3db1dc942503fa7 : FAILURE in 1h 45m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5361b6b2be15488db0b5daae16cab9b7 : SUCCESS in 1h 01m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b09e4506aed44a1ae25502b2782775b : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3379892390f757ced0033041a6f094ad027134bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 20:31:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b433735e0d28c4d42ae743772426b2f9d5008d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 22:24:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4fb7189990145739d15eb71f0dc81c6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/541ca3c645bb45dd8d2df72cefcfd722 : FAILURE in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597722be69a248ef8b631db6e9ef6aff : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94610a895d1f4d3293191c28c6008366 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3980b98baaf44d2b00076fad9a41692 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c830988022447c7b2ef34326c0a7ae8 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/7f0a1a5b18a94b8cac46ec1e2d240dd5 : SUCCESS in 4m 39s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/64d84f870af1437aa352ffbadb80eaf3 : SUCCESS in 1h 44m 48s","accounts_in_message":[],"_revision_number":3},{"id":"b93cdcaa8a881ebe6f588fbc1862c05035ef6423","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-05 12:40:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a30828b6b42e78fed0cf882cf12df5b91a201322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 15:19:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b3dc3fc59d7425bb4e53d21f6bcfdf5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/896c40da19014c85a8d4119770d8578c : SUCCESS in 1h 22m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42a94b6ab46445cfa9d119b2a06fb3d1 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/249c8d4d04ea4d568307598ff774a5bf : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c164eb28234811944073caad24583d : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffe596ed10964d41aae09df7ba9cf20c : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf7a320fb1364d82bae55629739b29f9 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e281777dfaff436887101a9039ae9f8b : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/831bcc7bb1c44d08ba04ef702e1c038a : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/cd1d2da4e1ff412687aee528970b3445 : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c11365e35d0648189620938d762800cc : FAILURE in 2h 10m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/410a96fed12048328bc15d36d7b22a53 : FAILURE in 16m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/63c64a4a43324396ab1a96c65a3f7d02 : SUCCESS in 1h 42m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a76d802465754177bc89961eb7044811 : SUCCESS in 1h 08m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd99bd23032941f58383aad0dbc76e3d : SUCCESS in 59m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79dc27256cad1acab1eac0a6ecd88c262cc02530","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-04-09 09:57:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e92e15dce5d131d0d53cc3f4d10b87ea6576f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 11:52:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d222d899c5477db32bfb81c68fb35b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bca39441b934563a7fa60b04eb9f822 : SUCCESS in 1h 12m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec2f8c4b38b949fe874335a904bdb655 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2805c2e921ad4749b905f4e917d66e3b : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99e3c078ee764ebfade98bce4f24014a : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce57166323c94d59b6f0b961856e8b2b : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc3b6aeebad04be2b5001406088108ae : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d909fdcacb4516814c412cddfbcad4 : SUCCESS in 3m 11s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/64dcddf998b14124854eedf0cd76985f : SUCCESS in 4m 54s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/38452feb7e0c43d8a5bfb2b4b7831257 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/38f87c33fd6242c898f6326052bf421e : SUCCESS in 1h 48m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d2081f986d284f4a849373c34d02e9a5 : FAILURE in 26m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7283984b3f504523ac8eeac081b53dc7 : SUCCESS in 1h 24m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c64d0b746ce24a0aa1f3d0fa45c38637 : SUCCESS in 1h 02m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2637aa2b81ee4b3f8cc40a42ab145e65 : FAILURE in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e784134080e01a4ff29a05dd3c0ea48725f048a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 11:52:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b62d033792294f952964b2e14c47138152711570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 13:47:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0834e0c172c14c8c9124c7b93e39aff0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b17f0e75c7a4125a3a383e00b214de6 : SUCCESS in 1h 54m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6944917491e045cb930106e869518f70 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6f609c91d1f4ff4abbe2fe7f937fce4 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b1776527d7941329fdddf34704f9e26 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a32ac9f72a4dcdaf48bea6d920f2d2 : SUCCESS in 2m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ed47cc8e866840a99b7e8d89ec743ca5 : SUCCESS in 4m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/08d8a43bbe5e481986f83ee296d69f5a : SUCCESS in 1h 38m 50s","accounts_in_message":[],"_revision_number":3},{"id":"15df2935074166aeb2dd7bb4cfb76e12cc9e9c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 13:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d3a48e0762dae737c05f0f0abf5c1833092801f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 13:48:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678c2cbbce044e3fb3ce9d6b52fa59ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2812d4e56e48ec94248b59874b6733 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b62208a54c4cfabf641bd9e6e5b91e3e24d54455","revisions":{"acf2bb45f1daaece882fcdea5fd3b5a2a454bfd5":{"kind":"REWORK","_number":1,"created":"2023-09-12 11:46:25.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/31/894731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/31/894731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/1"}}},"commit":{"parents":[{"commit":"7705b4f3023e0e63f3b37e9a25c774f309fec55e","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7705b4f3023e0e63f3b37e9a25c774f309fec55e"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 09:18:58.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 11:45:53.000000000","tz":120},"subject":"Use StopWatch timer when waiting for message","message":"Use StopWatch timer when waiting for message\n\nWhen waiting for a message in a queue, the queue.get(block\u003dTrue) prevent\nthe heartbeats to be sent at correct interval.\n\nSo instead of blocking the thread, doing a loop using a StopWatch timer\nuntil the timeout is reached.\n\nCloses-Bug: #2035113\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ie5cf5d2bd281508bcd2db1409f18ad96b0822639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/acf2bb45f1daaece882fcdea5fd3b5a2a454bfd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/acf2bb45f1daaece882fcdea5fd3b5a2a454bfd5"}]},"branch":"refs/heads/master"},"e11fa7d80e1fe4c0565ff065d14808c3b8bb53ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-29 17:16:28.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/31/894731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/31/894731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/2"}}},"commit":{"parents":[{"commit":"b5244bd05a7e712b0f3f6513f061de25c0048d30","subject":"Merge \"Bump hacking (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b5244bd05a7e712b0f3f6513f061de25c0048d30"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 09:18:58.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-01-29 17:16:12.000000000","tz":60},"subject":"Use StopWatch timer when waiting for message","message":"Use StopWatch timer when waiting for message\n\nWhen waiting for a message in a queue, the queue.get(block\u003dTrue) prevent\nthe heartbeats to be sent at correct interval.\n\nSo instead of blocking the thread, doing a loop using a StopWatch timer\nuntil the timeout is reached.\n\nCloses-Bug: #2035113\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ie5cf5d2bd281508bcd2db1409f18ad96b0822639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e11fa7d80e1fe4c0565ff065d14808c3b8bb53ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e11fa7d80e1fe4c0565ff065d14808c3b8bb53ea"}]},"branch":"refs/heads/master"},"b62208a54c4cfabf641bd9e6e5b91e3e24d54455":{"kind":"REWORK","_number":3,"created":"2024-02-16 13:51:01.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/31/894731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/31/894731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/31/894731/3"}}},"commit":{"parents":[{"commit":"b5244bd05a7e712b0f3f6513f061de25c0048d30","subject":"Merge \"Bump hacking (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b5244bd05a7e712b0f3f6513f061de25c0048d30"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-09-12 09:18:58.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-02-16 13:50:49.000000000","tz":60},"subject":"Use StopWatch timer when waiting for message","message":"Use StopWatch timer when waiting for message\n\nWhen waiting for a message in a queue, the queue.get(block\u003dTrue) prevent\nthe heartbeats to be sent at correct interval.\n\nSo instead of blocking the thread, doing a loop using a StopWatch timer\nuntil the timeout is reached.\n\nCloses-Bug: #2035113\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ie5cf5d2bd281508bcd2db1409f18ad96b0822639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b62208a54c4cfabf641bd9e6e5b91e3e24d54455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b62208a54c4cfabf641bd9e6e5b91e3e24d54455"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
