)]}'
{"id":"openstack%2Foslo.messaging~896422","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.2~I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","project":"openstack/oslo.messaging","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-26 09:37:19.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-09-26 09:37:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","subject":"Add is_admin to safe fields list for notifications","status":"MERGED","created":"2023-09-25 17:39:51.000000000","updated":"2023-09-26 09:38:14.000000000","submitted":"2023-09-26 09:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896422","cherry_pick_of_change":896451,"cherry_pick_of_patch_set":3,"meta_rev_id":"ba339cf22c8410ad14f34d6b494ade70a1fed8c5","_number":896422,"virtual_id_number":896422,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 09:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2023-09-26 08:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2023-09-26 08:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 17:51:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 08:38:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"79180f2d58ebb87902887aae4ef89e27436fc15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:39:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9989e023939e5734c7d923a42c19b9a3f1eb9626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 17:51:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c894476420e4f899b02879bce2c00e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/596461ef1a0c4f329eb655aae9ba5a79 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf7fc2c4955485e999d1565429f4a1a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6820bd889e9d43d28b1969153399b981 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5f5e8322b7f460b9c74d723a37b9091 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5013199dad314430abea4140b94c45e1 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4f56e6121f4dc7b27c7a764f69b9e9 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/6244ff313074445cb4680a068f1692a3 : SUCCESS in 4m 59s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/40983669b4c045c78ae3a7302d0a3934 : FAILURE in 6m 14s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/4701554a28244595a3e8aba5b3c701ec : FAILURE in 4m 09s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/433386844f4f432694636e6522b73f18 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b91e8e642195a9ddfc3b6754b3be052be726c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:51:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fabaf39902420769559ba3dc7aafa5a3d54240f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 18:06:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/315fdd8185cc48c9b7a2ad0dbfe60f0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d954a8ce89b44a3a5334e07e6758eaf : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08198da42454ebdbbb22c4264bbfd66 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f235ebb7dc4d6bb17bd6511792f46c : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfaf20203429411a8e7784b93fa524e5 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/686f59fb98914949aee11ffaa4369536 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9e2d32a9794be1b852d8fdcf971910 : SUCCESS in 3m 39s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/af0f7ae0c84b46cea61f6e04cb2f5eb7 : SUCCESS in 4m 53s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f79d6b80bda847328d62e82484dd6fde : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7e2017519c244460ae96f0b4caf12236 : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/c23055e46f624019be57326233580377 : FAILURE in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6115c4cdd35a0347714aae2bffa6ae7cc6d263d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-26 08:08:23.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf72c3fe2c6592d53500a11a442446e2b843dcfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:23:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e414cfc4270400a9a44846c103abc16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d88102dde242369bbc664a4afc5f6c : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f368d6851b31499ca1d980efcd906020 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff6c31d073e44e8ad6f6b2ea0af17dd : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/450deb3a07af42ac98555234d0269fc5 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abe64f7ca524480cab35d79cebee2869 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bde1f15d874898b8f291cf79550cd6 : SUCCESS in 3m 27s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/6e887069883a46839fcfc3dc17f6e529 : SUCCESS in 4m 46s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8462efd57ec84bc483b87f1c8aa875bd : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/d2db91faceae41019229144263875f0b : FAILURE in 3m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/5b70fb59ca2b4f74bcc3526d11a862d1 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"803c060a8f90e0a1de2d7cae1321dfb7abf3cc34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-26 08:38:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a6826d17f808fecaae5243f654795d1b57f8e97e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:38:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01366e82aa22bf3dda1c3e3f3233f5d458f844cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 09:37:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6930e557350641b49e0454508168adba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbbaeb8ea734ca089061b2adb456f4d : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7589ff04703e4109b6e4c4500bc91684 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/600b1e04838946a49722a96f6e6f7615 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709fde6014c640f08014ac72c44cd951 : SUCCESS in 3m 25s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cc61a54166004072adcc4df34ed29f76 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b18e41f752dd8def60e3cc8ba557eccd642d6b44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 09:37:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ba339cf22c8410ad14f34d6b494ade70a1fed8c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 09:38:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30e7fe4234074101a1a2111b0e6a6dc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1fb94adecb41f7a80d1e146f72d8c7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29623702fc1ebce240b6d368ee95ab9f479d5e02","revisions":{"03671dad33472e9edf08efe698fea000ae204ca0":{"kind":"REWORK","_number":1,"created":"2023-09-25 17:39:51.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/22/896422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/22/896422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/1"}}},"commit":{"parents":[{"commit":"2a76281e633878ff767eb2906b1116275633cdca","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a76281e633878ff767eb2906b1116275633cdca"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:39:51.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03671dad33472e9edf08efe698fea000ae204ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03671dad33472e9edf08efe698fea000ae204ca0"}]},"branch":"refs/heads/stable/2023.2"},"1ca2fe62a9ee1ea5644eaa6b34dab597ff8dec52":{"kind":"REWORK","_number":2,"created":"2023-09-25 17:51:58.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/22/896422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/22/896422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/2"}}},"commit":{"parents":[{"commit":"2a76281e633878ff767eb2906b1116275633cdca","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a76281e633878ff767eb2906b1116275633cdca"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:51:58.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ca2fe62a9ee1ea5644eaa6b34dab597ff8dec52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ca2fe62a9ee1ea5644eaa6b34dab597ff8dec52"}]},"branch":"refs/heads/stable/2023.2"},"29623702fc1ebce240b6d368ee95ab9f479d5e02":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-26 08:08:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/22/896422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/22/896422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/22/896422/3"}}},"commit":{"parents":[{"commit":"2a76281e633878ff767eb2906b1116275633cdca","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a76281e633878ff767eb2906b1116275633cdca"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-09-26 08:08:23.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n(cherry picked from commit c1b606f77e69a2829fcfdb74b8b3f646d4b60929)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/29623702fc1ebce240b6d368ee95ab9f479d5e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/29623702fc1ebce240b6d368ee95ab9f479d5e02"}]},"branch":"refs/heads/stable/2023.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
