)]}'
{"id":"openstack%2Foslo.messaging~896423","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.1~I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","project":"openstack/oslo.messaging","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-10-17 04:31:22.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-10-17 04:31:22.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-17 02:56:49.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-10-17 04:31:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","subject":"Add is_admin to safe fields list for notifications","status":"MERGED","created":"2023-09-25 17:40:03.000000000","updated":"2023-10-17 04:32:17.000000000","submitted":"2023-10-17 04:31:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896423","cherry_pick_of_change":896422,"cherry_pick_of_patch_set":3,"meta_rev_id":"c728ca25a1ae4bd12cdb25e84956fab6d9b43e80","_number":896423,"virtual_id_number":896423,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-17 04:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-10-04 09:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-10-17 02:56:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 19:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 09:04:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-09-27 09:10:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-10-04 09:53:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-10-11 10:43:18.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"ab01b2a85623dd9dba43a80f75f7fc8589968b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:40:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"f157b93f1b943e5dcdf68d2380ac78235d00ccea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:52:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ee91b8f71df0c7ec0e27055c11e7ad0905c99a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:31:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafb80da754646b3877e39b3e4e04a7e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e60a2da029df4300870ed2bd2dec379b : SUCCESS in 1h 22m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c111394a893a416797bb110c932fdae4 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555b1dee507d48249dea610bcbf6d424 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90661c06a1cd44e0930b456113b947cc : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92f755ee45e446a0b40d988d65e256ca : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b01a2f376d477bb47139146fefa75d : SUCCESS in 3m 27s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f628942614e243cc98684da7bd582c03 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e7641dfb25b54df78e86092002af2fc9 : FAILURE in 6m 09s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/3f286a6d8fb0488097ec9542eeb49afc : FAILURE in 4m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/30eb4f35fbcc40f983811a158d6dff66 : FAILURE in 3m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3e84815c646b47d18f794b2943a543ed : SUCCESS in 1h 32m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3af71434e1a348dfaeaf97a2d4fcc9f7 : FAILURE in 27m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/56e5904cc0fa4042b5fba51c2522431f : SUCCESS in 1h 13m 13s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/22a0b321e40d4e96bc4a9e33b3d3abd6 : SUCCESS in 1h 03m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/75713e6dc4194807a7541a1da1a1968b : SUCCESS in 50m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0de4a9c9c18c50e81f897296303dee988b699c81","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 09:04:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5580c168accf71983bd1c0dc85f335a51ad260b7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-27 09:10:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8fbac05932bf73bb60179ad3cc3cc66482e7cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-30 16:25:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"efa8d1a191155d801c3682017d07aef8fffb6583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 18:14:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67d8b71e6d844cdcaedee2dde13f2ba4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d518627400d540a5a0f0df67d3c38c39 : SUCCESS in 1h 07m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8f9f05cc264416099a3d631e609c6c3 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6870c2be90cb4b0facbf5509f07727bb : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad1e958bf3ca47a98b66d84c73e04aa5 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cd91bcfa00443e89387094909dc2ad6 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ace04a670a46239f9c4babbd6c3607 : SUCCESS in 3m 29s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/56676d9bc5284e43932c556fec8bdaf1 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/53807d99aa714cbdbb9dbdbee90db05e : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/22a0a72892424e1faaaeff9c2e8fe07c : FAILURE in 4m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/27920df3172749668213ee6d5f0a8285 : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4a576e22ecf247e5b425422c78f688a1 : SUCCESS in 1h 01m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f451b328f2a5475d8abdff2f724103ae : FAILURE in 29m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9ba5cc98767140ffb6d713f9b2a08490 : SUCCESS in 55m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/792b8246c92f46e0ae6746b2cf30631f : SUCCESS in 1h 00m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/46e076083f7a45ac8efd446564606c3a : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4add4f011eca7be1bb89518a8b431973dc1da456","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-04 09:53:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"53a13b014163558cc3a7f2acc164de01523dc55c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2023-10-11 10:43:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9a1f5c07ee904a31de6f6959f6ac187c47fe4eb3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 02:56:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"73babcb646a159d2059a884b24d9bb2144d03b37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 02:57:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a9c374c44423a1e6b7f8bf940c5b344916f9feec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 04:31:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1d6f2d495a4a7698eaed770bc4a9bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/491eda9daad149dd8e8322ad6b3e1a42 : SUCCESS in 1h 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce5876125914b26ae15bb3a0ef6331e : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5639a8bb6ddd41629b5b79b174651b2b : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/900830a557e74977a2df1de2cd134b0f : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b0bca325444ce8b332fe319716912f : SUCCESS in 3m 45s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/515e5933b6434a57840d6a239493f0f2 : SUCCESS in 5m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/86990cd65b094ec1816ebd1310e7180b : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0dc3aa1616807c4533af7e1b7e3ce482e8ee128d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 04:31:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c728ca25a1ae4bd12cdb25e84956fab6d9b43e80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 04:32:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/674a6e0632e04cbeb42fbeaa958ff941\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0730396ea8c34de4ac004e568266d035 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45a01414689e917f6adb17f99deb71830837fa97","revisions":{"68b64bcb6f53ceca5660182b305c805468bdde52":{"kind":"REWORK","_number":1,"created":"2023-09-25 17:40:03.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/23/896423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/23/896423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/1"}}},"commit":{"parents":[{"commit":"b8b4375885b5de03e3ae4e3344ed1305a3e82291","subject":"Merge \"Disable greenthreads for RabbitDriver \"listen\" connections\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b8b4375885b5de03e3ae4e3344ed1305a3e82291"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:40:03.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/68b64bcb6f53ceca5660182b305c805468bdde52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/68b64bcb6f53ceca5660182b305c805468bdde52"}]},"branch":"refs/heads/stable/2023.1"},"9137ac7b35a5d437d0959b50bc1cee2c5edee78c":{"kind":"REWORK","_number":2,"created":"2023-09-25 17:52:40.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/23/896423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/23/896423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/2"}}},"commit":{"parents":[{"commit":"b8b4375885b5de03e3ae4e3344ed1305a3e82291","subject":"Merge \"Disable greenthreads for RabbitDriver \"listen\" connections\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b8b4375885b5de03e3ae4e3344ed1305a3e82291"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:52:40.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9137ac7b35a5d437d0959b50bc1cee2c5edee78c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9137ac7b35a5d437d0959b50bc1cee2c5edee78c"}]},"branch":"refs/heads/stable/2023.1"},"45a01414689e917f6adb17f99deb71830837fa97":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-30 16:25:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/896423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/23/896423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/23/896423/3"}}},"commit":{"parents":[{"commit":"b8b4375885b5de03e3ae4e3344ed1305a3e82291","subject":"Merge \"Disable greenthreads for RabbitDriver \"listen\" connections\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b8b4375885b5de03e3ae4e3344ed1305a3e82291"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-30 16:25:41.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n(cherry picked from commit c1b606f77e69a2829fcfdb74b8b3f646d4b60929)\n(cherry picked from commit 29623702fc1ebce240b6d368ee95ab9f479d5e02)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/45a01414689e917f6adb17f99deb71830837fa97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/45a01414689e917f6adb17f99deb71830837fa97"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
