)]}'
{"id":"openstack%2Foslo.messaging~896426","triplet_id":"openstack%2Foslo.messaging~stable%2Fxena~I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","project":"openstack/oslo.messaging","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-09 09:10:15.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-11-09 09:10:15.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-11-09 09:10:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78b08edfcb8115cddd7de9c6c788c0a57c8218a8","subject":"Add is_admin to safe fields list for notifications","status":"MERGED","created":"2023-09-25 17:40:30.000000000","updated":"2023-11-09 09:11:10.000000000","submitted":"2023-11-09 09:10:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896426","cherry_pick_of_change":896425,"cherry_pick_of_patch_set":3,"meta_rev_id":"624a779e51cfacb71fd8814f75c3a528bbb4ac8d","_number":896426,"virtual_id_number":896426,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 09:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-10-20 10:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-11-09 07:51:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 19:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 09:05:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-09-27 09:11:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-10-20 10:17:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"66c272e1df23e5a5f6a0a331fe88ed99efd693b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:40:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"3eab687efad621f6a859cf98c18ef170ccfcbf58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-25 17:53:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ebdb92e8a079ab783028e30f0a8d8f8ada5f37c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1499579113854fdda51689a695a1d178\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99c8bbc47cb143808786b2cbd3261217 : SUCCESS in 1h 28m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e91b30ae384f4d7d84b6ce993b25bb61 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2787071f34d14914b3efd37ecf3beead : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fa488d9e66a4b12b57511783a62b2cc : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea157bd525a4fe5a1ff68b01a089ea1 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90c5688ca5a44eebb17739e13eab72ef : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3133c8c468044f3cb1b8d3d676bbda16 : SUCCESS in 3m 59s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/6606d854605541c7a96ff2e4d4b9f642 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/a439570565344badb0c6c3cdf7b02035 : FAILURE in 7m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/445a5d4550474d84a7e27e6242b7fa5d : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1953ef2d67494908a743e8dc71584e65 : FAILURE in 4m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e0e88e20532a495c83e908c9d3435eb5 : SUCCESS in 1h 21m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ed776e121ac941cfb86b08c1eb2e6d42 : SUCCESS in 58m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1796b7d5da3f4210ba151c139285145b : SUCCESS in 1h 18m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b568c223fa744c66a4260f100e76031c : SUCCESS in 1h 03m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d12b9861eb74331bd95f49a5a4e8c78 : SUCCESS in 1h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c67daba1f305fb5803d7aa8e11f44e724bf66769","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-27 09:05:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a1d00f333455dbb6e1b1e18c751efbbe8643349b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-27 09:11:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d4032851395ab07970925c2f688d23b6926b651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-30 16:26:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"99d5107942e02d238806c5ff792239c67d79a7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 17:53:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b659cb57bdb457d9ce908e0c5617e90\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba661f3e91b048d588c9b8142fa6b31c : SUCCESS in 51m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df3efd48a6b48a28220b69d46ffe598 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c452aac4de95402cab445bd3497ab6a0 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e30951b031d0457e9db7a46a7281f867 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/088e6bcb5ca7412a896e7e63d62c672d : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/522fad5d3ac94b94907e9c9659477be0 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f93b7f1734644909926cafc3a4175468 : SUCCESS in 4m 14s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/6704f3d991d44a70b55f33a1e3adcf85 : SUCCESS in 4m 41s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/1f59323a05864661a616b37aa435d8b2 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/cf29cadc94b84b5686cc921b2fb118d2 : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/802d470cfe384346a82e38ee42df66ef : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7de27057b9204818a042fcf2a7110b61 : SUCCESS in 1h 19m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2a5485cb61cb4c499c0cc8f7053e9f4a : SUCCESS in 51m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c4b636e7c2b9484a82366addd0979f88 : SUCCESS in 53m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2f73dd3fb98a44768842703d8477c81f : SUCCESS in 42m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bdbef38a5c2436298c096b23cb1fb99 : SUCCESS in 1h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c94433e18517ed88a05c96de5748d20bbd30040","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-20 10:17:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"36d6414a08e196b89daecd50fa3638198f783905","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-09 07:51:21.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"93a3b357e92216994f4997e8a76dae18ed67cec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 07:51:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1caca6bf23b73c07621abb48f403704e77073bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 09:10:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b570a1439b9b4e4fb53180cf571bf29c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13299f84df6f4ddaa5686c36613b3173 : SUCCESS in 58m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca54db40ca3141b391f9ec133166a51b : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/540ef82263744bd692966379d65f1d63 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/244ca67d42ba4c8db6da71307495544b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f7cb184da7479dacf919262e849e62 : SUCCESS in 3m 03s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/aa9e7b18e85648188d67ad770adf2fc5 : SUCCESS in 4m 40s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b939b55a84ac4790a0471a43c34dece0 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b8350727a7668f5101f52d4dcb69298c14c0ca20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 09:10:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"624a779e51cfacb71fd8814f75c3a528bbb4ac8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 09:11:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a574ff99344252b4fd995c57091fce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6a56bd10e84d8f92f35fa31a0737c5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1c52bab523e151f9e94ee6bed27c964d7f57a18","revisions":{"ccf9badbd96cc217ad2c005e9a0bb3982759d57c":{"kind":"REWORK","_number":1,"created":"2023-09-25 17:40:30.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/26/896426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/26/896426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/1"}}},"commit":{"parents":[{"commit":"44d112eb9d822e7432995bacdc494a0685798dd5","subject":"Only allow safe context fields in notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44d112eb9d822e7432995bacdc494a0685798dd5"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:40:30.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ccf9badbd96cc217ad2c005e9a0bb3982759d57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ccf9badbd96cc217ad2c005e9a0bb3982759d57c"}]},"branch":"refs/heads/stable/xena"},"7312cb566b2e3569ad8d7765f5c5fabaa2332c83":{"kind":"REWORK","_number":2,"created":"2023-09-25 17:53:01.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/26/896426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/26/896426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/2"}}},"commit":{"parents":[{"commit":"44d112eb9d822e7432995bacdc494a0685798dd5","subject":"Only allow safe context fields in notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44d112eb9d822e7432995bacdc494a0685798dd5"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:53:01.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7312cb566b2e3569ad8d7765f5c5fabaa2332c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7312cb566b2e3569ad8d7765f5c5fabaa2332c83"}]},"branch":"refs/heads/stable/xena"},"d1c52bab523e151f9e94ee6bed27c964d7f57a18":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-30 16:26:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/896426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/26/896426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/26/896426/3"}}},"commit":{"parents":[{"commit":"44d112eb9d822e7432995bacdc494a0685798dd5","subject":"Only allow safe context fields in notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44d112eb9d822e7432995bacdc494a0685798dd5"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-09-25 17:28:48.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-30 16:26:37.000000000","tz":0},"subject":"Add is_admin to safe fields list for notifications","message":"Add is_admin to safe fields list for notifications\n\nWe encountered bug 2037312 in unit tests when attempting to get this\nchange rolled out. Heat apparently will attempt to set is_admin using\npolicy logic if it\u0027s not passed in for a new context; this breaks as the\ncontext we are requested doesn\u0027t have all the needed information to\nexercise the policy logic.\n\nis_admin is just a bool; it\u0027s not sensitive; easiest route forward is to\nadd it to the safe list\n\nCloses-bug: 2037312\nChange-Id: I78b08edfcb8115cddd7de9c6c788c0a57c8218a8\n(cherry picked from commit c1b606f77e69a2829fcfdb74b8b3f646d4b60929)\n(cherry picked from commit 29623702fc1ebce240b6d368ee95ab9f479d5e02)\n(cherry picked from commit ee0bf90b6862c3e9469068d6dd65b0cff398ba75)\n(cherry picked from commit 45a01414689e917f6adb17f99deb71830837fa97)\n(cherry picked from commit 047da4e5ed8f853c1e69f73acaa9e655c88ab6e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d1c52bab523e151f9e94ee6bed27c964d7f57a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d1c52bab523e151f9e94ee6bed27c964d7f57a18"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
