)]}'
{"id":"openstack%2Foslo.messaging~896618","triplet_id":"openstack%2Foslo.messaging~master~I8eeb1b345033269a98989ef4a563ca65fe04e57f","project":"openstack/oslo.messaging","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-09-29 10:32:57.000000000","reason":"Change was submitted"},"35978":{"account":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"last_update":"2023-09-29 10:32:57.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-09-29 08:41:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-09-28 17:49:13.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I8eeb1b345033269a98989ef4a563ca65fe04e57f","subject":"test py311 job","status":"MERGED","created":"2023-09-27 08:34:27.000000000","updated":"2023-09-29 10:33:59.000000000","submitted":"2023-09-29 10:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896618-py311","meta_rev_id":"8bfd3ba592e55986453d5355187a855423289f85","_number":896618,"virtual_id_number":896618,"owner":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 10:32:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-09-28 17:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-09-29 08:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-09-28 17:49:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2023-09-29 08:41:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 10:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 08:45:03.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-09-28 08:45:03.000000000","updated_by":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-09-28 17:49:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-28 17:58:59.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"e52fcd3c2de6ac7c02bb19e61c62de14186b17ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-27 08:34:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed0546f1ad9d339efa78f38f4f4b8884aef18978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-27 08:43:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06659f6eca0a0d0ee4acbb2b6f9a3ffdd6ea5e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 10:36:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e189dad0329b43448e68849deb5bca1e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96f02e680c6048de9a04fec873e691e5 : SUCCESS in 1h 15m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a259c25832d42c58f2b11a7743dc933 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3661c2742154177898a64b07008cd51 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb87cf069440455fae917a63c81da74c : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf6e64a43f34b7c85808100493052c5 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acd7e910c0bb4061be597b3268f99160 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bb9ceee08ad48dc8510b29cb7b27c12 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d82b0a3549464172a4dcd68434737d73 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d05d009ba3064b5fbd4d86f133a1bec0 : SUCCESS in 3m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4a9ce0fee97047258f9383acdf0968ed : SUCCESS in 5m 48s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/23b48eaa1b204b1bb5e8ab969fe28e23 : FAILURE in 5m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c082531f627646b09fab7bed4e82fe41 : FAILURE in 5m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/20ccd45041b540b5b410144bb9e82f2e : FAILURE in 3m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/414ab50af76c4e7b807b890b1426a73b : SUCCESS in 1h 37m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/04b1677b6a344f35b0b78d6b83059662 : FAILURE in 17m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/baa964641d284fbf9cc4dc1213b5251c : SUCCESS in 1h 45m 20s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0fcbf234999e4cc2b5fd82351afe058b : SUCCESS in 58m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b944ad31d6940e9bfd1b43b8972f6d7 : SUCCESS in 1h 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"072f4f7e57c5ac22d02590601a5c325316afaf46","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-27 11:46:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"623054182f414b320a97043bc07f15a43fb82b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 13:25:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c58799419d4e9ca1bf62d7e6a3f0b9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/981e3f0b02744dcb8c7aaba0b641eddd : SUCCESS in 1h 24m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddf5f9112b7a416f907b7e0976bd5400 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5944d376e9d54b39abf067b83109b0c8 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c837442ee5546f691f62a6f33099e20 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/072c39144a8c4405b8a29f7f60af9d3b : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b45d396914244925a176bf4be20d0de4 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/510670cb56fc486da3466c6a581a9215 : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482e2c3980794483a585b2f835a2fc71 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48945ce019a24e56980f9056a590c803 : SUCCESS in 3m 20s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bf15e25b5aeb4ff28b97e2b2760db7d9 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/5a6bbaea28244be293658ede9de04dcc : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/25dd6639c0a1410bade92454a2bf68f2 : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d39a2be826d042bf9e3aafb614da953f : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9b6a7eda15804351a560417cbd9b26b9 : SUCCESS in 1h 24m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/fd35d13b5b2e41c0a93baff751ed51c5 : FAILURE in 30m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/80d69a1c10374f0fb76f5aec0be17686 : SUCCESS in 1h 38m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/66e375c4162a4a519cb1dd4849d5f546 : SUCCESS in 58m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/09e71c5206f249ccac5cc7a478546fde : FAILURE in 1h 01m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"027d729ccf265189508e0d1172fc7bb346bb43f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 04:38:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"91e5f74c2cd3af7076d3a733d63a8471ce2c8798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 05:35:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"608656aa19d8e30ef347633892a6943e70038b9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 06:26:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a02cf2feb4f3d6db24c8898433b9b23b4b4d918f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 08:17:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/149319b85ee246bfbe7536be680ce092\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2821bc653904579a6aa98f6b697d0b9 : SUCCESS in 1h 13m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f59f3b71efa48faa35a49821f991e28 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56e63d237e141ad8705c2a562645193 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4755f9126bd84e08b33f6e7cfa3784c0 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e87e1c8875f8429bb5bcddcf2899f8d8 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06bf51a9e06d4a8cba63c6a1ea2116ff : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa6f9d9b47e4428f815fae6a2841dcee : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6317f3174794a51a5dfb3d7a1ae8c6f : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74e0c73cc104489d877eef7cc60d489e : SUCCESS in 3m 02s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/94acf01275a5424d9650e6fac5743066 : SUCCESS in 3m 55s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/1c884a11a045450fb994c4ef61e67737 : FAILURE in 5m 02s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/694d20d65b1049438673c19fb0b81cc6 : FAILURE in 3m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/8fc1b0cf896f49dcb24cbd81d85d7a09 : FAILURE in 3m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5b817cd818d34c8a8612959df4455a98 : SUCCESS in 1h 44m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4dc0271d212446ab845bc6000d235073 : FAILURE in 14m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2892b7f4629c491c81710ffc7f11e037 : SUCCESS in 1h 33m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8914e340df0a47cda6d58f342f1377c8 : SUCCESS in 38m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff75813a4b0a4857ba3c9a6ef0b98dac : SUCCESS in 1h 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0985352d5a9969080527e27f9a76bc494d6ca39d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 10:36:48.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fcde46806914920b2bdc59c585c3bbc48b216b81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 11:04:30.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"f40d7da9ffa2d56bb1e394a8102eb64bd0723120","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"date":"2023-09-28 11:05:37.000000000","message":"Topic set to py311","accounts_in_message":[],"_revision_number":7},{"id":"bd6d817e65701eba50c794dbc8bb0d3378ff62f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:42:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed6ee01cc67434381cbda2d39dfb096\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a6e4308dfbf489896ea97dcc2d4eef1 : SUCCESS in 1h 36m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/052607b0d5b243dcb398969b9a5cafeb : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd8984b262246d9a0f87d0a149e8213 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c310068dd454a529b509fa59505e636 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/451a86814bef48f897b97927743c0efc : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c2104fdf9114cf09e7753479a7969d9 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3af9665733ab438c83a3da194ea950e7 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb45a4213f24f229bae977c324c0546 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01448dbaf04d48ca97992f3ded203f90 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/11d3422acdec48e799f3dbe4fcddd70b : SUCCESS in 5m 07s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f2bb9b436f1a425096be229ab4cdf19e : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/d3faa315bf024e8ab3d979cd7e3afc81 : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/e8cb9f0dfa914dcda60e1237a16a9943 : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/32cfa7acf1a24d6c906f9b81b4fb83d2 : SUCCESS in 1h 28m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4f26fb0f8e7a4908931c222585b1242e : FAILURE in 17m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cb2b7496271f426288bca438c08a4c3b : FAILURE in 1h 33m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f491db860b7346409d84ac71f6e943e2 : SUCCESS in 49m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ff947cb18d54244bac71eb859f241ee : SUCCESS in 1h 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b7e5d73ebc893090c4c0de513c0add1eadd50bf0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-28 17:49:13.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"93db8cdb6a7559e24f218e3ebd0ad4bb85ed677f","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-09-28 17:58:59.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b641729737b6492f4df1f59806fa41bcf77f573d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-29 08:41:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e5c15c8aa50ee46b55e85ba4a63b01866d32927c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 08:42:59.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5c804a8a6cc614de5b65f95b388c7c7a87c9453e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:32:56.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0624383480184817aaf897080408ccef\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9bdd337d69449c39c988f6f434e151a : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a298274dfbce48a0b016ef6db09c4538 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10861d2738d340be9dea3a3606271562 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ac2daf82b7f4a178d268c2477e10c5d : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4db3e8cd7c4b495aba215272a876ceed : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46aa1889696e47c18b6755c9e186f0e7 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383054581c9946ba92582322eb2dc60d : SUCCESS in 3m 10s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/58e459ce1e9343c7a3bc55cb53da1b00 : SUCCESS in 5m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/262f6e1d96d04b0a896e12d3d5e34b3b : SUCCESS in 1h 44m 56s","accounts_in_message":[],"_revision_number":7},{"id":"db7db04562fac689de6d5e9e2eae045f2b12fc17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:32:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8bfd3ba592e55986453d5355187a855423289f85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:33:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85dd26e88e946a2ae6da15a76effa7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c859d985e23e40ac9f0b2b5a68e4eb36 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/310948be3c6a4afb88fbb63bd1b400cb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a3715386de5b4f70ab396c2cc298fd4a495fe214","revisions":{"7c2ff4a7c6496c0b2e2c973df1ef2f4115a404b7":{"kind":"REWORK","_number":1,"created":"2023-09-27 08:34:27.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/1"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:34:23.000000000","tz":330},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c2ff4a7c6496c0b2e2c973df1ef2f4115a404b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c2ff4a7c6496c0b2e2c973df1ef2f4115a404b7"}]},"branch":"refs/heads/master"},"d12826017f4d507828c54659041bd7c09e1097e8":{"kind":"REWORK","_number":2,"created":"2023-09-27 08:43:06.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/2"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:42:47.000000000","tz":330},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d12826017f4d507828c54659041bd7c09e1097e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d12826017f4d507828c54659041bd7c09e1097e8"}]},"branch":"refs/heads/master"},"4156ed721d7e4af242087042c350bb59fad7b7f1":{"kind":"REWORK","_number":3,"created":"2023-09-28 04:38:49.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/3"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-28 04:38:22.000000000","tz":330},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4156ed721d7e4af242087042c350bb59fad7b7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4156ed721d7e4af242087042c350bb59fad7b7f1"}]},"branch":"refs/heads/master"},"1ab478b915cb71781d0217ed76bffe81cd39faa1":{"kind":"REWORK","_number":4,"created":"2023-09-28 05:35:35.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/4"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-28 05:35:16.000000000","tz":330},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ab478b915cb71781d0217ed76bffe81cd39faa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ab478b915cb71781d0217ed76bffe81cd39faa1"}]},"branch":"refs/heads/master"},"443d500d2d337d38deeedfaf49329ea2ddd2f634":{"kind":"REWORK","_number":5,"created":"2023-09-28 06:26:54.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/5"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-28 06:21:25.000000000","tz":330},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/443d500d2d337d38deeedfaf49329ea2ddd2f634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/443d500d2d337d38deeedfaf49329ea2ddd2f634"}]},"branch":"refs/heads/master"},"75ab91992f78279ef9a090b97d032319a12dad22":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-09-28 10:36:48.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/6"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"sairam vengala","email":"sairamvengala007@gmail.com","date":"2023-09-28 10:36:48.000000000","tz":0},"subject":"DNM: test py311 job","message":"DNM: test py311 job\n\nthis patch fixes the py3.11 job by updating bindep.txt\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/75ab91992f78279ef9a090b97d032319a12dad22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/75ab91992f78279ef9a090b97d032319a12dad22"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a3715386de5b4f70ab396c2cc298fd4a495fe214":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-09-28 11:04:30.000000000","uploader":{"_account_id":35978,"name":"sairam vengala","email":"sairamvengala007@gmail.com","username":"Sairam"},"ref":"refs/changes/18/896618/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/896618/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/896618/7"}}},"commit":{"parents":[{"commit":"3c8dd7d290a222abc864bd36c2180f4afc60a4c8","subject":"Merge \"Add is_admin to safe fields list for notifications\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c8dd7d290a222abc864bd36c2180f4afc60a4c8"}]}],"author":{"name":"Sairam Vengala","email":"sairamvengala007@gmail.com","date":"2023-09-27 08:33:28.000000000","tz":330},"committer":{"name":"sairam vengala","email":"sairamvengala007@gmail.com","date":"2023-09-28 11:04:30.000000000","tz":0},"subject":"test py311 job","message":"test py311 job\n\nthis patch fixes the py3.11 job by updating bindep.txt\n\nChange-Id: I8eeb1b345033269a98989ef4a563ca65fe04e57f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3715386de5b4f70ab396c2cc298fd4a495fe214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3715386de5b4f70ab396c2cc298fd4a495fe214"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
