)]}'
{"id":"openstack%2Foslo.messaging~901018","triplet_id":"openstack%2Foslo.messaging~master~If82c2175cf7242b80509d180cdf92323c0f4c43b","project":"openstack/oslo.messaging","branch":"master","topic":"fix-fake-exchange","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-12-19 19:07:43.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-12-19 19:07:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If82c2175cf7242b80509d180cdf92323c0f4c43b","subject":"Fix clearing of the fake RPC Exchange","status":"MERGED","created":"2023-11-15 11:26:58.000000000","updated":"2023-12-19 19:08:46.000000000","submitted":"2023-12-19 19:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"901018-fix-fake-exchange","meta_rev_id":"b4ea36c4f579a14624e00afcfd7de8a18d69260d","_number":901018,"virtual_id_number":901018,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-19 19:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-12-19 17:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-11-29 14:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-12-19 17:33:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 13:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 14:29:52.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-11-28 13:53:13.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-11-29 14:16:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-12-19 17:33:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a5f1b3f60b7d2b26cd6948cf870228be753c2977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-11-15 11:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb67b949bea4156037fde653ff95c0d1e2795372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 13:11:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77a218792eed4159b1d82e44135b6494\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/588567f17636470187ffed878bef6611 : SUCCESS in 1h 38m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a09fa6495c447db8ebf8b95236c96b2 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f3391b05b64f548283e82bc8077fd9 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15b8bfadabda408c900c8170d489808b : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8674b469bf904b5fb6757c06c518510e : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b7b09e936f4d80b13abc34e5944a49 : SUCCESS in 3m 52s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c12bce1c55ab45199470a4bb68813dee : SUCCESS in 4m 36s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/09f445b36a524d8c86cb4e8f7d34f7e9 : FAILURE in 6m 46s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/5c4dfcfa87904183bf719dc9eaa4661c : FAILURE in 4m 58s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/6fff756339644321ba70403b4eb53d2f : FAILURE in 3m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/25e77bc2e30849c0b906240f39b848db : FAILURE in 1h 07m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/bde513f55fc04e1bb884e802668b1f8f : FAILURE in 17m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7c6cbd6f184a469599910c356e68e97b : SUCCESS in 1h 35m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/934f64737b06479ba355635e27f099ad : SUCCESS in 46m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9cfcc0f989f4c50a360a8729a7df3a5 : SUCCESS in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9af2b26101b76bb0226fa86258851cd1d262b77","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-11-16 10:17:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20971040a85f63b932094f65c1f26662016538ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 12:06:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f82781431a8431f8027ce3944a7f928\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/613552a8906e41709dbd369acdc8c249 : SUCCESS in 1h 41m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa8123f4edc74747ba9db9e65d364fd5 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3205a92d7bb4ae2b5292d0ecf56f5a8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d41434d208324f6aa1b3813f57a32b21 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bd6b366a8744c479c08b6c9f7a4b257 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af9d472e0307453ca66eefe2e3e097bc : SUCCESS in 3m 46s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e3d85e8b1e064d3aa41416232eb889a1 : SUCCESS in 5m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/bc451fb196b143c6ad24de10c158c9ff : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/bdae5960b57c4109a8be2ae4ea4baf29 : FAILURE in 4m 17s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/25b25b85dd0a4c9484fb3ff14d481f9a : FAILURE in 5m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/629c8c674c4f4e789a30f220b7096e4f : FAILURE in 1h 42m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6176a32716ff41a597582dfc855a0560 : FAILURE in 18m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5515afb30ead4f81acd594a8d58dcd94 : SUCCESS in 1h 31m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4439b527707f4d0cbe5739668d996396 : SUCCESS in 53m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c146ac3a6d134b72a857396e4ba6a606 : SUCCESS in 1h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5686c164dc30665a1186714352831b61daa58b04","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-11-17 14:29:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"853e33a08205d995f9c0570c7ff9ac0e347a0f6c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-29 14:16:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b31a2a80569f8d7fb724f1a2e0e3660e23c4848b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 16:01:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c070a8a232894eb3a88b5d1203d3cb10\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9591ef12f7324750b097fd0bec592c45 : SUCCESS in 1h 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08f3530125f441b5bfcef4b717bdb2c0 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea3ffc7116c40949e2663f6d844319e : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5252cbba151f4a6b8b35b5d54055ace0 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/856d9512d4634308af341dbc5e3fced3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b15d2c6cf4f4f538dac79695f655e59 : SUCCESS in 3m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/296d62c87d064c0c8aa3406634cef4dd : SUCCESS in 4m 43s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3808bf290688489a8e70082865679f46 : FAILURE in 13m 44s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/1ed4f6ddf22c4f0d9d80454e05d5271c : FAILURE in 4m 37s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/4d8736dd699e4cf9b25891383fbc3f27 : FAILURE in 3m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/689a27ab3d104d6ab78f06b6923fd8b2 : SUCCESS in 1h 38m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/826a7bceb42740c7a0713cdc423dc5b9 : FAILURE in 29m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2b01281b902543dfb07f960712c3270f : SUCCESS in 1h 35m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/35ee9a2b7d4247eb9ec4cf544414f67b : SUCCESS in 59m 58s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8be2dd0f0154ad7a9262d33ac062833 : SUCCESS in 1h 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb68d7f3de96305e4a9d92e387f841cf9d777994","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-19 17:33:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b12616b9fa25504b5101583b4711ff52e7eb3cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 17:34:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd0ce6f84cff80064e6cf5bc016e0c749fdf7f7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 19:07:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f820e73c67154b14ab529cc9d49cf0d2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95cde579b963473abff17e5e1db115de : SUCCESS in 1h 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb7c0057e954f2d952ab3a59bbe56ad : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e50043f5bcb4e0e870e0b3440f33469 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/718211c1552440ef84f998440ebe841e : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2098b9db052248fa8f75ff13c4b07645 : SUCCESS in 5m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6cb2ee00af5545b99c1c0422d1e72128 : SUCCESS in 4m 55s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bab7b67691ba41aaad3c6e59c6df94d0 : SUCCESS in 1h 28m 22s","accounts_in_message":[],"_revision_number":1},{"id":"08c98a17ed5ed055ee3072943a87566ea1459fbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 19:07:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4ea36c4f579a14624e00afcfd7de8a18d69260d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 19:08:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8ffabd8d3d44f73ab4a44260d5d7600\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee53092da4b4a2abbd107385e47746e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f65607fa4890dff777d6461011b80671dfe3d627","revisions":{"f65607fa4890dff777d6461011b80671dfe3d627":{"kind":"REWORK","_number":1,"created":"2023-11-15 11:26:58.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/18/901018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/901018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/901018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/901018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/901018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/901018/1"}}},"commit":{"parents":[{"commit":"8e3c523fd74257a78ceb384063f81db2e92a2ebd","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e3c523fd74257a78ceb384063f81db2e92a2ebd"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-11-15 10:30:26.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-11-15 11:26:47.000000000","tz":60},"subject":"Fix clearing of the fake RPC Exchange","message":"Fix clearing of the fake RPC Exchange\n\nThe current fake driver does not properly clean up the fake RPC exchange\nbetween tests.\n\nThis means that if a test invokes code that makes an RPC request, using\nthe fake driver, without consuming the RPC message, then another test\nmay receive this request making it fail.\n\nThis issues has been found while working on a Cinder patch and has been\nworked-arounded there with Change-Id\nI52ee4b345b0a4b262e330a9a89552cd216eafdbe.\n\nThis patch fixes the source of the problem by clearing the exchange\nclass dictionary in the FakeExchangeManager during the FakeDriver\ncleanup.\n\nChange-Id: If82c2175cf7242b80509d180cdf92323c0f4c43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f65607fa4890dff777d6461011b80671dfe3d627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f65607fa4890dff777d6461011b80671dfe3d627"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
