)]}'
{"id":"openstack%2Foslo.messaging~905964","triplet_id":"openstack%2Foslo.messaging~master~Ide3d5885740d5023aeeaabde95fdd01d92aa43a7","project":"openstack/oslo.messaging","branch":"master","topic":"bump-eventlet","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-19 12:41:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide3d5885740d5023aeeaabde95fdd01d92aa43a7","subject":"bump eventlet to latest version that support python 3.12","status":"ABANDONED","created":"2024-01-17 15:49:43.000000000","updated":"2024-01-19 12:41:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2687f6fc8e327b5f3346a85395df215548ba021","_number":905964,"virtual_id_number":905964,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-17 18:34:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-19 11:51:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 18:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-19 11:51:29.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"e2c62b6425d0a513847bb98a76d697e73c5a800d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-17 15:49:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a28dd66359b202ca17a23695d8623d48e4ee255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ac542af369643958e47fe11c567c801\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab46fd7417214439be73a4a33c3853f6 : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80945d17305f49beb7ba449020a25bf5 : SUCCESS in 1h 39m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e46cd215aec436190119e01f4eed8da : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7967f70889649fa959b59991df58149 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e89abada7b5c4745a6f1dfcff6ceff2a : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bf9e1b2bdf7403ba228833974f9b2dc : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9e13dcce04425a93f21d2d58cb7fda : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d991a4982aa4458687e23db57cba2b37 : SUCCESS in 4m 14s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/14a1f30ceff94ba092501fb7e1087edb : SUCCESS in 4m 58s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3ffa61e0ad5a46b78748e033366e5eb9 : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/a0728d17c33949398360b45f0041feff : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/0d39a75844ab46d093a7a7c2021ca328 : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/13c331f66fd844fdae19b1d33c08a810 : SUCCESS in 2h 07m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8965d37a658f459a8ca7e11f5d321f3e : FAILURE in 29m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a4d8e2ac69a941af801b736a71e92508 : SUCCESS in 1h 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5c1895c9e25159d35c99a7fa6654fbd0085474a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-19 11:51:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2687f6fc8e327b5f3346a85395df215548ba021","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-01-19 12:41:20.000000000","message":"Abandoned\n\nThis version have been reverted from Upper Constraints. https://review.opendev.org/c/openstack/requirements/+/905905","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46e89f9dc4922025ef491c80707ff2660e3dd8a9","revisions":{"46e89f9dc4922025ef491c80707ff2660e3dd8a9":{"kind":"REWORK","_number":1,"created":"2024-01-17 15:49:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/64/905964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/64/905964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/905964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/905964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/905964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/64/905964/1"}}},"commit":{"parents":[{"commit":"800c58826ebc0effe00664f68aa5fe929899a971","subject":"Utilize the new RequestContext redacted_copy method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/800c58826ebc0effe00664f68aa5fe929899a971"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:49:29.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-01-17 15:49:40.000000000","tz":60},"subject":"bump eventlet to latest version that support python 3.12","message":"bump eventlet to latest version that support python 3.12\n\nChange-Id: Ide3d5885740d5023aeeaabde95fdd01d92aa43a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46e89f9dc4922025ef491c80707ff2660e3dd8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46e89f9dc4922025ef491c80707ff2660e3dd8a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
