)]}'
{"id":"openstack%2Foslo.messaging~906274","triplet_id":"openstack%2Foslo.messaging~master~I47fe04d6a39ed2b5f33b02fa6736d588d0383f5a","project":"openstack/oslo.messaging","branch":"master","topic":"fix-amqp1-paths","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-23 00:54:38.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I47fe04d6a39ed2b5f33b02fa6736d588d0383f5a","subject":"Drop unused function from scenario test script","status":"MERGED","created":"2024-01-22 15:44:16.000000000","updated":"2024-01-23 05:15:29.000000000","submitted":"2024-01-23 05:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906274-fix-amqp1-paths","meta_rev_id":"328804339e16f09d87f759ddd71e3bb33d6352d2","_number":906274,"virtual_id_number":906274,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 05:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-01-22 15:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-01-22 15:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-22 15:53:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-01-22 17:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d074017047ab5debc99c08ab5d0000b0582e6ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 15:44:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b48ceb7b7274b3703ef7d807f71a9509144b138","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-22 15:53:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c63bfe69923a641e3533ccf5ff0887a5d83b38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:37:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0cdb74dce743c48b8da90d99b7a79e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67e66248364c43d6b1e8f40eae3fa519 : SUCCESS in 1h 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/423dceb822ab4c9bae8586870d52a69f : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233819309c214c129e82557602f5fb3e : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514e4ccefe624dcda7a86bbb8a0571ef : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5f57fd0ae404d7cacb7f337857515f1 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77f4d0c81f349238ca411c1649a4c5c : SUCCESS in 3m 35s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/523834a558654213a3003f7f95d493bd : SUCCESS in 5m 59s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/94023a7603d14a628a832a2a510a4150 : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/f32aa0f0346a4cf29b2c9bd547402486 : FAILURE in 2m 43s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/1cc2b01bd93b4cae84e4763824e02274 : FAILURE in 2m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c82ddeabcb8a4c2ca10727835ee4edd5 : SUCCESS in 1h 46m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/01467dd469e244c99436c649bf59bf69 : FAILURE in 18m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e0bb3da7ad0145cf8a04bc5fb38c34b4 : SUCCESS in 1h 26m 43s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/915cd62b10674fc3b1245d7ce344c904 : SUCCESS in 1h 15m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a9eb69f9dc54095a83e9ed56e489aa3 : SUCCESS in 1h 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a946ec3561b1840c9ec829e7daa285818da94b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02442780c60938cd46cad694c041d9660cb54a36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 19:03:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b6a9248df26425aa2e9d2aaea72d0f1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1411066aea7d47eca6a16a261fc2eb2d : SUCCESS in 1h 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6fa5afa0504e5ea889e14a6440e0c9 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7119b4c0f20146808da719c4f714548f : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f2c23221de43a8bfb16f0a07d99358 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c5914581604aa5a9bb738399438a3a : SUCCESS in 3m 47s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/59a6ae71f53541ac9318b392b200c11c : SUCCESS in 5m 16s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a6221cfe8067429ba839379eb357c280 : FAILURE in 1h 17m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b0edf2871e29db31109772311e8223ffd80a26e6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 00:54:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07dd424b299d78f7deabd153d75177bc16cb1fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:18:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9331ff0aa384cb0bcc35f302b7a59c5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9adc03ed6af844a4a68069780a6dbb83 : SUCCESS in 1h 15m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0d0dc1219934fc4b18d451f352a4036 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4340ca9a6bd04c36b4c1e68e2bb3ac16 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c5f16d1621543379bf48eda6fd5f01e : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c5b7db442d54ebb8ce488b85202fab7 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763528e00a024cb39b3206df61b63b20 : SUCCESS in 3m 31s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/980593a295984bc7b6053f87809db24e : SUCCESS in 6m 53s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e8d6e4caa7c3433c88f863352dbb8ed5 : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/e15545d0dce04cdcbcea2278e2a40cdb : FAILURE in 2m 57s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/67acc63fd11a4638b5e4a51bc1a22696 : FAILURE in 3m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d43df59bc7074aeb96f97c553d6d95cf : SUCCESS in 1h 48m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1ecd3c26b10d4477af1e792fc3fbf153 : FAILURE in 20m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d16f9b993fd445e4974f3d1cd7437e0c : FAILURE in 1h 31m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0a88c66a420040b5acfc1d75a36c3d47 : SUCCESS in 1h 04m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2287145a61a24e5caf314b4cf6e4f85a : SUCCESS in 1h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd3d1448aa4d45c65227db89d05924cb09ad6c5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 03:18:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad34f5c711d42f803c85f870e135fe1cff0fabef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 05:14:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376ae64d1f76430b9b36526fd8c9733c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0604e39b45cc4d82b1b58d53bb79436e : SUCCESS in 1h 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421e383011964107a1b5c852f9eb7e96 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d67527a4b004db8b310c1625ab03ca1 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b08849829d884cf0808379071cc4955c : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37b836a5db44777b57795016f72e6f5 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/64535e19ffc04c7292c81de8c114a9cd : SUCCESS in 6m 36s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7dc631bac935435387e57bd5c696d012 : SUCCESS in 1h 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"836ea3f993123e941abb789fc5d91623c785a79f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 05:14:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"328804339e16f09d87f759ddd71e3bb33d6352d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 05:15:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f766b99324d45f2b08f1c0be2f71000\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad118e00fdc549c0a810b7780a5f2d76 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9202b01e27522c4250972a1037cea9d36b9b156","revisions":{"c9202b01e27522c4250972a1037cea9d36b9b156":{"kind":"REWORK","_number":1,"created":"2024-01-22 15:44:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/906274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/906274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/906274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/906274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/906274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/906274/1"}}},"commit":{"parents":[{"commit":"df0e4acdd4beabc371c0403755931b888d00f4a4","subject":"Update the python search path for extra qdrouter modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/df0e4acdd4beabc371c0403755931b888d00f4a4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 15:42:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-22 15:44:03.000000000","tz":540},"subject":"Drop unused function from scenario test script","message":"Drop unused function from scenario test script\n\nThis function is no longer used since we removed ampq1 tests, when we\ndeprecated the amqp1 driver[1]\n\n[1] 0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: I47fe04d6a39ed2b5f33b02fa6736d588d0383f5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9202b01e27522c4250972a1037cea9d36b9b156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9202b01e27522c4250972a1037cea9d36b9b156"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
