)]}'
{"id":"openstack%2Foslo.messaging~906487","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.1~Icc438397c11521f3e5e9721f85aba9095e0831c2","project":"openstack/oslo.messaging","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-29 12:51:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc438397c11521f3e5e9721f85aba9095e0831c2","subject":"Fix typo in quorum-related variables for RabbitMQ","status":"MERGED","created":"2024-01-24 15:26:37.000000000","updated":"2024-01-29 12:52:38.000000000","submitted":"2024-01-29 12:51:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906487","cherry_pick_of_change":873630,"cherry_pick_of_patch_set":2,"meta_rev_id":"1ceeb084d35b209dd0a54f97841369089c0ef818","_number":906487,"virtual_id_number":906487,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-29 12:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-24 15:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-29 11:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-24 15:26:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2024-01-24 15:50:15.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-01-24 16:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14ca5c1c8e5717a7e8846516d9762397e68e32a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-24 15:26:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"52ab5f6cc3f715aa68a68aa00e97bc91e92f0844","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-24 15:50:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c70b01cbde96a3f783d5b2388de54692f9f0acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 16:57:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17af7a4b2dac446999eb0c3eac81935a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9faf5a106de34ccca5f33024cfcabca3 : SUCCESS in 1h 21m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9490bfce3394f6c902443d5b5e8719d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42316488073747faaa1775cd74a9944f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e754848b14a94511a7c7071deea98f63 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ca00cb5abb340429979a8345da7e822 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d18b3148dc47c79953781d9825a979 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2db045f0533a4c82afa20f22cd651dfb : SUCCESS in 3m 09s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/41dd1bb2407d4178aa83984eb09d6b3f : SUCCESS in 4m 47s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/3e3f25588999434bb0ac6657bc339de8 : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/19d0cf24aa7a47b09d0a84aaca68934b : FAILURE in 4m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/648f0c98a8b4470583e04c50937cfa3f : FAILURE in 5m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0206c75ac1e4422a9987b9db70dbb9dd : FAILURE in 1h 07m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/13d28fb24dee41a5bf2b883582e64ba6 : FAILURE in 21m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b05afde725344ca681fcc9853ad9b24c : SUCCESS in 1h 28m 37s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/04ad6a3afab645c9a4bccaf4ca6f28e1 : SUCCESS in 1h 00m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2e11ff812eb404ba2432eb54227c096 : SUCCESS in 1h 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1fc545b186a95e6a72d19a9dd4e01b63d90a267","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-25 01:51:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52539dd069a4ef0e1affe604f82f1ac04c6cc43c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 03:27:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c844be45d934abca3203cf4cc5a22e8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d05a5be502244e2b7ed8e8b894344f2 : SUCCESS in 1h 32m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6611b90c2fad4d568dafa7957247e38c : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba1c98bbe4d64f26bfdee8fa40adc87a : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74522a15f58042349435e7622a2daabb : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4c52538420a4600b274dd1eea9f613b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb9b1510b6f4f7987fba0e7c9383216 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57901cce9cb54bd69e23a4a5e9a966ad : SUCCESS in 3m 18s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3748db69056e42deb90ea64bfdf1a4b9 : SUCCESS in 4m 29s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f51bb111d84646978f16452d9da53420 : FAILURE in 7m 20s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/02d72fe9692d4dc684538a4b07dc86b8 : FAILURE in 3m 56s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d0fa7bb4bb7149f9b2fa3a7f66f90556 : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b044dae1e6b14ad0aaa8214cfa01499e : SUCCESS in 1h 25m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0e60b4685d624231a5a60980073db506 : FAILURE in 22m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b00f11473adb484f9c97a0b60bba2a73 : SUCCESS in 1h 31m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f32ab20533f44e5995b98d3bc602191e : SUCCESS in 1h 02m 34s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d8c1c1c5f0d46d8a092f5b2283fc0fe : SUCCESS in 1h 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f4de55140ac2a9b11768ecad0c9bac8e75618ee","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 11:19:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"152473be2a3c5453838e8ed393a4000e8c36e052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 11:19:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54203e674ddb4c00f9ed1732720d2f4c849d02dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 12:51:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d823d85a734b7a8a8c389600337e17\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62480d612ba34cbc82e9bb43d96cc141 : SUCCESS in 1h 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0058c719a1043619771bb932218ff17 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d38c2df8524423cb05077386139313b : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be680df7b2fd41a6833cae3f75b65428 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbc9530b04a422f8fd0cf55a08cc615 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d662f2d5217d444393f3b28b7bdf3a1b : SUCCESS in 3m 22s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/511e1de868bc4069bad1ca61f121f3aa : SUCCESS in 4m 51s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4404593d3c2c4df69518f839b8235d4f : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7866f07c5c8c38d3bedb709912fe3a576b4f5611","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 12:51:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ceeb084d35b209dd0a54f97841369089c0ef818","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 12:52:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c973ad8156f042e88516e4f11d95bac7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a1493c0412438dbc73f9d28a8d0d10 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a56dc4889dc542e2907a0cf734ded6e0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00f839d1f442193af008c1da04f614ec37303e5a","revisions":{"00f839d1f442193af008c1da04f614ec37303e5a":{"kind":"REWORK","_number":1,"created":"2024-01-24 15:26:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/906487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/906487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/906487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/906487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/906487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/906487/1"}}},"commit":{"parents":[{"commit":"dcad86be5f7262f0f0323144dd47aea0b2dd1db2","subject":"Auto-delete the failed quorum rabbit queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dcad86be5f7262f0f0323144dd47aea0b2dd1db2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:24:20.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-24 15:26:37.000000000","tz":0},"subject":"Fix typo in quorum-related variables for RabbitMQ","message":"Fix typo in quorum-related variables for RabbitMQ\n\nIn [1] there was a typo made in variable names. To prevent even futher\nawkwardness regarding variable naming, we fix typo and publish a\nrelease note for ones that already using variables in their deployments.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/831058\n\nChange-Id: Icc438397c11521f3e5e9721f85aba9095e0831c2\n(cherry picked from commit 115cfb5b7cff2a9803d64d02b98250b7b0fc09f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/00f839d1f442193af008c1da04f614ec37303e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/00f839d1f442193af008c1da04f614ec37303e5a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
