)]}'
{"id":"openstack%2Foslo.messaging~906820","triplet_id":"openstack%2Foslo.messaging~stable%2Fyoga~I0f374134dd1bd7c782952c0d9a5d5931ea89d8dd","project":"openstack/oslo.messaging","branch":"stable/yoga","hashtags":[],"change_id":"I0f374134dd1bd7c782952c0d9a5d5931ea89d8dd","subject":"[DNM] Just trigger tests","status":"ABANDONED","created":"2024-01-26 10:22:07.000000000","updated":"2024-01-30 11:04:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e83031352dda5fb4dfe5e3c4c479f6bee9aa3ad3","_number":906820,"virtual_id_number":906820,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-26 12:22:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 12:22:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"caea051f07f61f809fb49213d3ffd0282e5d5eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 10:22:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4d3e79756b061d80adcf6285d17a449cac70a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 12:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e18347dac504574853705b4e32e5bb7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c947dfb4d9447ab9c9c74224979833b : SUCCESS in 1h 53m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/223b3eafc88a40138605dfb08fa34c90 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb91b36d001543caa2bf629a9fa283a8 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a2bad8dd6d4a4aa52db57bfc0dc694 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a44b1455b98433b96487987e809ec88 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a6fe0796704bdc8c696d7b79d2736d : SUCCESS in 3m 49s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/58a8fb28dd2047d4a2bd543ddedb8303 : SUCCESS in 4m 49s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/217449e3c2c443b59d8afa64157f3ff3 : FAILURE in 6m 29s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/2b1aacfc52724be4a0dc8a67eab6e8db : FAILURE in 4m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/223b27143adc43dd949d4b738ce2c336 : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/98a259fcbc1544ebbe311c3052e5de66 : SUCCESS in 1h 26m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/fa990be0c2134d9890cede6c9e843280 : SUCCESS in 1h 35m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/039c850b308c4da59f7b05265b1ba405 : SUCCESS in 1h 11m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b7ca5f28dc7c4ad791e755a8620f67cd : SUCCESS in 56m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d82dabe10db548e88497d0e9f66818eb : SUCCESS in 1h 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e83031352dda5fb4dfe5e3c4c479f6bee9aa3ad3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-30 11:04:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f07247edd663063001b04010c89dd01e7d59640","revisions":{"0f07247edd663063001b04010c89dd01e7d59640":{"kind":"REWORK","_number":1,"created":"2024-01-26 10:22:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/20/906820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/20/906820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/906820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/906820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/906820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/20/906820/1"}}},"commit":{"parents":[{"commit":"047da4e5ed8f853c1e69f73acaa9e655c88ab6e7","subject":"Add is_admin to safe fields list for notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/047da4e5ed8f853c1e69f73acaa9e655c88ab6e7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-26 10:21:01.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-26 10:22:04.000000000","tz":60},"subject":"[DNM] Just trigger tests","message":"[DNM] Just trigger tests\n\nThis is just test patch because I am\ncurious for test result with no change.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/880189\n\nChange-Id: I0f374134dd1bd7c782952c0d9a5d5931ea89d8dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f07247edd663063001b04010c89dd01e7d59640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f07247edd663063001b04010c89dd01e7d59640"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
