)]}'
{"id":"openstack%2Foslo.messaging~907267","triplet_id":"openstack%2Foslo.messaging~stable%2Fzed~Ib8dba833542973091a4e0bf23bb593aca89c5905","project":"openstack/oslo.messaging","branch":"stable/zed","topic":"bug-2031497-stable/zed","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-04-29 11:30:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib8dba833542973091a4e0bf23bb593aca89c5905","subject":"Auto-delete the failed quorum rabbit queues","status":"ABANDONED","created":"2024-02-01 11:50:00.000000000","updated":"2024-04-29 11:30:56.000000000","total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":889313,"cherry_pick_of_patch_set":8,"meta_rev_id":"de7c7382c7b963ca1be87d06b426de9caa74ef20","_number":907267,"virtual_id_number":907267,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-01 13:16:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-02-01 15:03:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"reviewer_updates":[{"updated":"2024-02-01 11:50:00.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-02-01 13:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 15:03:51.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-04-29 11:30:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"6569300b428df0ef3bec6c281e7ad9259ebf2919","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-02-01 11:50:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* oslo_messaging/_drivers/impl_rabbit.py","accounts_in_message":[],"_revision_number":1},{"id":"11e0212ef5b47e5045e321d2bbbcf2c2344ffdea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 13:16:18.000000000","message":"Patch Set 1: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f1e8fe786aa4fb68eaa94560642dc74\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53f001c5a3f14a58b531c6ae7e2906ab : FAILURE in 29m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a460a9942f4e0895050fbe835ac7ea : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2051f0d70b5409b86a53fe186a037f0 : FAILURE in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8649a03cda141019ebe95c36242d893 : FAILURE in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f97f9e9195484edb82b5dd21a2ce5a5e : FAILURE in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc9432ca31d84909a8d27625d29c6fb6 : RETRY_LIMIT in 2m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ece4d52ddf40409e284b24851ce8e9 : FAILURE in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94ee2e234032486ab7806fc59a6d44c2 : SUCCESS in 3m 22s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a12b9c5298424083906ae4c54cc2f21b : FAILURE in 4m 24s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/14c514656e964116995be1bb8a435ddb : FAILURE in 5m 07s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7f78baaa4019477e91d5a7dce528f4b6 : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/62b5166367c440129a6b121a1857ebe2 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a722cb0a62b04336aa1831d45a7e9616 : FAILURE in 16m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9e1fd2f0d367447984dfb07710ba54a3 : FAILURE in 19m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c54bd0844659479f936c28d9f3bd8842 : FAILURE in 33m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/054e601e4cbe45628fe6ec4291f31223 : FAILURE in 40m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/32d51b35fb4249c6abe2a4ac252cc928 : FAILURE in 1h 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20173f1fe6281300534f14b61b07c6b3e19a9a02","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-01 15:03:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de7c7382c7b963ca1be87d06b426de9caa74ef20","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:30:56.000000000","message":"Abandoned\n\nstable/zed branch of openstack/oslo.messaging is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4a7e34a924f5f242d23a3ec57505208920b3b44","revisions":{"e4a7e34a924f5f242d23a3ec57505208920b3b44":{"kind":"REWORK","_number":1,"created":"2024-02-01 11:50:00.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/67/907267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/67/907267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/67/907267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/67/907267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/67/907267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/67/907267/1"}}},"commit":{"parents":[{"commit":"f8c6c0a4cada8d33e495b37a8ad41a26cd6d744b","subject":"Merge \"Add is_admin to safe fields list for notifications\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f8c6c0a4cada8d33e495b37a8ad41a26cd6d744b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-07-21 14:51:51.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-02-01 11:50:00.000000000","tz":0},"subject":"Auto-delete the failed quorum rabbit queues","message":"Auto-delete the failed quorum rabbit queues\n\nWhen rabbit is failing for a specific quorum queue, the only thing to\ndo is to delete the queue (as per rabbit doc, see [1]).\n\nSo, to avoid the RPC service to be broken until an operator eventually\ndo a manual fix on it, catch any INTERNAL ERROR (code 541) and trigger\nthe deletion of the failed queues under those conditions.\nSo on next queue declare (triggered from various retries), the queue\nwill be created again and the service will recover by itself.\n\nCloses-Bug: #2028384\nRelated-bug: #2031497\n\n[1] https://www.rabbitmq.com/quorum-queues.html#availability\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Ib8dba833542973091a4e0bf23bb593aca89c5905\n(cherry picked from commit 8e3c523fd74257a78ceb384063f81db2e92a2ebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4a7e34a924f5f242d23a3ec57505208920b3b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4a7e34a924f5f242d23a3ec57505208920b3b44"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
