)]}'
{"id":"openstack%2Foslo.messaging~908234","triplet_id":"openstack%2Foslo.messaging~master~Ic017107ae15fa33cdfd4866f812af9a583a32f14","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-07 15:27:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic017107ae15fa33cdfd4866f812af9a583a32f14","subject":"Display coverage report","status":"MERGED","created":"2024-02-07 03:18:04.000000000","updated":"2024-02-07 17:01:50.000000000","submitted":"2024-02-07 16:59:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908234","meta_rev_id":"769521a0fc5033de609f421a4337c75d090754b7","_number":908234,"virtual_id_number":908234,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-07 16:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-07 15:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-02-07 13:29:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-07 15:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 05:25:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 13:29:06.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e537c30e4128edada1d165fbd82e373a53410d8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 03:18:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84bd80afe03e01386cb0e70f2edae752aa22b510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 05:25:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1837afdc13ac4209a61fbdbbd8bb3c17\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6aacfda2e5a64489b4df3453ff611c75 : FAILURE in 1h 13m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36ece8376612425b99389fd09c6352d4 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20087a0187824212a6a0c606016bcd65 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14eeb05337e74851894132029fe83f93 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6153b4405c5147ad94de981a050385d8 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772ed9e10caa4bfd89e5a82bd1f622eb : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e9ebfe51f934a189ae1b3ff9199e031 : SUCCESS in 3m 19s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/908d5213b36a4cef9aac9f1494d4f5e5 : SUCCESS in 4m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/aaed880367ce4836801f16c1fca0d509 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/22c81ddcb4b0446a8e7dd8c57eff2273 : SUCCESS in 1h 34m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3c116910f7044065aceac36f5decff75 : FAILURE in 17m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ea727e02e1a84f5199d3631e4a570244 : SUCCESS in 1h 24m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/37f1543970e9406185ab0c43cf0b47e3 : SUCCESS in 1h 03m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbbaccdaa8f248778265aebe4d1052f5 : SUCCESS in 1h 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6c4dc602d1d1a9c7fda4523a2522b9ee71b3742","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 05:40:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16869c1ec3d59e3eeb747b1690fe51c4873adab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 07:19:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f82728f0ff4be88288fa2e2c8a1de9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e55ffb434c56462b8372fef7f2c30b17 : SUCCESS in 1h 01m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0b1632910c5423d841244df16cb4299 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13417cfa4fb44413a090f6cf721418a4 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9134008874344904bdebfe69141446b5 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b12b40818882476ba8333b6b30a1826a : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a166a77205614fc7a89e2ff942030515 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52877cca45a4ae194feef4f2da2f34c : SUCCESS in 2m 55s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/006a20e11878472285a7f973a7e0b44d : SUCCESS in 4m 38s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/152398f05ac1462a8be519b482fdfff2 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9244befcfa3847f4a7eaffadb919bb43 : SUCCESS in 1h 33m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7ad680c008e74705947ba48d5eef816a : FAILURE in 16m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ebe675674cce47beb9ff66c496c1063b : SUCCESS in 1h 18m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c42acf36d89f4e62a77561b1b6e76166 : SUCCESS in 41m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca2cfebfd0974990bc6b94ccfd549eae : SUCCESS in 1h 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3276fdbb8a2951393b0f451040bca3b8210c588d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-07 13:29:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66f80276a81e48d167c64bdbdc7d6d9479228f46","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-07 15:27:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4b3badbde0657e9fdf748b959f3f808fbc1df09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 15:27:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c021f6acec0ac2ae01eeb907b276d22c716a66e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 16:59:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de75a68d337b469e9063fe241869c590\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44587f3beaf04e2489a56506615b782c : SUCCESS in 1h 30m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d1216bf57f4cf9b9072cded30a6c46 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fe19c515d24d3694577332d611f275 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be67bcc10a47427f87779e977fc2b8b9 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd815142ebd4e0da82fac3fb2df76f6 : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8ec396721f542e09bf687ccd838cbf9 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/468d176c1aa741aaa44178312189d3be : SUCCESS in 4m 51s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c5cb5441e1624fc8ace2051e02abadc3 : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"205338e8e7810953ec5d1cb2f35a6abb13645da6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 16:59:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"769521a0fc5033de609f421a4337c75d090754b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 17:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90d30d63bd94428963cab559b71bb5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151259ba752043a4b6b97ca9011d5d49 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3932a478032e4ba1919e4e5ba91a4faa : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"795185d43c42edb978ae90a8216a470b3622a415","revisions":{"795185d43c42edb978ae90a8216a470b3622a415":{"kind":"REWORK","_number":1,"created":"2024-02-07 03:18:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/908234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/908234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/908234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/908234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/908234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/908234/1"}}},"commit":{"parents":[{"commit":"b5244bd05a7e712b0f3f6513f061de25c0048d30","subject":"Merge \"Bump hacking (again)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b5244bd05a7e712b0f3f6513f061de25c0048d30"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 03:17:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-07 03:17:40.000000000","tz":540},"subject":"Display coverage report","message":"Display coverage report\n\n... for easy reference. Also make sure old data is purged.\n\nChange-Id: Ic017107ae15fa33cdfd4866f812af9a583a32f14\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/795185d43c42edb978ae90a8216a470b3622a415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/795185d43c42edb978ae90a8216a470b3622a415"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
