)]}'
{"id":"openstack%2Foslo.messaging~912499","triplet_id":"openstack%2Foslo.messaging~master~Id84adda53744ef55fc171d35b372d178f38ce068","project":"openstack/oslo.messaging","branch":"master","topic":"bp/oslo-http-driver","attention_set":{},"removed_from_attention_set":{"36129":{"account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"10115":{"account":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"33451":{"account":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2026-07-17 05:34:21.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Id84adda53744ef55fc171d35b372d178f38ce068","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","status":"NEW","created":"2024-03-12 01:09:54.000000000","updated":"2026-09-10 23:33:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":125,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"54ecf8a5d0e77b4741203ba64fad0b3c8decc7a0","_number":912499,"virtual_id_number":912499,"owner":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-17 07:15:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}],"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}]},"reviewer_updates":[{"updated":"2024-03-12 01:31:36.000000000","updated_by":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"reviewer":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"state":"REVIEWER"},{"updated":"2024-03-12 01:31:36.000000000","updated_by":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"reviewer":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"state":"REVIEWER"},{"updated":"2024-03-12 01:31:36.000000000","updated_by":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2024-03-12 01:31:36.000000000","updated_by":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-03-12 02:39:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-13 07:53:22.000000000","updated_by":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"CC"},{"updated":"2024-07-05 06:29:13.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2024-11-25 21:55:33.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-09-10 23:33:40.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"CC"}],"messages":[{"id":"95ec708b85b94e8905279e00973f4a1ccf231138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-12 01:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1f1cf0fa47363e215d3a61aebda4263c99d7d18","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-12 01:21:13.000000000","message":"Topic set to oslo-http","accounts_in_message":[],"_revision_number":1},{"id":"a26d4d988d3f73f80902e44569f718c743b3f614","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-12 01:55:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":1},{"id":"45092c1972760b8812270dc61110410cf1d9159c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 02:39:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffd676ddacce4358b8ad0ecc72b6c456\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ea58f4be6f004578960f1a89727ddc7b : FAILURE in 2m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85d57b3d6e1b47fdb292d1d66cef0334 : SUCCESS in 1h 28m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd9ece51a3e145519734389f06a930bb : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e097b9071546908c286ce30b40db23 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c3eacc4890744e9b17966ebf03c1e3e : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/934e6ed5a6ab4866b0afbebd92b8133d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c29c7f372445bbb008539468337c11 : SUCCESS in 4m 03s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8837ea339d5441f6adea2899b01b674e : SUCCESS in 4m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/760a43848e244afaab6a6cab2cd3fb52 : FAILURE in 5m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/87427cfcf6c64948afae94ecedc0e0fd : SUCCESS in 1h 10m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/08fed03edabf4945b4ee71f6f3e38384 : FAILURE in 15m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c4dfb17f0c9843e083794b7ed70f0d97 : SUCCESS in 1h 10m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a48570b104dd4e9e9206ef89c01fc754 : SUCCESS in 40m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/27c74494b09e415c945f5a3ae8692cde : SUCCESS in 54m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8aebf10b636c7097dbe098929517ca4563207884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-13 05:33:36.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e4af2cc54db970c7d038fb075c26509fef4636e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 07:24:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc787d487e7c454bb2d418204b2015d4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3077703b9fee46bcbd3d003e87041e85 : FAILURE in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af5e18bb9804495293e86a914f85f229 : SUCCESS in 1h 16m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85147b2790d424287b907368896795a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf9cf5f09414eb5b0dfdfcbb607352a : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f670529efba943e4a9007b7b0c8ad451 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b854ec7cfe8414ea402e7a6fbeaf1e0 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df41648b1e7f4bf7997e9f8a87e25c80 : SUCCESS in 3m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9b54582d5ff0414e991f4d77e582b972 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/11e6b9c41f0a4a7386a3610c5b9cd165 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e2f14d22bcb14d3ab9489781f23b8438 : SUCCESS in 1h 44m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1779c6f6d84f4279b3bf674bcd3dfa53 : FAILURE in 32m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f7d8c85f57964c8680e6f1f904bcb862 : SUCCESS in 1h 39m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8fa555afb6e74ccf93207e59ae5bc4eb : SUCCESS in 1h 02m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae9b9e48746048358808fbc221a5c964 : SUCCESS in 1h 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91bb1732388368ba8ba3d666e6d3bb53185ccecc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-13 09:54:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6ee05c6261df8fd8871a627c3cb2b81f54741f09","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-15 05:21:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1cb1c298f612b82e0c08bb73205a292b4b1c959c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-15 05:47:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0fc45593a8265b79ee45769c57d7c3aa5d1bce1e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-15 05:56:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"842bcaaefdf4d0629d18bcf41332f9ce59724f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 07:29:43.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab6cf88272194a08b11fecd6d4ce91a3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab67a6f0a9fc448dacf7a25cb2035828 : FAILURE in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58f788a245f54baa92715bcc41e1fda8 : SUCCESS in 1h 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb47739e51e74f2a9aeaa0d5466a6946 : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbdde6602174a56971bf2788a6ea7aa : FAILURE in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab64a7a9d1f423eb7d1b2293345ea02 : FAILURE in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e51f814dbdd448cb8b198798468ab6e : FAILURE in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c1ea5d68ee4d46acfa761e86c18876 : SUCCESS in 3m 29s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/148254189d5c49c2b4a683deae4a1c11 : SUCCESS in 4m 53s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/fd1e1c0f345347b0a7cc40a78da377fe : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0954cb99ee83413ab529bd42861dd81b : SUCCESS in 1h 35m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/155caebfe49b40588037f4194a59a3f6 : FAILURE in 27m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8bbb627a61e44b4baec1742890fb06c4 : SUCCESS in 1h 36m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/943573dc4e3046cf82f67b108e4eeb2c : SUCCESS in 1h 04m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/41a2706af3e14303a31902776683b401 : SUCCESS in 1h 08m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f484da4db5496f8e167e4e72f4e15058653d8960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-15 08:11:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fa1e499da4890a347b5ad262525159d6b19241e2","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-15 08:23:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10a403b5dab9666bced060fa6e395e9290709292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 09:47:34.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/025126c1042e44d787589a162e29e2c2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5faf767c539247e396d9aa1fcf156672 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52de2741f2ea4922a7be3603b71b8843 : SUCCESS in 1h 22m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84c18e74c9d8443c84e538e8e5672e42 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb8ebe3115943e49e9d4c9fbe889357 : FAILURE in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/933b1fbc0b0e40ddb8150254f4c30fac : FAILURE in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebcd3c598536467b9a5feab3068ee5cc : FAILURE in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7431cf49d99648bc8fcce2afd819e020 : FAILURE in 3m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/cd1822453e59461b919d6e1d68cfc437 : FAILURE in 3m 44s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/43f580937fc14fe89eec78b21c51389a : FAILURE in 4m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a57ed3e4a94a4a19a4783448b6e53319 : SUCCESS in 1h 33m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e287e565324d428e9b104e15b7250508 : FAILURE in 15m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/af2489ca9c6b4ae2a34214488dda0701 : SUCCESS in 1h 23m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/53b09ad7ca994e38bac7242fc78ab164 : SUCCESS in 1h 17m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7dfa75adbe5414bbdc704a26897d714 : SUCCESS in 1h 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6806026efe0e38334811611dc0a3f6358f0ec200","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-15 11:18:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"029bfe08bf561bcbc56b75da2c42f859c1602f6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 08:11:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9bdf88f4bab2bab2ba77798c9ca57b1a18bfcf0b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-22 08:52:22.000000000","message":"Topic oslo-http removed","accounts_in_message":[],"_revision_number":5},{"id":"adc72531175cb9d29480de109e554ae9702cde1c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-22 08:52:29.000000000","message":"Topic set to bp/oslo-http-driver","accounts_in_message":[],"_revision_number":5},{"id":"be8c2e9a6e91bf048d6e9767b2108fa4716938d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 08:56:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7ac77d69bcf715de98634cb316d2f7df37ce4e13","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 08:57:44.000000000","message":"Topic oslo-http removed","accounts_in_message":[],"_revision_number":6},{"id":"aa7739e27e813c7eed001bdfb2b61ab0e4a19b40","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 08:57:47.000000000","message":"Topic set to bp/oslo-http-driver","accounts_in_message":[],"_revision_number":6},{"id":"43c0b7fc73ace80d2552208dfd71ff6c861dee11","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 09:21:59.000000000","message":"Patch Set 6:\n\n(22 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf0a8f157ba03803944467499794d3bf185e1614","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 09:28:27.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1fe0966891d4edc6171fe5b7fbc4ef300ac732c4","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-22 09:51:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c7a8c8d7aae0a2c272989ad78aa8e4c931cfbd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 11:05:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/678416feaa7f441dbd2b8482c79c4603\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/79727fac55a54530bea4742f44206858 : SUCCESS in 2m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cfc01d6f319433abaa7c9a31ea8d415 : FAILURE in 2m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df1943a8a2b549df9860e3448599aa11 : SUCCESS in 1h 30m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41b3b0d766604a0580ddfc4a8f00b09f : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cea6f56d01a4e0b9ef3013ebd598537 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e99e44e04de643d49734f4a7aa29bb10 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4a91b7122144188921f000b4eb6187a : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/987cd839020d4dd096ce85fb29c8b6df : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916d3661986c46d1ba8d1175a9317a83 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb0e299034f44ec384d29cdcf82f3e50 : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/2c572691a3a341c495166739b4684cdf : SUCCESS in 5m 16s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/ac4358a77fbc4b4f885e70c9541e4b1e : FAILURE in 6m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/deba94433e1b4f31b19316f6a0f92c24 : SUCCESS in 1h 36m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7f89d8b00119413a93daa46c345b261e : FAILURE in 29m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9d25c8d020294fd4bed0489b089e53b0 : SUCCESS in 2h 03m 13s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9e83c4214e03428493e48ddef2b5d6d9 : FAILURE in 36m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dbbbe92cc944a9291f1ef570289fd62 : SUCCESS in 1h 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a97f734065ded2f6f00b6b09ff30b98595be45f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-26 01:08:09.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"55562fbcf4a453a6568c0813c6eecbce699fd53a","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-03-26 01:14:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9497b82771523a1e1af835233ae94636799c9feb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 03:05:09.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"733879d5e23894630bca7f77d1b1fb5f8c479e8b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 04:01:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"fc04bd6a2e995b6a850d7e73b90ce3d770b080bb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 04:37:38.000000000","message":"Topic submit-patch2 removed","accounts_in_message":[],"_revision_number":8},{"id":"fcfefaaeea9d1307ef4c8193a6ebd3470c5ad946","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 04:39:20.000000000","message":"Topic set to bp/oslo-http-driver","accounts_in_message":[],"_revision_number":8},{"id":"a48810f1fbf350a26a7da03d40f3191658b26399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 05:18:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea36c16dc1f4f5d9f74f50b8d4bcb43\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f9401c7b11d4497827334881c3c2582 : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2abaee423e70414e8d9454db047c4526 : FAILURE in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8445f61795f04c20845b65634ffda3c3 : SUCCESS in 1h 20m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cae88236e444619870c666939accb94 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ed1dc0ad854915a63b6191a21bdfb4 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ff4f6b6996f493a8e52681061840f78 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc37682ac28c49d292310ed638374b47 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f5a646eb69c4c1a9af780f5ebc9ee6e : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6bb1f4d9a84f299cdaff82770972d6 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbee7b07c81347178b760269dd05a908 : SUCCESS in 3m 07s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/050fc16293954170ad0908671f0fda65 : SUCCESS in 6m 04s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/85ade07dcddd45679c8bfcac9536b5bf : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1fdaed4411aa443b9300533a02bb9e28 : FAILURE in 1h 42m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2a45e4b1dff245c6b34fda3174fd0084 : FAILURE in 21m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ed963499d36e40509df53104cbcebb30 : SUCCESS in 1h 32m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ab47ce1fa28b424980818b073c81de66 : SUCCESS in 43m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b01eafcf50fe4412b8e2a49f87844775 : SUCCESS in 1h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dbb9f29d86c0f14b89b2f36be16e56ea1f51ceac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-03 11:17:48.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"95f234d3261bde3e575932f2f71b60dfb3554017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 13:51:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/792025b5fdbe44c2b68a63de22bdb0ef\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d40cd8a706e4a24946e941269a9e21a : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64dd88ffe666494cbfdd9adbd0c5aaad : FAILURE in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/141becb75a1c415fab45aa30998afd00 : TIMED_OUT in 2h 05m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9993fa6d364a4fb776e0ef77be3029 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5b6380f5a54779ade84c01b3640cd9 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c81f4309fc5240f29187295f201a73eb : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/146e1730f48446dfa33f7a658a5a2014 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f30c322f1192431fa0754ed2a362f4f1 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0cc255d5bf4111b60be53f3558963f : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe38c3e078134b9b986ec961c377b0e4 : SUCCESS in 3m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/865f25cc4fcb4b9ea89e2a60acfefd49 : SUCCESS in 6m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/57bfc176933240658646c855557c2ad2 : FAILURE in 7m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1ca81c769b044bd895f6e9fe9c02ac88 : FAILURE in 58m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5e24e05cc0f64dd2af18fb299320a161 : FAILURE in 20m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/43ba9b9e4ea54761a1e8f192e7520ec6 : SUCCESS in 1h 32m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/40f7b9d5df5f49ef83f76fd4aae0a4a8 : SUCCESS in 1h 04m 57s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/7e2c4b94067c4f9cb859b8126f0626d9 : SUCCESS in 1h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"40335ada045c9d8c96ab265185aacc9fde5a8d75","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-05 06:29:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e3d875750fabdd2ededd7783bbb93edf0edb8c06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 04:26:20.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0d1109521c6715617542c6f89ced031b752fc9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 04:31:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bf14b4d664b9d97937650c935ec51117588092e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 05:50:44.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e8202ef2593be8da1fc18d3ecaf6bfb3db1fc15c","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 05:51:37.000000000","message":"Patch Set 12:\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fe787fa0809bc676c2be11d7c10b87db5e12186e","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 06:02:12.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"de314bbcd53ae6f11667e5532d67b78f1c945b1a","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 06:05:32.000000000","message":"Patch Set 12: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b190284a22b5995e7114e4fbf301703133869b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 07:24:06.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64556dc728904031a44ce1b3c210b661\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c468c7e143a64f2e8b4d973ec86e4543 : SUCCESS in 1m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a48308ae013d4505bdb737df9b6fa601 : FAILURE in 2m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27a95ec578c54773b335eb5cc402e315 : SUCCESS in 1h 25m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1208466ea846508d2252846485a7a8 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb489078506948bdae1296254a754a35 : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d19537071f44f29a36b94dec89fa5c : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6558538ecd944b81b8576ea0649d4fde : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4719edbf2d4e638952bd0f4f780f48 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fbe7bf62d554ff7b234fc194bf2ebaa : SUCCESS in 3m 21s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6af5a32950004dcd884d126e5b751405 : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a7c15a953032431987fbe5560375f327 : FAILURE in 5m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2e4c6daaab8a4dffa0948655cacbfed0 : SUCCESS in 1h 21m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/484ccb8358214c0bbf9001969280d403 : FAILURE in 16m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c977ff5c267b4e7b8cb6843b823d686b : SUCCESS in 1h 26m 29s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ff63459cec2a4feba840099b5babc04a : SUCCESS in 40m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f916722b1f874b0d90e997cb593abffa : SUCCESS in 1h 07m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"44222a30a141a4673ed14ccb8478b40d4795d784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 09:10:06.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"23118ff459ba9baa6f2656c20e0c7706ea581f74","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 09:38:08.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"2616c9757420d95d3d7bae01c22e123ab9172878","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 09:39:51.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"379150e469abf34a49dfe7521970d62d51edfa48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 11:35:26.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d980a3e7cc475897542903bfd52996\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/73fd2eb8a3a8443a9ccbf048f89d85c8 : SUCCESS in 1m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7f873a593814d6ca4047bf264ed6dd9 : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca76d9d21e2a4dd4840d4266a97bab34 : SUCCESS in 1h 26m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a81627c2638474cb4ba329e561ea493 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3721ad94004a2788f4dd40a8be47b4 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97c1621f202743f7b4b64835995e07f8 : SUCCESS in 8m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8de1ee1694434d23aa2801b8c9c2430c : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e63e05c32049978fe8d18c6a5201e7 : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/370c3a01a6af4916afd7588a83efd2cf : SUCCESS in 2m 55s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/16bb95d7a03e4ef799391801f382ba79 : SUCCESS in 4m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/99ff42946b944326a1cf027a1a72097c : FAILURE in 5m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba7895ab54964c6196061a98d510a901 : SUCCESS in 1h 45m 26s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d145bee2871949fc93c33894ff0a2474 : FAILURE in 25m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9e45ebc5f2864eb4921ce3c347e1f351 : SUCCESS in 1h 29m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/970dfbc1c4604afdbcb2eead1a8fa513 : SUCCESS in 1h 07m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/32dd159d92054fa7adb33cc98ece6eb8 : SUCCESS in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"573c6dc8f1b905489bdae106597e3afd6c4c2a0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-12 09:01:54.000000000","message":"Uploaded patch set 14: New patch set was added with same tree, parent tree, and commit message as Patch Set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"58513fee4bb58f2742ad0a86789ea4d20deedf70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 10:48:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731f369199be4ca1be1ff7c59402bdf7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/febd6576b45249a881f1630f28bd7a66 : SUCCESS in 1m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a12c6b35e27d4cd09c51f5e23c8b68a1 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a289ecdb349486fb714b2906baf59aa : SUCCESS in 1h 37m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d7860fda74495aa9d5c5804bb93475 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4f38ae2dcb4168a0c14d7b27a28dc8 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7267063322844e169c9ceb0ad8708ffa : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b76fa0682cb4d55b28b4f34de12e800 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18aef30af9534cc895b9709eb85ce24a : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8206468ed2344aad9e0a07e267383bfe : SUCCESS in 2m 56s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c52102f8c5694043b22d47eada31047b : SUCCESS in 5m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c3aeaafd777f40118ed0d1e5b5362bf2 : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f66428fcb234a759262b7363f81a6bb : SUCCESS in 1h 34m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c0db0df0ca79475ea43920102dc771be : FAILURE in 17m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/487fb2ae5cf74f4c9b48c072021fa8a0 : FAILURE in 1h 14m 46s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/db3ed06aa8f04697a2cff6da3adecebd : SUCCESS in 59m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/abe2d82fcdaf4b0abec5cee386b44e56 : SUCCESS in 1h 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a223ddd28f74b9416401e766e3aa92fcaf0f6938","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-12 12:50:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"bc9bc4d1aead46611683b7a0f3972bc7681dbd18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-13 03:06:03.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"9cb67b76a031725dbd6b58c0212fffaf561ae6c0","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-13 03:06:53.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a5767392f9c6cef6805cbf362b57d5bf8ae51330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 05:01:44.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7644001e2654e19b14e176e8c957b94\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e6f61695fbc4553b998d8f5a0e52d71 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4f85b5c96e824f78aaf57f1362a4b089 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fce9841346e492ebcd0da13dcced892 : SUCCESS in 55m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/825a96bd4d4248d9a30ce46409c03cb8 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e209c6a17044caa256080d3f8c7f3e : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea834d52be0641099c056e9161593754 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4488b04e60ca4be1baee66f02e0a767f : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c680bc3cf1641dcafd6fa9c77436f74 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b6bdb06ab1441dbcff26eed2c028b7 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c1a9e6739fea4990b7a4a9fe4516d1aa : SUCCESS in 5m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/aba1e903bb204bdd82fead346a6ce7f9 : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6a04c52ace4a413ea5e27d0fd6297886 : SUCCESS in 1h 49m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/989ee8971aa443e79f6bde72a2a0becf : FAILURE in 16m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9d6256c16e754f5c9c39bde7af636fa3 : SUCCESS in 1h 31m 46s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4c77716f17864e3da5ffcb02f8b6adcd : SUCCESS in 1h 02m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/05e768e28b214539aa824975d1c18bc1 : SUCCESS in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b654c65b32384833c72d6f74a8847bc450ea2347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-21 05:07:09.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"bd8d06d59b27dda43a85386cc8e29143adf65b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 06:53:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465313dc862745db9c287e3efb7fda5a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2bf0da699df401cb8d9e34623534b30 : SUCCESS in 1m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3dbf5220576d4a598b4ae90c5485f0fe : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11897415eaf14dceab4165ee3016b847 : SUCCESS in 1h 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c51f56f4e3734cb3b8dccf113e3bca8b : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede58de14dbe4bf8b147e6fc4e65f0fd : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65748adf252847d4990df0a852d4061b : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b741d9725cbe40338e174788d37abc5d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb925c92b6e244a897adde2ac30c915e : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dbb73bd2ce249e9a49dab328752e552 : SUCCESS in 2m 45s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c582df16588f40b5b4f0c8114e7f5feb : SUCCESS in 4m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/204fc77da7734705a7e9e4df994ff2fe : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b8e713e04c8d403ea9eff60721f844f8 : SUCCESS in 1h 35m 14s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2ea72d182db8474e9cc974dd902a6e16 : SUCCESS in 1h 24m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3c045110e24043cbb4b9bb839def4cf6 : SUCCESS in 56m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c85ee579f9bd4a6888c7dbc600bc1df8 : SUCCESS in 1h 04m 39s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8228d3bd42e6185737dd041e4262aa0bf8d3393c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 06:03:22.000000000","message":"Patch Set 16: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c50b9bb746c244cdf839a05772e836a3c3f98ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-25 07:12:09.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"199af4b83428fa88d39d4f68c147e19b49fd8c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 08:50:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f692030403df41f2aaa5353532888c30\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/586e5e69005a41e5911a46ad4d9143ab : SUCCESS in 1m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/647abd91a1e748f991f15656db0cce20 : SUCCESS in 1m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d92fc8e26d544dce900b8384dd48afbd : SUCCESS in 1h 29m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41049c296fc74288834aa878748b6dbd : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa812bab71d4eabbc7a4ca963ddde5c : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33322c7616a646049611e97cdbee5707 : SUCCESS in 6m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/799eee40c0b547c3b7b1752be614e775 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ea70969374491aa1281a5e44ea9cfd : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21df81891f0643229a7ff7a3585d09ca : SUCCESS in 3m 47s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/279895e24ac44e71b01d3f961082f5f4 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/cb067b99ccc64840bf35b94f085b00a5 : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0983b335b8ce42d2835e395eb4fcdff8 : SUCCESS in 1h 26m 21s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/760072f89ca54f1d9d898bbeb531f618 : SUCCESS in 1h 32m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/91c7a947b8ba4f6aabb669641ef150aa : SUCCESS in 58m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/92bb6a1475f94ad89ac1eb660f333088 : SUCCESS in 57m 23s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a30e21dbb4470e89273f660b96616c6b9b74ab79","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-25 08:54:03.000000000","message":"Patch Set 17:\n\n(9 comments)","accounts_in_message":[],"_revision_number":17},{"id":"8de794a11ae4ccee404da7e9c53243e27f9e7484","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-25 21:55:33.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"de8ae48a34672622852b7799ac9261718176aaa1","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-29 07:36:23.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f575ccb713a78e2ed69cb82a2225778dbb4f1772","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-29 08:37:12.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"afbc82525e330c15a41368158bcd0ca7de55dc3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-02 08:20:21.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"30fff69625b9c775fcf6502a581da305623331e3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-02 09:41:11.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"d5c86aaaa99c4d1a60bc197618c84dd4bd9f439c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:57:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb4a447cc454031892eee551874148e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90734e6bf30f4713a2d878aaf3818b3c : SUCCESS in 1m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c15bbf25fe3548b7b317e44ebe8826d7 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c6396347d2a428a9b83e7d96d8b69be : SUCCESS in 1h 30m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25a445f215194a48962a02c76123aed4 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77f6d7352c640b8bace58f78ad1928a : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf268bfad6b64269a6b036271514f951 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecaf51d0afa3428a98a450499b9734b5 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce38608eb2654da0a3fb5b25627e8222 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea97752d6f4347a5b77d80cc33e50aa7 : SUCCESS in 6m 06s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bf8cb7129b4447c2b77793a16fa468bf : SUCCESS in 5m 23s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a30e43254a2040c3a9dc52d75186b22c : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5451aef7690c4e93b45aa85f8ca73e1a : SUCCESS in 1h 30m 08s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1819a1ed077b4c63a6d6ff76d1fd2a1e : SUCCESS in 1h 31m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fcd7dba7278b43a7b222e10bf8a4cae9 : SUCCESS in 56m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/25c5b408b1624944ab8084586acb70ae : SUCCESS in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"627e1ef80dadfdda675a675cee25f4ccedba1471","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-02 10:42:54.000000000","message":"Patch Set 18:\n\n(5 comments)","accounts_in_message":[],"_revision_number":18},{"id":"1569a440cdaed75c6e39eace40ce72b3740d81b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-09 06:11:57.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"c3b3bd296b6ffe414d2040e588ac8a97c1ca9e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:06:12.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/286aaca0dd084363b12fb47aa4db9903\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a325559b0ef24df0aa3984e624160ba4 : SUCCESS in 1m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/52c518f1fc2d4c09ad43d737e0b2298a : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d420724bbfa7485a9b6a486d8b5762b4 : SUCCESS in 1h 05m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fac0ab235574846bc314768846d94d8 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4eeff5a1444c019e1d3dc797357367 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd5088487730448486529ef7f7deaf26 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c89da33778384152b55d7dc53103eeac : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc58e32cffba4542ba7f4f363af294ab : SUCCESS in 2m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e4bc94e49be426e81237df490d5b226 : SUCCESS in 3m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ccf8029ccff24d4d87c7e8e4e5f02e10 : SUCCESS in 4m 58s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9d103e0712904fd9aa102dc87c6fedf0 : FAILURE in 6m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/efe4aea58b8a4a61a143134ab113ddd2 : SUCCESS in 1h 47m 10s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/96f5535268e843dda23a9ecdbf22c729 : SUCCESS in 1h 45m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5ef579b408254b0e959180bfa83c4f10 : SUCCESS in 1h 12m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e7a006f27d94861aa1c8fc7ba53273c : SUCCESS in 1h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5fa6908136fac5f4462265ed7cb6cfe541055f85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-19 05:58:24.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"253658895bd6b8bfab2298c27a9584376ef97696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 07:51:29.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e7a689704cd4c60b445e85a5c662bd5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0887f9de4374e078a9ec8ef5a84235d : SUCCESS in 1m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1a3f38a246040cbbdccb935fce4a715 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90fc5ef6325044cc8f36363020759189 : SUCCESS in 1h 45m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b891f2f43c3443d89db77f13fbc6073 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18280644d53746daa178bcd169d05fb2 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0ca516b9f4449b8159a4f87b116fc9 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92d143a1c6a84edb8f22fe1255a738e1 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c10c5713fe44000a11101da144588ff : SUCCESS in 2m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/344a4cb5690b42f1845d46d92fcd1087 : SUCCESS in 3m 06s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9ee7e61e4f864963adecc154c16e6158 : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/784d86304e474583844b77157dd49e65 : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/66c3901eca914e0e97ad53f2e3a37938 : SUCCESS in 1h 40m 46s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/27b3ad40895247dc9e70394c3491fb1c : SUCCESS in 1h 47m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b6c71bc06a664673b23218d8c8880084 : SUCCESS in 59m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/11bd1e092c334e6fb93ebdd7f0e9d424 : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5a05a43ed624951eededd4f9f7230cc8b733bc24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-27 11:06:22.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"c80fe9ec986c9875b770119dec0f1f020a75a3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 13:41:03.000000000","message":"Patch Set 21: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e87ac9d28734c838c7ca5f44aed423c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0fad40dc833043ba9ab485e2a5d1ca71 : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0a953bb72db4838ad1b55c08bb0b068 : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/868d3cf6005342a6ad6e5e4b1b8a0f7f : SUCCESS in 1h 56m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b80f8a2b932e4612bc28a3c74f332895 : FAILURE in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2e02da32324b40b320c5ca9bbc3b9c : FAILURE in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58cb387c003e4698b9f9aaff7c0bf07d : FAILURE in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de03b1973ace44138af84ae62f2910d5 : FAILURE in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bbd1f8382e4ec5a115fd48081ae159 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea90caf892874d5385cac45e9fbba3ae : SUCCESS in 3m 11s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/7947109b928947d98e79051e53810a48 : FAILURE in 3m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/456d1ec14c0e4856bf3bc27278970ebc : FAILURE in 3m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/18194d1ecb2a4abc8c3e005d262ab3e0 : SUCCESS in 1h 20m 12s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f5eefe036dca41c48762d95389abd170 : SUCCESS in 1h 16m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b9af9cde366e47a090d783f319ab5f0e : SUCCESS in 43m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f19d865deabb47e18b193114b69eeaf5 : SUCCESS in 49m 38s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"36d5591bbda67a401967c0f8d3fd4b6dc6c3ef63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-27 14:53:30.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"8020c984352ebc7fc5284bc9e74f14659c5de1a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 17:35:45.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d55bfd5d77c04f819b9e414ce8e8479a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c5407086a124252b512320ab161a1d9 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/250c10c2bd8d4b2382984c259d6c5158 : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f339e2bd8304fd2b52adee99c679672 : SUCCESS in 2h 05m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b38da877464d4c899e9593975b6224a2 : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4718cbbc69eb488a968542787a93f32a : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c068dddc6b5c4be28e1d3fb8ed62b71e : FAILURE in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a48c79e16b404ffb8d335e21e48a5162 : FAILURE in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4b978757a94799abb7db1a0d1f7e49 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a598c0c587840edb53a39ca6f59ee0b : SUCCESS in 3m 32s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/12ca54823bf0443aa6f7fd2cb074f897 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/1fcbbd2546364f07a18abcdc16c266eb : FAILURE in 5m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/db48e37e626f4cafa01be65a51b64e9d : SUCCESS in 1h 36m 55s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7b2a165799124354ad1fbaaf544f8e37 : SUCCESS in 1h 21m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0472d4be69de4c2aa1ffe47c910b9b5a : SUCCESS in 44m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7adf24b82bb418aa20a1459a6cb761e : SUCCESS in 54m 45s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ea3b3b32182dc97c933d6f735717bf9061f87cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-28 06:59:53.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"b45e2ad0ca9e8d417fcaf796ae500e7e8c23f430","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-28 10:22:28.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb73d5f88f5340d1b27c860548482807\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0e1c7c6050e14e4e82afa1b0ab173d2f : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0fdf3a61ebc43568317dc99d1ff8a4f : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a36c2dc0a1e04180bec378a431bcef6d : SUCCESS in 1h 13m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505faef91f1247c4826be7b8fc3647e8 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b9d3757d894dbca0f6b7b8b99d215d : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7963c94307274002aea90d8e1d9dbffb : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35fd76a4c93a46d196bb3deceb82c7a2 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a8e03039744d5f8c7497b5f7d142cc : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a63cdb95bae44558b17528a28d659a4d : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/694f70cccdc94cfeb614d6bc3da7f718 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/10447a875ee04919b29c759763f93964 : FAILURE in 5m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a08478f9a5804f78aff741c3417da661 : SUCCESS in 2h 07m 35s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/90fb84600b304806a408bb39230b9b54 : TIMED_OUT in 3h 08m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5ef130e8784c4210baeeeafeff35cdca : SUCCESS in 1h 04m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f0b877a85f14c75abe5bf45b7fb3614 : SUCCESS in 1h 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"b8f6f0abbe1a5ee3a8aa6785fce5415e4a44c066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-07 07:47:38.000000000","message":"Uploaded patch set 24: New patch set was added with same tree, parent tree, and commit message as Patch Set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"0f67d71f4e0b974f2d0cb93d9d5b2aed54fb2941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:48:37.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af2200281e8946a394433cecd0233445\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5fca24372ae04f6cb01c06ff3c643c93 : SUCCESS in 2m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30dcb01c7cb54b6c8b4f613f6cedfb41 : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d8651773d0e4cc7997713cf4c4ee8bd : SUCCESS in 1h 42m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/623b5acf7e28400c8ac52f5eb92d2ef5 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234c680cacde40d3be6de0c4df578ec5 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c2bb32b70245e88214affb3946e6db : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a046a9194281484ab0d9a948cc68bcd1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4f2b8852604f19aee5c9af4b18f9f7 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b3b996efde648f2b3df76fa0ba19f06 : SUCCESS in 2m 56s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/fcb236978fd242a885db2aa48746d9d9 : SUCCESS in 4m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8c7e2b196bd44ecfb06a1ddc67f0afc8 : FAILURE in 5m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4d75a77f9d154c40ab4f8c8f1b678d09 : SUCCESS in 1h 37m 09s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3bb20e989be7495f80d8d4fe10de7862 : SUCCESS in 1h 39m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d66acbd3c1eb484cacc559320f306688 : SUCCESS in 48m 12s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/56d8d627b10c44b1b6a7ccb2dd79e25a : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d069651d1f466fb5194968eb209949e26a7f1dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-19 08:34:25.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"9dfee0baa0cd4fe320bbd4d7f871466cb109a9b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:21:18.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803a0fbdc03e404c8f6a5f19b8062e5b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/939ba2b80efc44b1a5c529c2852f6532 : SUCCESS in 1m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f71196a39a8748a6bd1deb25b78376b5 : SUCCESS in 1m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/862e28153e8b47898cb9a1e3e9a3b0dc : SUCCESS in 54m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/346aba5d02c242599d4d03b4993adb93 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1fdefa5ab74ba48a3848610be2689a : SUCCESS in 2m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a031680ff964b48ad16d4269f8a9a94 : SUCCESS in 7m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/294cb9d2c2d54dee8eedd8043f674f0d : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa817b2ce43440a85d028a814ab12ae : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212142c51b284cb9a258986362003595 : SUCCESS in 5m 11s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/13da4a01d5e744fe8047ea524efb01ef : SUCCESS in 4m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/48ef452ab5bb4e8ebc8215b19afbce94 : FAILURE in 19m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1f8ae13cb5cf4ce69ce6645c76c25570 : SUCCESS in 1h 40m 15s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/31045ff73ccc44769cdefe87f333b768 : SUCCESS in 1h 45m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ed1446d3797a4b6784dab5ccae7fe56e : SUCCESS in 1h 02m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/58be5b3971734111bdd875127c262d07 : SUCCESS in 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6dc36dca4269485121eb7854f3d6c23c6f5800a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-21 11:03:04.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"81b1ec010083febbcb7373a72e954c1ad2b1f043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:51:52.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7c2cba18dc4f27a79267ed408a3485\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/586449dca8db49bbbe0c8248bdc23854 : SUCCESS in 3m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f045693dc49f4786a3db7c0cafdd1cd6 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2eb51dcefd61444f854ff51df2e4ba8f : SUCCESS in 1h 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/304ba35da7544771bded71eb7acdfc88 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154aea41edac4b6b8248c3e1ad59b123 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40e96604f460432ea9138ce7c31673d3 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cba200d495f147c6a27e66e2b20ae4dc : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfffcec4a86d4a3e997fff63e19513a5 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf9e26f5f7040539c98962ef08d618e : SUCCESS in 2m 52s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a84af84bda4844a3a4a8f90ec8935175 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f0e1223946d446f2831faadd77f46213 : FAILURE in 13m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/61fbc7264471499d81a775401d512fe4 : SUCCESS in 1h 36m 02s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/99640c6b16ca4eb49b45235b9e35331c : SUCCESS in 1h 38m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/54319606f53b422eb127c70b84c5b0ef : SUCCESS in 47m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc1a622f43b048839b3bda5de2601adf : SUCCESS in 58m 53s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"f284ae2cb7897c591dc66c2e324ad57bdd280ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-03 10:46:00.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"6d8c82dd824f01c3138d12cb7587d8053a20432f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 12:37:14.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29023a4b7b0499ebadb3b88114d9617\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7791aa408d3043e2841ebbef99e50515 : SUCCESS in 3m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e609b54e25524ec9b5fb8fb0740af281 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78cf9bf3f8754d1e8038d7620cc11f5f : SUCCESS in 1h 39m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efd5d08559be4d13ba8eed3eed68cabd : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4653f4f0834b94ab33c3fd9656fbd2 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/133ed2e39b904266baf683a4448c9f04 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6690fa4d68b74db6901fa7d216e98cd4 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a03a672e3028441db268022560395238 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3cc27d3e91f45b5af0c8f7df45f4ec6 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8503341126b24afe942ca717c9a465af : SUCCESS in 4m 04s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/d473f4041f354c88be53c43282ac219f : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e8546b519ae1451295dbe268a6014225 : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c2a849917230497e860d1f7d48413d1b : SUCCESS in 1h 21m 20s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8a2c61c16fb34901a9cf588ddd88bf20 : SUCCESS in 1h 39m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2c5de7ecfd39481abe0f11c187234dcc : SUCCESS in 59m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4dbe6156f1b541c793b67fe88f1fd76a : SUCCESS in 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"513207a5262541a7bf9fb036793ac972121882b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-30 10:14:34.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"6dce3c44683b37b993401da3bb5d4c7ded706560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:00:52.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dac6480fcd34af4b1a01ebcd852f3be\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d32ffc41ce94c4395bc4c70cd287a9e : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/30de0b21ba9b4417b3dd4a663b27b574 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d903fd509c8b4f52a91c2d69ec04bcd0 : SUCCESS in 1h 29m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0e0575ee4b744e2b1008234743b48b5 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d22868c215472c823f86ab429de3eb : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/167b63b64b594d4da56da9262815b2f8 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e889aed5faa4a578f862a4fe4f92a8b : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efe34e580446402d9e21ae8e54ae3d89 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a228cf757fa7432ca8c2d72293f02e93 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a98c3e1abc34f70a88e45360035c279 : SUCCESS in 2m 44s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e2b1f42974a94810b61ca818b97f728e : SUCCESS in 4m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/45c0517223754917952a522fdb17193e : FAILURE in 5m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a0bb0b59336840e7a0df977048e37c0b : SUCCESS in 1h 44m 42s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3fe895261ff74a1989d06c946c2df66a : SUCCESS in 1h 31m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3e40900c690e42ffa3a603888b95a510 : SUCCESS in 43m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e9c93fd8bee46f0881f5c32e4dd1d21 : SUCCESS in 59m 19s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"8d4e1ea8bc2a17ec143f4d475713baf4245b6d37","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-03 06:10:52.000000000","message":"Patch Set 28: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"1a6e7ab04b5620fb129b4ddadf4e06d5ed94a509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-07-29 09:10:49.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"7a4364e18e16ebd4c19728c61a6ce6299a938913","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-07-29 09:14:59.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"6bec2e3aea9112369c0f1b3431c98c4917deda93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 11:14:17.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605fb21ee54a4765bbf74a829e2366fd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58628d6e1cc147c9a93c9157db99ed96 : SUCCESS in 3m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c88564a0cf834da6ab0c9aa378284f5f : SUCCESS in 3m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6280ff1d42464e5e9bf783d49aadff91 : SUCCESS in 1h 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f80599dabbc4423bf5ef2bc6c814c86 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb24b618b7ed4e94bf13a0ba7fd3ab5c : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe931164cc604c94b5a255c301d39788 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0bc96210f5fa4654a1f43c29eb9da67e : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3763631b76e54f2295358d12c6154453 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7373aeef8ce42d4b3b5dfaf8b9fb2c6 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/092dda4317b6422d80e4f637741643fc : SUCCESS in 2m 48s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ac6729c0d9624aa5a9463185dae68b2c : SUCCESS in 5m 06s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/2ab73e7d98a3448f830df54a8e38a428 : FAILURE in 9m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ee72ab26e651480092b3e71f53c746a3 : SUCCESS in 1h 38m 42s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/87e76933476f48d69ce963a68a42dbd4 : SUCCESS in 1h 51m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/49568eb0ba0b4439adae7244ace5ebe3 : SUCCESS in 1h 00m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/12f753dc0e804e30b352189e42c00c7e : SUCCESS in 1h 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"8b8c431df6f0b1f6d5ac5eefa51bb4cca1a477c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 04:49:40.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"a1e6f9e32cc829bcf8297b1fab7069ee4c50c860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 06:08:08.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47a49f3da0d5411ab29fc38328deed0b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a9f2c7ccf59415da9448213a3ef9b4f : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdef52143d9c4167b6b4dbac35e0b96a : FAILURE in 11m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5cc33d38f854c05a9da418133e476c7 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a880da4da6664af5b05095d05c9cc412 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e23d99709964b91886908c10efb2c5e : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a53030c116c4edbaaba7b11d37caa79 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a2aa07660364ada8c28396c57847cf1 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e1afb3038f46e6918d98da60b30148 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02950e1976ab46b1b2131b6122a3aa76 : SUCCESS in 3m 25s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/761114e5a78e448c9443e670619c7ca2 : SUCCESS in 4m 22s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6d9cc9557ce74b13aeef38059ad6fb1f : FAILURE in 10m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bc068a89788044ea8dc2f727bbb94712 : FAILURE in 21m 47s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/429df1eb88db482281c84dc11ca5f2ed : FAILURE in 24m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9be1acc136274b77bca195eb8f27c940 : SUCCESS in 56m 03s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1282b71cd6be4e2b9766c19e11ff181a : SUCCESS in 1h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"dbc5fb16be742bd95d42a300d5a55ac1f20031bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 09:23:14.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"1dd030265fe808bdd8b0369213b2a6d1897dea1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:06:41.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64aff4d3ea154d3cbf4dd51e1bf7000f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ceb8b00b90c46c9a7b403ff0e41cb60 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d1b5039f7864922a90738187773a6ef : SUCCESS in 1h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a8bfa92041348c28df5880ce13898f9 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbe60e154b54269815baa67af387da5 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26c530188a794397ab3f2cf20f4eb407 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de378e51157745e882812532dc1fe7a7 : SUCCESS in 6m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85466191ebef456da4d9bbf23d40f8d2 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43313139a464437faa19b7f7bda92a6f : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/155a1b160fb9443ca61ede55a2acaed1 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/4c1f850da8e546538e51eafc009bd1f0 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/971fd4b062124526819ce77db088866a : FAILURE in 12m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5c63a6d7d7fc45f8b5fe774adb3312fc : SUCCESS in 1h 35m 18s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/73cc2290603a43cd9b6f6feaae6ced58 : SUCCESS in 1h 31m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6ae807608c6a4f4ebd8dfe1694863997 : SUCCESS in 54m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c31f81234a60432ca4a8c767d29ad264 : SUCCESS in 1h 06m 46s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"4b560503b98494cac2e7773c7e2f3f21bd17d8f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 10:13:28.000000000","message":"Patch Set 31: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":31},{"id":"fb326dc196936e7243ececffc871eebdc520d0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:39:20.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"f83f4969eab400e66c25719252781f9a6f3dfcf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:50:49.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"e846757386e299912481ebf49172f7da56a93df6","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:56:09.000000000","message":"Patch Set 33:\n\n(5 comments)","accounts_in_message":[],"_revision_number":33},{"id":"5de0e1783c3ebab226c5adc3d6b524201522af19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-26 10:37:24.000000000","message":"Patch Set 33: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa090512db894b7da994fa46955d8a64\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1c198d10412438fa63a32a3ee805d90 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/967dac8c423f432e89276cc9741d4b65 : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3759cc0dd4a74fdc969d2b6356969ead : SUCCESS in 1h 45m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/314582954f1542f1a80723a82afd9e5e : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c86206a2909449ae8acf1e03222c1513 : FAILURE in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c44b53ed930e4b7196d059723c216723 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6527cc3d09914136a1c35ef7681a1e88 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb22ecce17fe4883baa135e4c061e2fd : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca63dd6f06e04a61ab8be5064debf69e : SUCCESS in 3m 35s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0e18dfae1ad14eba9e0d9ad01ad6f93c : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8587af96fa974176aa31572aff96d261 : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5bd27e2a51df488fa60f92b434495e07 : SUCCESS in 1h 15m 06s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/09bf788f7e3442ab9a156bf4772abeec : SUCCESS in 1h 17m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f19032415b3e4c099b2f6e57274e83d1 : SUCCESS in 58m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a31fbe20491742fb8b95e9c084ad074b : SUCCESS in 1h 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"ef1189d9638a4af0f2a04d9a5a2d93b613cb3fcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-05 09:09:45.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"5eb6dfb0c0e8a98301010e7bfc1fbed3f854129a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 10:58:03.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddf8280b1b9146b6aed6da56e89d6cb2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f0008634fa8b4fd89e8055c16220090e : SUCCESS in 1m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2deb825051544370ab92da64a71ac509 : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34f6fc8b4ca248a89f5bbfef726ea2cd : SUCCESS in 1h 37m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6fde425797415bba6c79b4d34e534d : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9175205ff94a7f9a4363cbde4b636f : FAILURE in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af76699b7df5488ab664e2fd71e52692 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d6a0dc68d7d42f3b79d163ea0a0a6cc : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f7b5b7a09249fead3fda9c605dbe50 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87054f391e1f42b78fc530fd080d2035 : SUCCESS in 1m 59s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e4f2edeca75d439daa4ddae4f9fe578b : SUCCESS in 4m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0c27ab8722684ae081c2a188e328b9ee : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/483e819d873c4e9290e8e2fc34141469 : SUCCESS in 1h 46m 34s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dfb5739363ca445d96f761242c7e63d3 : SUCCESS in 1h 27m 33s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4567e69c1d8f4f6bb6b1c07f84de64a5 : SUCCESS in 30m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c636da814f384327b266e9a1cc1b9e04 : SUCCESS in 57m 14s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"5a4902a64cc7374c6972f8a2f700baa95fb7d5bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-05 14:31:48.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"d96a6b82663c39a330f5c21a315b278b29848a07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 08:30:42.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"25342430e0ec25b3c66b0693c7a26ac2b4f1f687","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 08:50:06.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"ce80b07e213330e1f25d8087ca6ad99f06e4045d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 10:37:16.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef12696d3944439a2cf6b2060d54c9e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7a00fbcd1304ee08b3191e82286994e : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/155c0b9b6ca545bfa220895713f0b8ca : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdc0d487fda34888a39f816c0937604f : SUCCESS in 1h 57m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6476a3c07c4090b61b38e50a6a963b : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f478e4446d014283b7342bc1672eebba : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e75764103e7e4df093ef72131c0546cd : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fdda273d9834bdfa4e2b2221cde8cab : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927cceeea75246a184b58f86955be078 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79fb72c2bba748ea97afdef075ede147 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/181e3a44534240abadca695bbb2ab4ee : SUCCESS in 5m 47s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5dc1d768a2304f56a88f4c1412489158 : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e658e6d5664a47428eb1410adb5e55e2 : SUCCESS in 1h 40m 29s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b83996a0e1be40d7975a94519c7b42e6 : SUCCESS in 1h 44m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/02663444d1c64b0a9caf2c25056e57da : SUCCESS in 34m 27s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9abfe8f23436482b9372840576919215 : SUCCESS in 1h 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"01ccc33a2cda88bff979e9e58131720dc2dbb30a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-15 09:35:57.000000000","message":"Uploaded patch set 36: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"9458ed422f162095b0cc0d7f0092859305178d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 11:33:59.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ee0ae2f8e884ed7a0a99302599b9d7b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/723f011f13c241b3883f0636051d5ab9 : SUCCESS in 1m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d1aa4d863e344e0b0ebf080fcd3fc47 : SUCCESS in 1m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6050b8bafe6843aba4107f9862bf6cdc : SUCCESS in 1h 52m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaac231788cf443d8c97653013d8d9f5 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae9d5b5cd1b40f58abcf1207f7a5e11 : SUCCESS in 1m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1eeb7c57e3e45b7a4d1ff75e544bf81 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebade920c7304007a9c25482501aefb5 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747d784e01134a5f903b7463d2f54983 : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9456495bce94fa3b73516c2d8617c98 : SUCCESS in 3m 29s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bab1b88d6bf24096b374e122d9946fc5 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c8ef9e014c5248c1a6c8ede2ee08e399 : FAILURE in 6m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e3ca543f8c4a475589f58434542675c5 : SUCCESS in 1h 47m 00s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/403f25cc292a4c40993f1ff2817dadf5 : SUCCESS in 1h 41m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/61c6c6ccb29c4ed690298ee6e5d20fe3 : SUCCESS in 1h 23m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d554bf98f8c94b51a2bf8b497633fff3 : SUCCESS in 1h 09m 34s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"a960bfd3d3d0d576b3011a93ae378f5b9eb0ffcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-19 14:48:53.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"51196a5ce7483766a58f75af8a292b7b3b4c406e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-19 15:30:18.000000000","message":"Patch Set 37:\n\n(2 comments)","accounts_in_message":[],"_revision_number":37},{"id":"bfb30c5d9893a623091555e7872a1f8af3b3d78a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 16:51:58.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0335bd04b72a45dfba2992dc42b82f26\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10efaf1ace0848dfacab095332c3262e : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b51133f2ba054d12979c2f4c09029b96 : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e426d59bd404558b7ec3c99b4922179 : SUCCESS in 1h 43m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca9ad6001d3d40afac062bf3824f4fe5 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b94f1d69e994e579dc9434a96a4305a : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/816c633f53664a53ae780c79d0a53c4a : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5109c755b6624747910850ec760124c6 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2aaf0aacf146a58ec7d1342b719577 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1107b23592024b30a843ccc957b8edd5 : SUCCESS in 3m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/84bb388eebc447b2a7757243a550b93a : SUCCESS in 4m 55s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d614bdecee3a47e3a244e58dc8fcdbb0 : FAILURE in 5m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/23dfcb537e314d22a03168c44d8c3ed4 : SUCCESS in 1h 29m 20s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7275ccb76289428fabf268da1185b4d5 : SUCCESS in 1h 50m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0b12cece79de4251a45b7d7aa991aa24 : SUCCESS in 1h 08m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a60d7092f234ab18142761a87bc991a : SUCCESS in 1h 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0360df6b4e60b646dc4e362e6139bd416b5d540d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 07:45:37.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"860463ba8e7a981d7feb290eca1488d4540daa42","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 07:53:40.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"4a3da821c2ff842e64aecdbbccbf8ba544e5a3d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:28:50.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/202a30224a6d4014a1147d8281731d41\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/596787260cad43d7875ec6c5005cff91 : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9f8525638944c29a9838c917152720e : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/401e600ae5de4c83bb61aeca6a4e11d0 : SUCCESS in 1h 32m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3eec0474fd24f579824522e079646ee : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca85972123648fb9095238f5c294082 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf62fbff9cbb49e287bdbd5bddc8d4d6 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca67b3dc7e9c4624ab2551b682f31d8c : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0702425c5b614b14be21be55f18f1c1f : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33cbb00acccd4d6e95a318bc96791134 : SUCCESS in 3m 05s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9ff6f128a31745c2a7755e0d3098f78a : SUCCESS in 4m 43s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e03f8d08c7c54048a791d56f3bca7ddb : FAILURE in 5m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ede912136d20404085aeea6c04f1b74a : FAILURE in 1h 21m 22s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/44b7401271cb49ce8746afcf46ad4385 : SUCCESS in 1h 26m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8e1bf99052c24d3a8b336ef2f465170a : SUCCESS in 1h 00m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1539e255d234834829fac21c54ebf6d : SUCCESS in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"d879bf50bde16189a97a8642e330e52b537454bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 11:23:17.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"b97061d7095d4c811475b9b077791140164f4248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:18:06.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c46460fb4882483bb63bcd9a2826705b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b968944176834bb58d8d0db5754bf989 : SUCCESS in 2m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/705de841f6204c7a80db629323f29f2b : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89af3112f7174bac8a26af14122f6f87 : SUCCESS in 1h 45m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8886b7b140604b6099e05b3f8b7c7f0d : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd1251c1df54530bed18738f678ac8b : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57c2013c13ee44f89ca83d73a2e28c29 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79bf2e6b98bd4829a80289e5a02345f6 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96228b5820fe46339650e9923a5866d5 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/843eddc879f74699abc7160e88b3b6d7 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/dee6df69ea614b038e816927459d7b7a : SUCCESS in 4m 34s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e73596d8e34a48159178b6f144bd0441 : FAILURE in 5m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/67891cc050464d7fae952d05d661d5f5 : SUCCESS in 1h 45m 58s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/30806e2561f441ad848bd59b5deb7cc5 : SUCCESS in 1h 49m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/1a4180126d854e68a402959dc44119a2 : SUCCESS in 46m 03s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1e51970d5a5499b848b13da8dc7372a : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"3e87e530e25f48569a282f5a6871d3b70beba520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-26 08:24:40.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"2dfaf09ba2a880594c73a3af63f3811eadddb8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 10:21:06.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eb733fc8bff4e81bca51740a543a5f1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4934cc42b7744f8ba0d8efefacb6dbe2 : SUCCESS in 3m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/600417ebfebb4fb5b85cc5cadb047c0a : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7763cdc713142508d56465f658ff8cd : SUCCESS in 1h 13m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdbbf7b3c227432c9f2afcc4044aaaa9 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc67240156f94727bebb8b939589b5bf : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c4b6e9695d546a3a7760adaf2980c0f : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af6a320c90bc4693b4f1cd7b46a50f8a : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2377c58b6ffd4725956f920526e8e744 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae47df651014a44aa01778cac06e6ae : SUCCESS in 3m 00s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ae640b32ac264793b94062a41c469083 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a95315d9c16242fdb693989ed8c3110d : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/db679d8b96e94bccb4da191977a572c0 : SUCCESS in 1h 51m 00s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/aff0e5798a4c477caa5e80aa4ecc7e7c : SUCCESS in 1h 46m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/765d4df9edee482da5f085246158bf2d : SUCCESS in 46m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/333d5aebd48145ddbf44c89009570f20 : SUCCESS in 1h 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"afb71e176461e11d6ec5f92a43d9b2df50b14c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"date":"2026-06-15 01:10:43.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"e2a7db8915b8862b4e5293a98bd316c5618bd0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 03:06:11.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7bca153ee94c7db3a34ce3af52ed49\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b219ee6d61e4434faa821df2741c062e : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85e60c1963b94f6f83bef8e2a4200632 : SUCCESS in 1m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ceb93488409b462ab3b4acb470885c4b : SUCCESS in 1h 39m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c490eb4383d4e5daab62a77e54dc8a6 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71502c15f334071998c8b76b848bbcf : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5757c6102c4448d8ff3bd15e821882b : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/356270bf049e4e14a0d9f87fd172b7d9 : SUCCESS in 3m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/853468720e7e4f0f84dee89773d54f58 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69eefa644b7a459fae17dcb8f4038bd1 : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d955b05084145d58969f87208862069 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/a885e87ccd4f4963a914d1a4ff813f79 : SUCCESS in 4m 35s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/13a953509c1742ac96c2dba4b2639f51 : FAILURE in 6m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c4437d3516454078802ea5f290e33d68 : SUCCESS in 1h 49m 01s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/994721877dc84de19bef716c68bcbace : SUCCESS in 1h 13m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/72909beaedb44f0e8c57695b99d32892 : SUCCESS in 1h 10m 02s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ab4bfb98b424058abd2c2178770fe90 : SUCCESS in 1h 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"0e5480f6aa78f272fe6b01ad83b3538036110dd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 21:34:29.000000000","message":"Patch Set 42: Patch Set 41 was rebased on behalf of \u003cGERRIT_ACCOUNT_33451\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"_revision_number":42},{"id":"d006d1e8faa71582475e26ff2e0443694a40379d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 23:16:41.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7080e1ec233241b6ab52e536e570179c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fb60af88bd24a0f981c261352940064 : SUCCESS in 2m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e363efab91474516b44825dee60797f8 : SUCCESS in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27f87e0b38cc4069bdb0cbfa51101468 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd611dd3fa04c8798a33de57330a107 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aaed23884a644a9b637d62d7b506144 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b257d32e77804f549781d6f8119a14ec : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ffe31b593804e16b23ea316dd233351 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3529859270594d8cb16520a1e3eeec79 : SUCCESS in 4m 25s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/098b45a2002a427ab8879ad4e4efa244 : FAILURE in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc14c3ba246541948348c2d768c87e95 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5096427714974fccb894952212d65b2d : SUCCESS in 4m 36s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/6b1f73fabdd741cb91202708d0938ab5 : SUCCESS in 4m 00s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/4d9b11c41a524fb2b40e1f19b8994e02 : FAILURE in 3m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bceaa93876dc4030a25e491ba9719755 : SUCCESS in 1h 36m 56s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7e827c4284604028814ffb290c474954 : SUCCESS in 1h 36m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9000dad1f46e4ac3a53ae595f5fa1770 : SUCCESS in 1h 04m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7acf0698bb94ce1bac81b850b0c7fb3 : SUCCESS in 39m 18s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"029aac4b4f65ea288c378c6a2236ab7930b604a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-17 05:34:21.000000000","message":"Patch Set 43: Patch Set 42 was rebased\n\nThe following files contain Git conflicts:\n* pyproject.toml\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":43},{"id":"47c68d124efb3dc0b49ebad49820386187fbd6ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-17 05:35:43.000000000","message":"Patch Set 44: Published edit on patch set 43.","accounts_in_message":[],"_revision_number":44},{"id":"ae112bdd807120346e9e4dbe3ec84bb7939fc418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 07:15:13.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650f43bc4dee4edfb4705a889d429533\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58a034977a8b47efa9638a51fc536546 : SUCCESS in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e3331b7f38f45ae835ee4ec502b5c8d : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50991265c3e54479a9fdd15f75ae4899 : SUCCESS in 1h 30m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfbe129758a74483b2592724b7d1745d : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3ea192d0c8480aa3c709c0656dd808 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35f6ab127b0f4ad9b75e3156b4892651 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8418bb7987a144fea3906f856e22e609 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2592f3b9fcf04c2ba99c8c66439f9255 : SUCCESS in 6m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a8c8b4c3ab964d82aee3ae464119c26e : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6f006fb95b43b8a8dc2f95ee53f9c1 : SUCCESS in 2m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a531a3017044dea7376fb923f96065 : SUCCESS in 2m 39s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/3f827b11b50645e7aebec5a520af5968 : SUCCESS in 4m 42s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/e7d08d48285e4c1ab9b2b12ca2f03bf6 : FAILURE in 2m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d255fca2370e48b5ac94eaf80a0b3528 : SUCCESS in 1h 26m 51s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9edb2e821d1c49289a5af014993c9da0 : SUCCESS in 1h 35m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ecd3d8c585ea42c186b37d47b1c3499c : FAILURE in 36m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/997f66fbee15490dbff2b5509f59cec6 : SUCCESS in 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"54ecf8a5d0e77b4741203ba64fad0b3c8decc7a0","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-09-10 23:33:40.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"6862df61a9d5619379864939e75f0886af986cb5","revisions":{"d369c78e774ed13e18413ca0e99345cb9e9ff5bb":{"kind":"REWORK","_number":1,"created":"2024-03-12 01:09:54.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/1"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-12 01:09:50.000000000","tz":540},"subject":"Implement ConsulOperator as a server broker to support HTTP Driver","message":"Implement ConsulOperator as a server broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker for the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for RPC communication.\nIt communicates RPC client and server directly over HTTP without a messaging queue.\n\nImplements: https://opendev.org/openstack/oslo-specs/src/branch/master/specs/bobcat/http-driver.rst\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d369c78e774ed13e18413ca0e99345cb9e9ff5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d369c78e774ed13e18413ca0e99345cb9e9ff5bb"}]},"branch":"refs/heads/master"},"763ceeaca5c9060afbca88e3c35f8a5501a562ff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-13 05:33:36.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/2"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-13 05:31:39.000000000","tz":540},"subject":"Implement ConsulOperator as a server broker to support HTTP Driver","message":"Implement ConsulOperator as a server broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/763ceeaca5c9060afbca88e3c35f8a5501a562ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/763ceeaca5c9060afbca88e3c35f8a5501a562ff"}]},"branch":"refs/heads/master"},"6dcf3a9d0cdd8e6c6e9c1797137373ea2e13a00d":{"kind":"REWORK","_number":3,"created":"2024-03-15 05:47:29.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/3"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-15 05:47:16.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6dcf3a9d0cdd8e6c6e9c1797137373ea2e13a00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6dcf3a9d0cdd8e6c6e9c1797137373ea2e13a00d"}]},"branch":"refs/heads/master"},"f878bae9d54f121dc40ffad4c17fe203bb6ace76":{"kind":"REWORK","_number":4,"created":"2024-03-15 08:11:41.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/4"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-15 08:11:16.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f878bae9d54f121dc40ffad4c17fe203bb6ace76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f878bae9d54f121dc40ffad4c17fe203bb6ace76"}]},"branch":"refs/heads/master"},"d1af2d09183db845a5fbbbeacc9f55199d4e2088":{"kind":"REWORK","_number":5,"created":"2024-03-22 08:11:58.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/5"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-22 08:10:23.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d1af2d09183db845a5fbbbeacc9f55199d4e2088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d1af2d09183db845a5fbbbeacc9f55199d4e2088"}]},"branch":"refs/heads/master"},"c13bedb9f5b6724afcf3f7e4a6bcd646244283a6":{"kind":"REWORK","_number":6,"created":"2024-03-22 08:56:16.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/6"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-22 08:56:03.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c13bedb9f5b6724afcf3f7e4a6bcd646244283a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c13bedb9f5b6724afcf3f7e4a6bcd646244283a6"}]},"branch":"refs/heads/master"},"2092c76255aac57d89f2248b78bbfa9636cdd54d":{"kind":"REWORK","_number":7,"created":"2024-03-26 01:08:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/7"}}},"commit":{"parents":[{"commit":"ebdc7db19eafec27b1a231e1ccfb19d141e17a93","subject":"Merge \"Display the reply queue\u0027s name in timeout logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ebdc7db19eafec27b1a231e1ccfb19d141e17a93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-03-26 01:06:24.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2092c76255aac57d89f2248b78bbfa9636cdd54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2092c76255aac57d89f2248b78bbfa9636cdd54d"}]},"branch":"refs/heads/master"},"7d0660b64cb1fdee33ac7168369837fd22bc58f1":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-04-02 03:05:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/8"}}},"commit":{"parents":[{"commit":"986cd4ab240ddeb5022513f6eccc4cb48e8d3de0","subject":"Merge \"Fix incorrect desc of rabbit_stream_fanout option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/986cd4ab240ddeb5022513f6eccc4cb48e8d3de0"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 03:02:30.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7d0660b64cb1fdee33ac7168369837fd22bc58f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7d0660b64cb1fdee33ac7168369837fd22bc58f1"}]},"branch":"refs/heads/master"},"d06ac93054d0d850e544f9f600915e26b072227d":{"kind":"REWORK","_number":9,"created":"2024-04-03 11:17:48.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/9"}}},"commit":{"parents":[{"commit":"986cd4ab240ddeb5022513f6eccc4cb48e8d3de0","subject":"Merge \"Fix incorrect desc of rabbit_stream_fanout option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/986cd4ab240ddeb5022513f6eccc4cb48e8d3de0"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-03 11:11:06.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d06ac93054d0d850e544f9f600915e26b072227d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d06ac93054d0d850e544f9f600915e26b072227d"}]},"branch":"refs/heads/master"},"4d224d6cdc42d2a716d56be570fd2db0a64a57de":{"kind":"REWORK","_number":10,"created":"2024-11-11 04:26:20.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/10"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 04:25:45.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d224d6cdc42d2a716d56be570fd2db0a64a57de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d224d6cdc42d2a716d56be570fd2db0a64a57de"}]},"branch":"refs/heads/master"},"4e9e4ab8a347dd73173385b74e208244f9fc8921":{"kind":"REWORK","_number":11,"created":"2024-11-11 04:31:46.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/11"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 04:31:10.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e9e4ab8a347dd73173385b74e208244f9fc8921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e9e4ab8a347dd73173385b74e208244f9fc8921"}]},"branch":"refs/heads/master"},"926b9dd11d89e8659427afa551857907b54c0aea":{"kind":"REWORK","_number":12,"created":"2024-11-11 05:50:44.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/12"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 05:50:05.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/926b9dd11d89e8659427afa551857907b54c0aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/926b9dd11d89e8659427afa551857907b54c0aea"}]},"branch":"refs/heads/master"},"fe64ba0bd5833fd859292673779ac90c09692d81":{"kind":"REWORK","_number":13,"created":"2024-11-11 09:10:06.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/13"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 09:09:19.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe64ba0bd5833fd859292673779ac90c09692d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe64ba0bd5833fd859292673779ac90c09692d81"}]},"branch":"refs/heads/master"},"b453991c852d8816f18668883c57bd08fdcd2cf5":{"kind":"NO_CHANGE","_number":14,"created":"2024-11-12 09:01:54.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/14"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-12 09:01:28.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b453991c852d8816f18668883c57bd08fdcd2cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b453991c852d8816f18668883c57bd08fdcd2cf5"}]},"branch":"refs/heads/master"},"be13b041020e6d75c409b8421fc144fa0e867ca7":{"kind":"REWORK","_number":15,"created":"2024-11-13 03:06:03.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/15"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-13 03:04:42.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/be13b041020e6d75c409b8421fc144fa0e867ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/be13b041020e6d75c409b8421fc144fa0e867ca7"}]},"branch":"refs/heads/master"},"8480a283131843bddc07489c06b16f4d5cb565c0":{"kind":"REWORK","_number":16,"created":"2024-11-21 05:07:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/16"}}},"commit":{"parents":[{"commit":"7997a199acada382b0c61dc84aa956e4c65de986","subject":"Remove deprecated AMQP1 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7997a199acada382b0c61dc84aa956e4c65de986"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-21 05:03:48.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8480a283131843bddc07489c06b16f4d5cb565c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8480a283131843bddc07489c06b16f4d5cb565c0"}]},"branch":"refs/heads/master"},"a31647e5ef2c1425b7b5b590037c36546650b90a":{"kind":"REWORK","_number":17,"created":"2024-11-25 07:12:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/17"}}},"commit":{"parents":[{"commit":"c39c0bca1c15544d663215c03f61c202afe4ca4a","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c39c0bca1c15544d663215c03f61c202afe4ca4a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-25 07:11:49.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a31647e5ef2c1425b7b5b590037c36546650b90a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a31647e5ef2c1425b7b5b590037c36546650b90a"}]},"branch":"refs/heads/master"},"43a6bbf778852590152bea5a1404da457cc6e2d4":{"kind":"REWORK","_number":18,"created":"2024-12-02 08:20:21.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/18"}}},"commit":{"parents":[{"commit":"c39c0bca1c15544d663215c03f61c202afe4ca4a","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c39c0bca1c15544d663215c03f61c202afe4ca4a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-02 08:06:15.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43a6bbf778852590152bea5a1404da457cc6e2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43a6bbf778852590152bea5a1404da457cc6e2d4"}]},"branch":"refs/heads/master"},"6ad27d975a3ef5a666c6581751a4cf7fe62d6ab5":{"kind":"REWORK","_number":19,"created":"2024-12-09 06:11:57.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/19"}}},"commit":{"parents":[{"commit":"c39c0bca1c15544d663215c03f61c202afe4ca4a","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c39c0bca1c15544d663215c03f61c202afe4ca4a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-02 10:12:16.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6ad27d975a3ef5a666c6581751a4cf7fe62d6ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6ad27d975a3ef5a666c6581751a4cf7fe62d6ab5"}]},"branch":"refs/heads/master"},"3128dde9e5a4e17b79755056e6bc926f19d530b6":{"kind":"REWORK","_number":20,"created":"2024-12-19 05:58:24.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/20"}}},"commit":{"parents":[{"commit":"9e34e238101a6cefdbc54722d73f6627f586f7da","subject":"Merge \"kafka: Remove unused connection pool options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9e34e238101a6cefdbc54722d73f6627f586f7da"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-19 05:57:02.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3128dde9e5a4e17b79755056e6bc926f19d530b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3128dde9e5a4e17b79755056e6bc926f19d530b6"}]},"branch":"refs/heads/master"},"2e7bab65c4bc16288cff80efcba0e4493156a540":{"kind":"REWORK","_number":21,"created":"2024-12-27 11:06:22.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/21"}}},"commit":{"parents":[{"commit":"60f7bc2012ccd47dd671c33deaa476fdc9f68873","subject":"Merge \"Deprecate the Eventlet executor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60f7bc2012ccd47dd671c33deaa476fdc9f68873"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-27 11:06:02.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e7bab65c4bc16288cff80efcba0e4493156a540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e7bab65c4bc16288cff80efcba0e4493156a540"}]},"branch":"refs/heads/master"},"264366c4b218299cdfaa442368e3c82264b3a1b7":{"kind":"REWORK","_number":22,"created":"2024-12-27 14:53:30.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/22"}}},"commit":{"parents":[{"commit":"60f7bc2012ccd47dd671c33deaa476fdc9f68873","subject":"Merge \"Deprecate the Eventlet executor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60f7bc2012ccd47dd671c33deaa476fdc9f68873"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-27 14:52:48.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/264366c4b218299cdfaa442368e3c82264b3a1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/264366c4b218299cdfaa442368e3c82264b3a1b7"}]},"branch":"refs/heads/master"},"6478d11fb0a9b24ba156f6aac2f90b7503c7eb11":{"kind":"REWORK","_number":23,"created":"2024-12-28 06:59:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/23"}}},"commit":{"parents":[{"commit":"60f7bc2012ccd47dd671c33deaa476fdc9f68873","subject":"Merge \"Deprecate the Eventlet executor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60f7bc2012ccd47dd671c33deaa476fdc9f68873"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-28 06:59:39.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6478d11fb0a9b24ba156f6aac2f90b7503c7eb11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6478d11fb0a9b24ba156f6aac2f90b7503c7eb11"}]},"branch":"refs/heads/master"},"ad86ff9dad926c7a22775d7b5db578346a59845e":{"kind":"NO_CHANGE","_number":24,"created":"2025-01-07 07:47:38.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/24"}}},"commit":{"parents":[{"commit":"60f7bc2012ccd47dd671c33deaa476fdc9f68873","subject":"Merge \"Deprecate the Eventlet executor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60f7bc2012ccd47dd671c33deaa476fdc9f68873"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-01-07 07:47:22.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad86ff9dad926c7a22775d7b5db578346a59845e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad86ff9dad926c7a22775d7b5db578346a59845e"}]},"branch":"refs/heads/master"},"d19e8d64de45aaddc5b62cdfe0aa0665f23eac3b":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-03-19 08:34:25.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/25"}}},"commit":{"parents":[{"commit":"3a0b6cb9813614077c6712d10df57190a98583b6","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3a0b6cb9813614077c6712d10df57190a98583b6"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 08:33:18.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d19e8d64de45aaddc5b62cdfe0aa0665f23eac3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d19e8d64de45aaddc5b62cdfe0aa0665f23eac3b"}]},"branch":"refs/heads/master"},"5359e6030822b0dd8e2110ce3c294be2ebec9f5a":{"kind":"REWORK","_number":26,"created":"2025-03-21 11:03:04.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/26"}}},"commit":{"parents":[{"commit":"3a0b6cb9813614077c6712d10df57190a98583b6","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3a0b6cb9813614077c6712d10df57190a98583b6"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-21 10:50:24.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5359e6030822b0dd8e2110ce3c294be2ebec9f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5359e6030822b0dd8e2110ce3c294be2ebec9f5a"}]},"branch":"refs/heads/master"},"e3b3ad8919faa2b63712eebae93080fcd89815e4":{"kind":"REWORK","_number":27,"created":"2025-04-03 10:46:00.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/27"}}},"commit":{"parents":[{"commit":"1d67be04bcc440897ad5487b31781a692b355644","subject":"Merge \"Drop workaround for old oslotest version with six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1d67be04bcc440897ad5487b31781a692b355644"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-03 10:44:12.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3b3ad8919faa2b63712eebae93080fcd89815e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3b3ad8919faa2b63712eebae93080fcd89815e4"}]},"branch":"refs/heads/master"},"40ec233e64f69b1354b4b9b31401f461b365a2ef":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-04-30 10:14:34.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/28"}}},"commit":{"parents":[{"commit":"c058e79bafb1976a2e40a1a6dbd05036acf1a33c","subject":"Merge \"Do not un-patch threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c058e79bafb1976a2e40a1a6dbd05036acf1a33c"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-30 10:11:42.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40ec233e64f69b1354b4b9b31401f461b365a2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40ec233e64f69b1354b4b9b31401f461b365a2ef"}]},"branch":"refs/heads/master"},"ad98e52609d4817e45e21daf20ba2adc7209769e":{"kind":"REWORK","_number":29,"created":"2025-07-29 09:10:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/29"}}},"commit":{"parents":[{"commit":"c058e79bafb1976a2e40a1a6dbd05036acf1a33c","subject":"Merge \"Do not un-patch threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c058e79bafb1976a2e40a1a6dbd05036acf1a33c"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-07-29 09:09:10.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad98e52609d4817e45e21daf20ba2adc7209769e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad98e52609d4817e45e21daf20ba2adc7209769e"}]},"branch":"refs/heads/master"},"28e3dc1dada5a0b0763744388fb3e913c54a94e6":{"kind":"REWORK","_number":30,"created":"2025-09-18 04:49:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/30"}}},"commit":{"parents":[{"commit":"4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 04:32:31.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/28e3dc1dada5a0b0763744388fb3e913c54a94e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/28e3dc1dada5a0b0763744388fb3e913c54a94e6"}]},"branch":"refs/heads/master"},"44db5e7a62a8621d7c0da0fb265ed8873114dd37":{"kind":"REWORK","_number":31,"created":"2025-09-18 09:23:14.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/31"}}},"commit":{"parents":[{"commit":"4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 08:38:39.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44db5e7a62a8621d7c0da0fb265ed8873114dd37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44db5e7a62a8621d7c0da0fb265ed8873114dd37"}]},"branch":"refs/heads/master"},"a573b63b1d9ec23602ef30576eb0ee37e798c600":{"kind":"REWORK","_number":32,"created":"2025-12-26 08:39:20.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/32"}}},"commit":{"parents":[{"commit":"4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:25:34.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a573b63b1d9ec23602ef30576eb0ee37e798c600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a573b63b1d9ec23602ef30576eb0ee37e798c600"}]},"branch":"refs/heads/master"},"d2204c5b79222858cdf20da850e8ea19291cadf7":{"kind":"REWORK","_number":33,"created":"2025-12-26 08:50:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/33"}}},"commit":{"parents":[{"commit":"4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:50:13.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d2204c5b79222858cdf20da850e8ea19291cadf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d2204c5b79222858cdf20da850e8ea19291cadf7"}]},"branch":"refs/heads/master"},"2c0d4c4aebb2a2538441d606c9eb43b6927a3f39":{"kind":"REWORK","_number":34,"created":"2026-01-05 09:09:45.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/34"}}},"commit":{"parents":[{"commit":"4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4b1941221f1f51a0432cb96bc0b2e8203ac2fcb8"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-05 08:22:56.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2c0d4c4aebb2a2538441d606c9eb43b6927a3f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2c0d4c4aebb2a2538441d606c9eb43b6927a3f39"}]},"branch":"refs/heads/master"},"ed0dee904916fd0e6356384018a6d0a76c8a229a":{"kind":"REWORK","_number":35,"created":"2026-01-06 08:30:42.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/35"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-06 08:09:43.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ed0dee904916fd0e6356384018a6d0a76c8a229a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ed0dee904916fd0e6356384018a6d0a76c8a229a"}]},"branch":"refs/heads/master"},"c1949b0a73411765eddf5c050968519d3c9ce588":{"kind":"NO_CODE_CHANGE","_number":36,"created":"2026-01-15 09:35:57.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/36"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-15 09:31:36.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/5)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1949b0a73411765eddf5c050968519d3c9ce588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1949b0a73411765eddf5c050968519d3c9ce588"}]},"branch":"refs/heads/master"},"3ed02c36cfa5a5fe4db165f1f26dab25cefce924":{"kind":"REWORK","_number":37,"created":"2026-01-19 14:48:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/37"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-19 14:39:42.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ed02c36cfa5a5fe4db165f1f26dab25cefce924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ed02c36cfa5a5fe4db165f1f26dab25cefce924"}]},"branch":"refs/heads/master"},"b458ee1b1bc9d3f114873c431a5f22ba3f854740":{"kind":"REWORK","_number":38,"created":"2026-01-20 07:45:37.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/38"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 07:43:19.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b458ee1b1bc9d3f114873c431a5f22ba3f854740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b458ee1b1bc9d3f114873c431a5f22ba3f854740"}]},"branch":"refs/heads/master"},"024b618ae0fec4d80a7e4e1dc55e8672732d95c4":{"kind":"REWORK","_number":39,"created":"2026-01-20 11:23:17.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/39"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 10:42:17.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/024b618ae0fec4d80a7e4e1dc55e8672732d95c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/024b618ae0fec4d80a7e4e1dc55e8672732d95c4"}]},"branch":"refs/heads/master"},"43263a15f722fe01fbc2b941bba61fe684ac19c7":{"kind":"REWORK","_number":40,"created":"2026-01-26 08:24:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/99/912499/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/40"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-26 06:46:23.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43263a15f722fe01fbc2b941bba61fe684ac19c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43263a15f722fe01fbc2b941bba61fe684ac19c7"}]},"branch":"refs/heads/master"},"09c8f167ca737eb6c0ff9a65af5274a32dae1b3a":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2026-06-15 01:10:43.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"ref":"refs/changes/99/912499/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/41"}}},"commit":{"parents":[{"commit":"b595618bcf94676066c5e50e80c095d3a1c75c01","subject":"Remove Python 3.10 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b595618bcf94676066c5e50e80c095d3a1c75c01"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-06-15 01:09:35.000000000","tz":540},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09c8f167ca737eb6c0ff9a65af5274a32dae1b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09c8f167ca737eb6c0ff9a65af5274a32dae1b3a"}]},"branch":"refs/heads/master"},"3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2026-07-11 21:34:29.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/99/912499/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/42"}}},"commit":{"parents":[{"commit":"8ca4a4bfed0af710c90af95a7821471f898b3d93","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ca4a4bfed0af710c90af95a7821471f898b3d93"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-07-11 21:34:29.000000000","tz":0},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b595618bcf94676066c5e50e80c095d3a1c75c01","ours":"09c8f167ca737eb6c0ff9a65af5274a32dae1b3a","theirs":"8ca4a4bfed0af710c90af95a7821471f898b3d93","merge_strategy":"recursive","contains_conflicts":false}},"1c1afd11250aa57d60f485ea035dbb59776cad54":{"kind":"REWORK","_number":43,"created":"2026-07-17 05:34:21.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/99/912499/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/43"}}},"commit":{"parents":[{"commit":"3b02f8abb4e694543ddb303c14a3347472ee77ad","subject":"Merge \"Add OpenTelemetry tracing support for RPC calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3b02f8abb4e694543ddb303c14a3347472ee77ad"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-17 05:34:21.000000000","tz":0},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c1afd11250aa57d60f485ea035dbb59776cad54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c1afd11250aa57d60f485ea035dbb59776cad54"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8ca4a4bfed0af710c90af95a7821471f898b3d93","ours":"3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7","theirs":"3b02f8abb4e694543ddb303c14a3347472ee77ad","merge_strategy":"recursive","contains_conflicts":true}},"6862df61a9d5619379864939e75f0886af986cb5":{"kind":"REWORK","_number":44,"created":"2026-07-17 05:35:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/99/912499/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/912499/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/912499/44"}}},"commit":{"parents":[{"commit":"3b02f8abb4e694543ddb303c14a3347472ee77ad","subject":"Merge \"Add OpenTelemetry tracing support for RPC calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3b02f8abb4e694543ddb303c14a3347472ee77ad"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2023-10-30 06:02:53.000000000","tz":540},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2026-07-17 05:35:42.000000000","tz":0},"subject":"Implement ConsulOperator as a service broker to support HTTP Driver","message":"Implement ConsulOperator as a service broker to support HTTP Driver\n\nThis commit introduces the implementation of ConsulOperator.\nIt is designed to manipulate Consul, serving as a service broker\nfor the HTTP driver.\n\nHTTP driver has been proposed as a new oslo.messaging driver for\nRPC communication. It communicates RPC client and server directly\nover HTTP without a messaging queue.\n\nPartially-implements: blueprint oslo-http-driver (1/6)\nChange-Id: Id84adda53744ef55fc171d35b372d178f38ce068\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6862df61a9d5619379864939e75f0886af986cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6862df61a9d5619379864939e75f0886af986cb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
