)]}'
{"id":"openstack%2Foslo.messaging~914840","triplet_id":"openstack%2Foslo.messaging~master~Ic6c01c7eafb20527e061a8c98225c6f6c9439788","project":"openstack/oslo.messaging","branch":"master","topic":"bp/oslo-http-driver","attention_set":{"36129":{"account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"last_update":"2026-01-19 15:31:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"10115":{"account":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"last_update":"2024-04-03 02:10:24.000000000","reason":"Change was marked ready for review"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-01-14 01:03:49.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}},"33451":{"account":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"last_update":"2024-04-03 02:10:24.000000000","reason":"Change was marked ready for review"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2024-04-03 02:10:24.000000000","reason":"Change was marked ready for review"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-12-26 09:26:52.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-19 15:31:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic6c01c7eafb20527e061a8c98225c6f6c9439788","subject":"Add HTTP driver to oslo.messaging.drivers","status":"NEW","created":"2024-04-02 02:51:22.000000000","updated":"2026-07-17 07:22:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d07f473e2c68e8afd0987abcadf7893542702dc","_number":914840,"virtual_id_number":914840,"owner":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-17 07:22:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":1,"date":"2026-07-17 05:36:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 02:55:58.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-04-02 02:55:58.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"state":"REVIEWER"},{"updated":"2024-04-02 02:55:58.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"state":"REVIEWER"},{"updated":"2024-04-02 02:55:58.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2024-04-02 04:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 09:04:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-11-25 21:46:45.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-09-24 21:43:52.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"}],"messages":[{"id":"acf682101500ae6f35e10551be765b90d08860dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 02:51:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eedf0be1aca41ab93569b7304ea327bac813d743","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 02:55:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6156e1ffa81fd3b3048ea34a1d66052282cc0812","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 03:00:37.000000000","message":"Topic submit-patch2 removed","accounts_in_message":[],"_revision_number":1},{"id":"609e048eb2df6b1a48c5f3d8fec1a9a58a0babc6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 03:00:48.000000000","message":"Topic set to oslo-http","accounts_in_message":[],"_revision_number":1},{"id":"c52ade8c9ebb1bd373907d4292b8b98b67642a1b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 03:05:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ee552510641feb3c6180f78b2a362f2bfd6af0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 04:38:56.000000000","message":"Topic submit-patch2 removed","accounts_in_message":[],"_revision_number":2},{"id":"cb7728fc7331728b440d74250205ea69b3e3be22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-02 04:39:37.000000000","message":"Topic set to bp/oslo-http-driver","accounts_in_message":[],"_revision_number":2},{"id":"5a4f23e11dcafdd1dbb11df9f8de765994f5dc5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 04:58:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0da4bbbb091b4e80905ad890ca509229\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63110c548ef14b46b60212b0144957d7 : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/185d036fed134efea25b8ff92b7e8682 : SUCCESS in 1h 34m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df37a28a5e4d4aa6a2edc1cb97033274 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06931a5f2f8a46838dfd214130eb0c4f : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39769aaf7c264d8580674e758ef99b1b : FAILURE in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/822c855bc23b4864aed220b912066a17 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1ada8e6986d4b3d92dea6e8e81350e8 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d63390bd494783bd6676ebd58e3f58 : SUCCESS in 3m 07s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/401396b067df49eebb7ba5b59e9b3253 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/43ce53f5627b4152b6d7f1d13f491028 : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/766651deb39747159666f22bf328f77f : FAILURE in 1h 18m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/425dde79c45c422088e7683cdd6bed85 : FAILURE in 28m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/961520f851b3453ca90eeed128b40113 : FAILURE in 1h 28m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4ded11f6293f4a45b42a7eafcabddfff : SUCCESS in 44m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2aefaf8b46c54c8487fd9304414bfbfc : SUCCESS in 1h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1eb8c2bc9e92a3f7c910d3eb9c34562065bdb87d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-03 02:10:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"d311935c1901cbd8a051b7be2ead8e953176bdfc","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-03 02:10:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0d4835df98384e8aad9e4a5c8cf98fdf93bcf14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-03 11:17:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3ac7b1918d7b4ad8357cdb04be6e25df15f0febc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 13:51:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6efc7cf2ecae488ba07145769e41ad83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac8dee04c990479685b33e8502a4538a : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/233ca3988eed4d1586663545502c7a3b : SUCCESS in 1h 42m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94396332def1471083c3ee641307e317 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e31f375ec14ae2924a5c081a45279c : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60e270989c74e499024873c31f12373 : FAILURE in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15368cf28dc64f1986e442a0231137e3 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63c8e33fd79d413bb0825395c42b7b50 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bd05932fe2436baf391ce19493c37e : SUCCESS in 3m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/85a3d8ea25f045019809ddf78b91cb8e : SUCCESS in 4m 42s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/21197c6ebd6f4d33848d9a8f661df9ba : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a74d21a0959e4ce898d3e5b452ee81a6 : FAILURE in 1h 33m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7e8675fef0934bd983c0c82a18151a97 : FAILURE in 27m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fecf1048d8bb4e4a88265d88faafd120 : SUCCESS in 1h 39m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9c541efe7a0b411e8442ecf70b9c7246 : SUCCESS in 1h 07m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8447ccd43e7a42fd93fe079c904806e9 : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24bef765b5e1f2a32c5f96765233f3973df66471","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-04 04:12:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fed7e6f121f5e6d91f7d178d427e68e2a2eb8d90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 05:48:57.000000000","message":"Patch Set 4: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b8b5f1ba6e746a5a0f9cc7c275ade4b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7980a264ed4473fb5f00175ead9b974 : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82c9530eb62d440d97d932197b526555 : SUCCESS in 1h 03m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bcc5b7237a64bcca23c1f9dd90ff1b4 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4578f5052c7245fbb7af7907d46101da : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92236eb1a404e61a4da501ac2024c2e : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/050e7c5f5539494a9821de1476d8890d : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a09761568edc4b9fac07fb85fae83a7b : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e99ad74999b4f099d164779dd922d61 : SUCCESS in 3m 16s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/44e3ae4132f044a18ea56981c4ab234f : SUCCESS in 4m 43s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/068f947f80474ef3ac0927a70587697e : FAILURE in 5m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c309d38a07614b38a751cb3568e86a3f : FAILURE in 1h 29m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/998405eff11b4d3f823c745794dc1bbd : FAILURE in 28m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c11691cc33ec45f7b3e3ce77c5759b94 : SUCCESS in 1h 03m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6b354d6f89b14a0594885bf7f4990c89 : SUCCESS in 50m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f08324a839964bb8be1dbe478d14a8b9 : SUCCESS in 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3b823b91e8c92b06c031b99519e8d29559100d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-04-04 05:59:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a758a0169290d3b69b3c852e559856fd128b7e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 07:34:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a35a775dcd8747dda704e13fb7f217e5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/55a53ddca6a34d398f3da67ed67f759a : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a91d6746ee24dfc8f748c2627054f2b : SUCCESS in 1h 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5c446a5e8d497782160d9ef0004837 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dbe60028054d638720291e706fca64 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42f9b597e60d42c2af6e87e1d4beed3a : FAILURE in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee17f8ac0900456e82ecd46fdb2a692a : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c585cf67246b4b7986fa2067f33fafdf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9ea4cfc72e4e669b7d2efff5600c53 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ea7abd500bcd4f498a5157a1d1b61c57 : SUCCESS in 4m 35s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4143b265647c42ff82607b50b06ab65f : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/92b02f487c504bcd8546e44727dc35e7 : SUCCESS in 1h 17m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c48f2c8d74c34bd3a4814618d68e7dd7 : FAILURE in 28m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/405cda61e0ff4a608b3219581be476a8 : FAILURE in 1h 14m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e3d06bc66a904d2abcee57a2ea34f15c : SUCCESS in 1h 01m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e89538c41cc6445293d70812d82fbffd : SUCCESS in 1h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"85172082b1f6fb86d72fa63709387cdf84dcf4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 04:31:46.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5f88febf62499193a39eb7a9365ffa4a277493cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 05:50:44.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b4e2acbe9dc82c1b4eb6384a4937fa838e88012c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 07:40:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62362027b1774e2fb608067b43986f88\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcc8039d20494074bd1a3e8f36bee58f : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c502930a8ed945e795dcc4595f9a8234 : SUCCESS in 1h 21m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b51f3924b94fd2a0aa4feb50c226dd : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fee9afb16e4232808e127109602a46 : FAILURE in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e71eb11534e4f38bba5c29834c30118 : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49d9a6de7d3f4ed3aad933fc1acc7710 : FAILURE in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41fb45e8baf4a9fb3ab9673a89dad4e : SUCCESS in 2m 40s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e972f558cd9d4fe3bb2e7b38c18551ce : FAILURE in 4m 13s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c99f910d849f4a39ad03c45a0366a4cc : FAILURE in 5m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7105c7083c594b1482f9029e9068548b : SUCCESS in 1h 38m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f99a7bcddaec4bc0b38011127a31bc7b : FAILURE in 26m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6eea42b6c3474791b1c9ca0f2588ad24 : SUCCESS in 1h 29m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/785fed9d85b74b73af217992240a447e : SUCCESS in 52m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e333d1d76bd645fb8b5c7b6374e5ec61 : SUCCESS in 1h 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5957044a0fea6389f2746f42708777f29180676c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-11 09:10:06.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d0ed98fdbbff559e7fb95449ddfc91ee4c8a6e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 11:08:10.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40a9c9692da441bd96c1de995c30a373\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4ceca0df5ff4922b00b855f865fb293 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9c59fffed5d4d26a8847b118a1566db : SUCCESS in 1h 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3e003e203394447be87d9b265a0619d : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bee2d2efa37428295b9c40be45ad153 : FAILURE in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/740607ead9af420fb798484da9c2d582 : FAILURE in 11m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c38c274ab71a43b897246c61838768bb : FAILURE in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece63b465f77472a97179d216b056129 : SUCCESS in 3m 04s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/45ffe575ad42473d9acc8cc8ad845565 : SUCCESS in 4m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/88e1320f0d5e414495af497b57808d5f : FAILURE in 10m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5b6ba592eb6b44cda77cf1d2ffb72f08 : SUCCESS in 1h 46m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0240c71301bc497b9925d1ed8618e32e : FAILURE in 28m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/df0137bc139244a4999eec1b477465e9 : SUCCESS in 1h 47m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9681cf341d2941f29bdee0992a5194f6 : SUCCESS in 1h 01m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/59482ec6a1b742cdaabe1f7fe20edf8a : SUCCESS in 1h 09m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"17ba4e9ea7ff94c760636eef29a8f29f7e8b92c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-12 08:55:00.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ccf386270a57c2797ec9c65bce2b1faaf85af4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-12 09:01:54.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"b117ac0752c0b1e7fd099d510b8adb4225dfeb7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-12 09:13:37.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bca50cb8a1002be15e73ef803db3f1bfb1dc1371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 10:58:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80a896ddf24e4a78804bcf9138a6994c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/77e278b82afe48bebeb9c86b4baaa91e : SUCCESS in 1m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b18f03e259e459491654e03bb12217c : SUCCESS in 1h 40m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55a566db052745c99399d85c1868d046 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dc651c3bb0459789a15e438dc6b7fc : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e263fbd86486406b876e1c6e9dcfe034 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40e8ef90cbb048bba666c690a3cfd399 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2041f13a887643fb914cc7882167cd5c : SUCCESS in 3m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3dbc69539f794c45a909aa34659ff627 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/857277e120ab45109b770ab158b675ec : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fc893b37ad2f40259056725942c55629 : SUCCESS in 1h 34m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0f25da40df5d4c3394da6995e0d4f38b : FAILURE in 25m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/66025101341441758ebb64278c158788 : SUCCESS in 1h 44m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/dcd8c87304d04e839ac28bd33b887d58 : SUCCESS in 58m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c554872752dd4e5b8d517874191f8a9e : SUCCESS in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5f43a429207c1dc02a7f08d640a920fbc3500cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-13 03:06:03.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"d0a7d243687713cc8c45d1ee5fb3a17fae060efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 04:56:36.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aa71ed2f1e842b98d3197b27a76e31c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d70e143c8ad24afe993b2fb730c93bfe : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/638f3530a24a4ce2bc4678e203a3c396 : SUCCESS in 1h 36m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/971bd18c50e54dfba25a457ce6b7a66b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba82fb23c08d463e929de4cde268b778 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c02642bafb3247c3b771a530be3e9fbd : SUCCESS in 6m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c1e3a0a9a1f43e8ab348f3075478dc6 : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5fea4c2f5f4350a9d976e91a85a91b : SUCCESS in 3m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bdf52a63c4944651b258bd650eb324eb : SUCCESS in 5m 15s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7d111fce19f24478a597cc48a1399069 : FAILURE in 5m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4e06eb16ba0e4e318389722cd17fbff7 : SUCCESS in 1h 23m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/56efda5b4fa6481699ec5215309b405b : FAILURE in 20m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/624052a161a042949709a8cea279b23d : SUCCESS in 1h 49m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8db1d1caa5b14bac86ff14a5ac982d58 : SUCCESS in 54m 35s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fff9458987448d8be8edc6f816e8a92 : SUCCESS in 1h 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ca5c0c099e4d712f4748203cc3df59ce4f364836","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-13 09:04:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d4d0ec638291857435608d2e6a8af1fb560a1e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-21 05:07:09.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"754f799ff5b31ba561ac520e941bce221e49b2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 06:52:35.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5916a8282a1b4ca3b0f2287c990666ee\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/80a38b7d7e94428592c28c57827171ae : SUCCESS in 1m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82cc0939d82b4eae888f69ab245587e7 : SUCCESS in 1h 39m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f09e5e55b3b42ce98c97cdcd9c6cd3b : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/498f5e44dd3d44c991983b934f4acec9 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4febcf73856470abdf9b1e9adce6589 : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/823f95a5b7064953a1b66c442446616d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e1291cf8a84192b47dc9f1b34e2108 : SUCCESS in 3m 03s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e087c23909144e21afa777773849dfa0 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/94bc0d4131f942d7822af6eb0b820239 : FAILURE in 5m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bc3ca2a47b0f40cf80b1f858314cb42e : FAILURE in 1h 05m 50s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/06ece72e7830411684c4a646457e3dba : SUCCESS in 1h 29m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/82c76aa14d674a09bfdb20963b04482a : SUCCESS in 52m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/92f01126db464cce9582da315f89030f : SUCCESS in 1h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"13bfc5a9210d6a4780853244c2a000fc63bdc7b8","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-21 09:37:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6db91a34ace17d61cb63e5753aa0acd428a0e2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:57:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0244165f9cc4050a9f2f978254d8468\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d3259f9b5d2941a291aa3de9c7ad896d : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b77b921fb938404aa892729ef9da2d7e : SUCCESS in 1h 39m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/630d1b751cbc44da8cf470c124f6f580 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5b7337bc2444ceaf0f7fb5cecdb90d : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87752261240649e9bd799430449468d2 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9fd56fbd39c4707a0e9457e9c71db31 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce97e4592e2403090e791af02648531 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b390927b4a5c4cf1993ff2cccd10dab5 : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/1e72d9ce926549cba4f0e7b8d33f146d : FAILURE in 6m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c4f7373c7c3d4f289ed97304675c46d0 : SUCCESS in 1h 52m 20s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4084996331c84e0b85970101160c966d : SUCCESS in 1h 26m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/75c4d407c532458996aafdb8fae69af1 : SUCCESS in 50m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/722fa5e3710c476d8804ea60875acec2 : SUCCESS in 59m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fa84227c423cedbd40d8e0de02070c5416cab13c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 06:12:30.000000000","message":"Patch Set 13: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0fd21c3ecf3e1ccadbd155fdb0f83792ffb4c4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-25 07:12:09.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"abadd0f7ea177f1f1ac01d1857b067d6c4b2c224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 08:45:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5286aede8a643bba0bc6a455772b55c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7fafa77d38c4120afb40adc4e11083b : SUCCESS in 1m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac567979497f4102afcbf965022f2242 : SUCCESS in 1h 21m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2eec0e92c31d45f092e5d69d6be54c11 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71117ac94fbd4b7a9faf20812c7ba4c2 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/300a2d1d8cab4cc896dbbf652bc411d6 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be06e44f5aa5442981ca39904a42960f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843659faa4ff458c9111e3c9918d2f2b : SUCCESS in 3m 28s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f9d49ee821e14ec4b7d8fcd842a5c6a2 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8fa82483567b4381a3f7db8fc4f0d3ba : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c24caaea56a246ed95f6f23b111e2a3e : SUCCESS in 1h 31m 17s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0be5ec15613e418890f0a656cc9bd0eb : SUCCESS in 1h 27m 46s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3539822ff2414420b7f7dc935aa35ee0 : SUCCESS in 1h 00m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cc41c8b24504fc6864c3486e3e69d70 : SUCCESS in 1h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2ed1421f9a1019755b3e5ddd24b77dd944b9635d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-25 10:10:34.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"70a22640476da519cd41da4f0d39eee5ede82cd3","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-11-25 10:11:17.000000000","message":"Patch Set 15:\n\n(7 comments)","accounts_in_message":[],"_revision_number":15},{"id":"f8f1e62802bbb6564c5d8170cf8dca7ccf93ae3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 11:54:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcce097f4b3e48fb848dfa964889ed1f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e15638a93ddd495080062f2f91e0e397 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eca34df0f5fc41c69569c093353d7c8b : SUCCESS in 1h 21m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f211f6f95dba4699941384c665d6cf16 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a5b675a9134471af00d5363a9c41f6 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eac0a1a10deb4413a65c8e718cc55e91 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c921b107632441d8b99756137263f2f5 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e39c1dbff6412bb52408ddf1649631 : SUCCESS in 2m 35s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c386c50ecc4343de917135a1f1c7b19e : SUCCESS in 4m 03s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5aee0da97ebf42039ed56ec57e04d629 : FAILURE in 6m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/54eb4dcadff74ae09e784f43b6d2dd2d : SUCCESS in 1h 41m 51s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e77d20fd346a41f58ef8ea1ad03ced59 : FAILURE in 1h 31m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/864d15f7d99941a6a00d89be42ea12f3 : SUCCESS in 56m 58s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/92b8e4889e38495ebf496b4ed502f559 : SUCCESS in 1h 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5be5cfcfd75a3ee632deac20709720d905b99b64","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-25 21:46:45.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a997364a1749b5d9a43c6c96b97a66bcea7f8baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-02 08:20:21.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"c40b877f1a5501385c90e8f202020fd8dec36647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 10:40:38.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b93f6d19c14a8f8873e4733e91ae59\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d757789d62534356a8908d396e40f8be : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c13d89d04044b9189545ab30f78ed32 : SUCCESS in 1h 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd712ce5d1ac42d9af0417dbd7a3abd7 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26234508af9460194a546232b1521a2 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd49b5eb3ea4cf98905de4d1c45937f : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4387b0eb134f493a9dc0971989948682 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2694ce18d94ce1939d75a29db9c962 : SUCCESS in 4m 06s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3ac6e75fc9fd4a0a975fd7df4f4021c4 : SUCCESS in 4m 48s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3b02438940344414906518696fb1cdbe : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bd959baaf6be4fbdb9fc12002cc07a66 : SUCCESS in 1h 29m 12s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ae046f9d5a2f4e048307d19094a45c09 : SUCCESS in 1h 59m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/652616235b514ff2a31818d3d754cfbd : SUCCESS in 35m 57s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/411531e460764c0794e385da6ff60a96 : SUCCESS in 1h 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5202bf219a39bf4304b6c63364ffac89022c415a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-09 06:11:57.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"fa385155a1303dbdc0e28c389ddde6b0f08d5114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 08:06:36.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f6b8d93b3744da9fe29902da6f4e75\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a2b8e1d6542e43e794545b186bb648d2 : SUCCESS in 1m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9489caec69a14dca80d302385dc6b272 : SUCCESS in 1h 00m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b18f2661794a45ae5e8c409c1019b8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/815f10143c41450a957757f11f0d0e95 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49ed191fc2574a41bfb7e7616784d465 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30039e378fe948068cc3b7b0a50fc1ca : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59626d0691c4a9ba6b43125e9fae87f : SUCCESS in 3m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f87ef7bee7094f5a83f2fac729be183f : SUCCESS in 11m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/b9537b236f1549a5980dc790bf3cbd37 : FAILURE in 6m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/75375a78bc74444f82c3a52511cfe44b : SUCCESS in 1h 30m 08s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b36ab857fc5a4dd29950d2227892498e : SUCCESS in 1h 46m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b8b92813854b4bc0914be69ec41314cd : SUCCESS in 1h 06m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e8d92999fae4d1a8c5bcfedb92b4bb8 : SUCCESS in 1h 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a9df46f750686c809c0d2020b1c0d3805c085f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-19 05:58:24.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"bb43498c89bfb5c7c960e2f710f974b265810b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 07:46:36.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2967d347375044cabd379ddd8886445d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32ea20379b4c479fa47c51690261e7af : SUCCESS in 1m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b94c6668db243518d976ad9b03623de : SUCCESS in 1h 42m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76660e85bc7c463b9825bc4618c42e8b : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8270c3f7e314f26b49579dd476c98ed : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d6d2e8d719e422ba8751e0f92ada68e : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8f2bd43885d41559ea53d8bae46678c : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711a6e6cfec14d31bfd591b8edcd4de3 : SUCCESS in 2m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8dcb5be35ba244de86d40e6ca68238d1 : SUCCESS in 4m 26s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7a724dc152604222af54922bce78c390 : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0771fec4a66048859e255f9308fc62a3 : SUCCESS in 1h 42m 08s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6138d2de660949748410148391b6edac : SUCCESS in 1h 45m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/10bad90e13db466bbb768cc67db5a75b : SUCCESS in 58m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3565e3e2c1fb4eff9aa075bab5207240 : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1249d5a3a33a7e4b4da007872df24d6becefdd08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-27 11:06:22.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"f318dbb905e0419c84805b35910b90f8bfd59f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 12:46:18.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b4cdef9f2ac4f98a716d3de11ba1f0f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fc04d3c200846f785c172363179bd2c : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/837fd1a74421454c935af485fa35131a : SUCCESS in 1h 14m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9108a722d8ac4200acd888241d604d55 : FAILURE in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748b790b60ca402599e1c10f992347db : FAILURE in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9610fe6fa874695b7c64a1ea29fca21 : FAILURE in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f911215c9f84c5a8418b0c318b8bf7d : FAILURE in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/236346a4169d4e7b98094a10d969f57f : SUCCESS in 2m 50s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8458dc909791409a9b5349610cfbe02f : FAILURE in 4m 03s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6b8f4406586a4e76a72a3785b3e3dfbf : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6d15358793284bd7a4ca416c36b31d43 : SUCCESS in 1h 15m 19s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8b72cbea86234119945af3f1ed673829 : SUCCESS in 1h 39m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6d463edab16a441b9054afa8aeab7b68 : SUCCESS in 44m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1bb93a686544bdc9ca70596b66b9af1 : SUCCESS in 1h 13m 46s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/tests/functional/http_driver/rpc_test.py\n  Comments left for invalid file oslo_messaging/tests/functional/test_http_driver.py","accounts_in_message":[],"_revision_number":19},{"id":"920e11cc36c31a9f26108de31fda3a56cc970c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-27 14:53:30.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"870f042dfd1f686b74a2d317f8d30c5efeb32c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 16:39:00.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ca19568af5840f29b3d7b3670ee7cdd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c10ff977aac24aa48f1132f9eb208a0f : SUCCESS in 1m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42db5d04a6004d0c85762d748dcb19a5 : SUCCESS in 1h 29m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22c65440cfec47dbbfb6bb70ef7585e1 : FAILURE in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e33bbf2390430f9964d4ebda972bd6 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b3036978716435f83cfe9cfbfb633ee : FAILURE in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e75dc89ca0f4e648b53f33cb7cfa4e7 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be61aef5870b409fa11977441a8ef693 : SUCCESS in 6m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/7d533dca81b74850a747e63ca5cd7e86 : SUCCESS in 4m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7580012d3726469fb3cc8a40bd707f4c : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e51f5562c4a541d5b1e6f83fca680948 : SUCCESS in 1h 43m 39s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2f03db093176408c83737f5d7a17f913 : SUCCESS in 1h 27m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b9453e498428416cb139d548aa60e06b : SUCCESS in 1h 13m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0bb8f82022f46019db7717de173ec52 : SUCCESS in 1h 00m 25s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"97621b06ba0415cf4b03858077dc2e39fb0e161b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2024-12-28 06:59:53.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"7fc3338fbcd886ab666f618ebc31602bbbb379ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-28 08:22:30.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f83ba10ac4564d099851285445a4cd05\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7cef2c61d0fa461a9c0a20cb93a7dfc8 : SUCCESS in 3m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3d9482013614869bf4c4d24ad72bbad : SUCCESS in 1h 11m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3b499a825204c729a3c844254738641 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1754bdaea8b344309836fab2dcd76ea9 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd1ba229b827486abea2c4d27aae25ca : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/559b2b0996b54a4b8b1c346268fc9489 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c6e31d1107472ea7ba30d064a2b07e : SUCCESS in 3m 18s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a93379e10a804563a9cf88f2da1c8f07 : SUCCESS in 4m 14s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c77672dcfe4049b1b6f2c0e766a17dd5 : FAILURE in 5m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6fbaf143bbdf4f92996c53b115a4ed9b : SUCCESS in 1h 09m 23s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/96d300edf5fb4f399af2a0d6dddbdf6d : SUCCESS in 1h 21m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7171cea598ab4edab931bd7cbb354bc7 : SUCCESS in 43m 23s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5f643bc114b485090ef887a0a5cc3c1 : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"5f57cecf7cd9ccae9a4ae876db51e1a80cf7813c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-07 07:47:38.000000000","message":"Uploaded patch set 22: New patch set was added with same tree, parent tree, and commit message as Patch Set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"e520759721f6146f59af687e5db06f857e27f192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:41:10.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d1df4516af4d54a42b7d124ef180eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c86948dd03a46658a08f27abca5c3f7 : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c596803cfa7474e90ecb4be9b6e19a3 : SUCCESS in 1h 34m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e09f3fd05cc466fbb3c1321a06be038 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c784596ce5aa4af59b38a83fa7f94517 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e7ee346d85459c81fa03b259086c8c : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b2f9a452d534fd092ebc46fef5a7147 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fc404eeeed4707beb9bf11526cfb31 : SUCCESS in 4m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ef752703404041e7a2739aac79c5a7b8 : SUCCESS in 4m 35s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/cf2393c173944631adb419a73eeffa04 : FAILURE in 5m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/451806bdf4ec44eaae0d75e3be825954 : SUCCESS in 1h 28m 59s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8c66ed8b7fe84263b2f8ed1801f9cb18 : SUCCESS in 1h 43m 59s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3824e1c409ba4669ad7427f34ef5ee69 : SUCCESS in 1h 01m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/387217440d6b4200a58be5bbdc8a84a1 : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7a52095d9bebdc9b37ce0856339edc4af25df20b","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-14 01:03:49.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"ddad96392a5a2c829497c27dac8d46765faa7b69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-05 05:19:19.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"3aa7135f9a1821ed3f0c244635049d3b680b32ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 07:01:43.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10481a4cb5fd4926832084981bef0460\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bde275fe4efd4b9cb79b13038444e905 : SUCCESS in 1m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7a696313df7455580f558e2b83b27fd : SUCCESS in 1h 13m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce61e6e3557e44cc97d5d582e1f5f573 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0118198315ed45949aff521d9a880b56 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24a56b424eac44a797bdfa2683aae999 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fa060550be54a73ab78f8066efd5925 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4226c2456547b2828d29c5be54b067 : SUCCESS in 3m 42s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8dcffc624972436c9ee931ab07493e9f : SUCCESS in 4m 41s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/006989b34e804ac592bd5d669ec2c8ee : FAILURE in 5m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0e8caf6c83264fdaa03db2d6f8fa4505 : SUCCESS in 1h 29m 21s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1e80bfb99e1d44408747b0509cd24368 : SUCCESS in 1h 37m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5337a1e21f8d46ebae596af02b10db7c : SUCCESS in 52m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/23252e779ac4404494bf48dca6916d52 : SUCCESS in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"278995f9ec3c12851d139cdd248fe2af35b5d79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-19 08:34:25.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"422521aad7fba116644f2f8fb7cbf39f6ab0bd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 11:01:18.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e72c66bdb14f47aa628f6da162e3fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ad80714c622472385ac14499d343155 : SUCCESS in 2m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3f9b91754584cab9cd91d14393f798d : SUCCESS in 1h 28m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dbf3d9563264f5c97145d32661a10ac : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9827b46d9a814688ba37d526ef875ce8 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43ebceb898fc4b10b6c99ccef98b289f : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68f4bbfb86e246b8adcd6cdbc1c46c96 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039ce0f9a3994f4d83b93caa2a4dfcc8 : SUCCESS in 3m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/55de9737295a441f9d715bcec7485c4c : SUCCESS in 5m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/18a9e364fe6d46e4a3ba80763a19dc06 : FAILURE in 9m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f67a04b2f7a48f99e90893849606708 : SUCCESS in 2h 26m 03s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2f07ca541cb940e4bc32def694a3ac33 : SUCCESS in 1h 25m 26s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5235ac3162cd47d980718c477073e41d : SUCCESS in 53m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cce92e211c845e984899069416694f2 : SUCCESS in 1h 11m 59s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c9fd5ea05d2c562ae6c9c3fd4e3c1e98997d3aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-21 11:03:04.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"4d86039516961c872e1b48dd889138a29c216927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:47:45.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd808a9b9d8434fb3a8bddef577b184\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc02bf249e874b178bfaa61d5c482445 : SUCCESS in 1m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6707c1a0aed74c10b8c29ae36233f4b7 : SUCCESS in 1h 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9c82e1be0df428f91f925c6003699f0 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/751f5a1a070a43428bd9f09aa9238a81 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb0b868966974b8aad8d610cf7484bf0 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/048ddcb4aca24e3e8e07d4687708a81b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a44117e516bf442fa64500786dec37ff : SUCCESS in 3m 26s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/fabdc186908e45ddb48e7d7f0a97cb4f : SUCCESS in 3m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/47bd01e0556e4a49a0ee7d526e734249 : FAILURE in 5m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2124094647f745f3bc6408cbb9cceade : SUCCESS in 1h 19m 14s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b8e965a716404c34a9572be1e81471bd : SUCCESS in 1h 38m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f7d27e32867947efaf700235ed0ca09d : SUCCESS in 33m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/622bc31bcfc64481a0e405d038a50ec9 : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"056c28b92023b290f2141a78779d0690d53d0091","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-31 10:04:25.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8aa00919bc3dcf30e31492b1185d2aa81ca795fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-03 10:46:00.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"a792a937dcbbc6a2771bb8e2d3775e4142ba4e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 12:46:09.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db42d3cd99884f26a4afe0c410f5b779\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/919ccf196ec44d488933433e74f69e35 : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f77d1961021a4f1188badc37e2f0784b : SUCCESS in 1h 39m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f99ff2c8265450699586c8c7d06fa58 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4f6a8a3a314301bdeb3507a972236f : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96966c5b738446d091df23515bb2eff3 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b81b33c4f88480b8406a26342decbc4 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98cb9e893e8a4819ae3fe62b95bc1f79 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d20fc81ebb46b7b9af9663ceea8f38 : SUCCESS in 3m 39s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/951e59b606e14d0fa57a853624b4b000 : SUCCESS in 5m 14s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0c831f72e044475498d18ed311121218 : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/585635e48c9c4220a0eb1c43114838dc : SUCCESS in 1h 47m 52s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/046d1cbb64ea45138bc56043fb5abb32 : SUCCESS in 1h 32m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bb96b785ee8640a4a2d6dd84fe263d26 : SUCCESS in 42m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/08c55e3bb226440789ce1252a85aa4b7 : SUCCESS in 42m 20s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"084c4a9157ab229e5aadac79c8cf3dc425e31938","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-30 10:14:34.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"3d4fdabd5eff11c732d8c58f279bdfc67ebee901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:22:43.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c4330f4a5942f79c11ac847e7e9866\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8f27d0f23aeb4da4a53cb5b727f7408d : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84bf606bd01d49ed9debc350479fdb5b : SUCCESS in 2h 01m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a5c75ddab6344f0a8962bee5c02d296 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3639ce5c73144d60be1a1140f5c4b144 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c67ac2187f184d0ab0b85ce915766b0f : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7469da5744db42ca92367bf4b41cd9f6 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65cc10e74fbd4164866cfeba3e6edb40 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35eda7ce42743e5b1fd926bcd166445 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1f34766796ef4c0bad01532e03601118 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f4330f6a9ef941de831a0258656f46a3 : FAILURE in 5m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dfb1a9cd728f480c879e7ffe47a3f1de : SUCCESS in 1h 25m 44s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/60c3a69baa0f43eeb3eba902db75850e : FAILURE in 1h 20m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/29db5aa7babf4e0f8b0424f18af5f276 : SUCCESS in 28m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4dfbbdc9e5c43aba1460d2a87880f4c : SUCCESS in 1h 03m 15s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a1db7ba7913e40e4fc59f58b9263b9b9ef4246f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-07-29 09:10:49.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"37712cb7e67ee95e75d67cfce877e3162b6aed06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 11:09:37.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7058e35307cd4d90b71d0560a8f2579b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/34afc7727e4c4b5da388968ae0f37da1 : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a270cbf366446fcb39c21472ab6b94b : SUCCESS in 1h 40m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2630214f26b94b4ca0e6271fcfdfe34e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d27130a385946f3bae86d1dfdea9281 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61584679b13347bf96cd50afc5aee4da : SUCCESS in 5m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c870c990d7494d6f991df5178872cce3 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98bc354db2ab4c55960d336aa23000dc : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b7d966ee554f48b7385b4879d0808e : SUCCESS in 4m 26s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6d6873705bd64af1898c0b43177d50c7 : SUCCESS in 6m 27s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/2158baef030d482dbf42cd7bd1bbfa08 : FAILURE in 6m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/20663cd91d734cd4bac049f52f103db5 : SUCCESS in 1h 54m 59s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/be052ee2f3eb4ade99452973fbd8c49b : SUCCESS in 1h 39m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4c73bba76fa043f58300c33dd8c0f12c : SUCCESS in 1h 02m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/88e98d71fc064c709d50d6e750f9589a : SUCCESS in 1h 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3a344c2d09010b94fb491b855fbc5ddd3fc66818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 04:49:40.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"390ebfaf72077a78aab278f845e751248dbb73ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 05:57:21.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99fed351bfd14714ab3ff4123b53ac5d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b477802aae444ffb5b12fed8717b209 : FAILURE in 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a53112c6b8114d779e93cc5bb08187a1 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab29e118d75499aa72c62aea34fa31e : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c484a565f15542768a05097ec25a2685 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8a884b344c4446094b5a47f072ca962 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ced9a7921957493fbdab9c65f750db66 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e9f559fa02489399611b3859f6b065 : SUCCESS in 4m 00s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/7c8bd70acf094385a3a8fafc10d58586 : SUCCESS in 7m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/54f93891514a45a7bf0faeaf91dd5921 : FAILURE in 12m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c8cfbcafa3b7420f9f78fd6561197e90 : FAILURE in 21m 38s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d31f06a527e64f578ee20a84422ce7ad : FAILURE in 30m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ab2b968fa2454856a366cbcf09765066 : SUCCESS in 58m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/09bff178f5fe44fab6d2fcd0de8ba3fb : FAILURE in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"f407dc2622fcf5961caa5b6be5f64332acaec6b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 09:23:14.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"0d50e78c94ed3d7d80275ecc09e69fa4eeef109c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:28:09.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb588493da4c460da41dafc0e2babb9f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9032efe2af1745738d5699ba842ee1a4 : SUCCESS in 1h 20m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e2880cf6802494dbf2d477fee6065f8 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016a5a91f8a24b879fa3c75f885bed4d : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3333ea099b7f4d7288c8833f5bd1990e : FAILURE in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ce9497e2b5e4fd38901645c0eab050e : FAILURE in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1127d3f5ba324f30a9653d6bd3a6b2de : FAILURE in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09133b5dacc14d91b05e0cc473b5a7a2 : SUCCESS in 3m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/824b4be225954067be80700d71ad7c9d : SUCCESS in 5m 20s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/efded4246a724a6492b7cd0f6e9ac7d4 : FAILURE in 11m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/201bf896fabe4b64998e990118537417 : SUCCESS in 1h 38m 40s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f6014dcedd47400b923f2b99bc5e1181 : SUCCESS in 1h 48m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/804a7c844d904f349dc4605cc276dbb5 : SUCCESS in 58m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/695c80870c2247959a513311c999013f : SUCCESS in 1h 04m 22s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"71c55decf16279a5095ab3afc549b3efb27d4a4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-19 01:37:23.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"c27f6357b43e062f32ce033a1ba9c08838321243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 03:25:00.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a11123343248658513ef9c2f7316f3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75688035091441c7983f10324f300907 : SUCCESS in 1h 44m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad938401bd1c4152a3c4a43b5fc32f7a : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e170962a6e7e409da842060493cf7e3e : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/751ffc78d5454d18a18a1b7977b08297 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f4f52b6aebf416bb606375703af15cd : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3a16c1aa5c347da992989c197d8f668 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f28fa239ce044d99919daca7d5d5663 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/09248bbfd80a496baa62dffa45ed2aee : SUCCESS in 4m 59s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e035c4275ab549efa5589865507c7311 : FAILURE in 8m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b7139d73125e432695e66057cb880d7c : SUCCESS in 1h 25m 24s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e213b590a24446e5a1236c59b4676ead : SUCCESS in 1h 42m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3f8198c30662458b848d1a8f9d200522 : SUCCESS in 1h 03m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5431a85f25a4e43b99e0cc129849d9a : SUCCESS in 1h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"0719e76e7b10d59cf99c0afaef7ef7c819440ce4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-24 21:43:52.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"8ebaa249b3aacabd16d716eb103a9cd053aabf66","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-24 21:45:04.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4141616a977265637d00dd5579e3ef9f73bbb260","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 10:22:12.000000000","message":"Patch Set 31: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"80ba2c798cd77573a2a58eee24d51affa87b972f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:39:20.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"888aca506bc407cbfaf86ff4cc0da617be4e3792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:50:49.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":33},{"id":"ff470df50c8c27f902b457c31220eb34c196a669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 09:19:44.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":34},{"id":"f4bd01759621866ce45ecd4c24f26c03a0fe5f37","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 09:26:52.000000000","message":"Patch Set 33:\n\n(3 comments)","accounts_in_message":[],"_revision_number":33},{"id":"b5438796bf9192fe6bf2b79b862d8522b22600de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-26 11:20:06.000000000","message":"Patch Set 34: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e0ef951997c4484bda05226264236fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0151c4ff5759465fae0bbc9055b086b4 : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9113c48ed4604699917dfe02e7c61fda : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42c596e4bf504ffdb274565d62788cd2 : SUCCESS in 1h 14m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb53688af68748828073d239187843b1 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b860e418de334daca4c2a137266f34d1 : FAILURE in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c56cba266e74fa586985657125c337f : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9e8f5e117744f78afb92da9b1bfa071 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee60a68635f94ed4941a0f19e1684250 : FAILURE in 4m 02s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/18b84e79e48e434eb7299fc5f7b01ac9 : SUCCESS in 3m 55s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5606cc7abe9d447a973b27ee0abab5a8 : FAILURE in 5m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c49c136f4b964013bb30cb8c36901000 : SUCCESS in 1h 55m 11s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5d4c8dbabe834d30bd030d973c07bbbe : SUCCESS in 1h 32m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c2e4283b603c4270b650bd4aeb12cbff : SUCCESS in 32m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9239f5ffd7f45bfb31bd7342a7bb25b : SUCCESS in 1h 11m 03s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/_drivers/http_driver/consul_operator.py","accounts_in_message":[],"_revision_number":34},{"id":"3bd795f1f0e628a8f6fe11ff292f1cfdb0592daf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-05 09:09:45.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"c86b072a1924f7706fa579ad93c8fcc7fce69b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 10:57:14.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bab48643fb5048c7824970b15ac82247\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee5ff89012fa4a8d8139e10cdae48652 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/87083d23302e468fb8676e28399f29b1 : SUCCESS in 3m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49873f2247eb4ecfa13bd35d28db65bc : SUCCESS in 1h 41m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1f782aa137489c8ef8878cbcf4c0e8 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb33d23a5b4d1b8b084cae3f2829f6 : FAILURE in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f934494e41a64157aa2b25d4870f6328 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26fd27f9fa124e7cbd2aff32b2c20cab : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c2f487bccf4042992d1bdf24440e64 : SUCCESS in 1m 59s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/69218cb78aac4bd08b51bfc1b473ecf7 : SUCCESS in 4m 01s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/ec8ae9ab78c443cdaeb38b1eec2069ae : FAILURE in 8m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bcd234f242cf4e57a1a4afc98e345ab4 : SUCCESS in 1h 42m 34s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/86e51433a4064fdaa0ff453515b957e9 : SUCCESS in 1h 30m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/515982e793b249e887560cd48a7fc287 : SUCCESS in 1h 16m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/00977366e6b24fbca7858c3e849e086a : SUCCESS in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e4bedee318971150438ca74dd4031a96803da4e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 08:30:42.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"2c981df004a4ecd0e67d7178ae8e9db43c31098d","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 08:57:05.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"6f0643b8e01f9bd3de4ed271ce562ae7f29bc425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 10:12:00.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc626612ece46c6b005e42c475330de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15b93f99ee78422f9ef3cbe3d8f6e8de : SUCCESS in 2m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/69675f029abb4ec7af906ff77350ff92 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dbdf05049ce47c4a5f246d9918e8d91 : SUCCESS in 1h 19m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69c0d328b1f643e9bdc543afbd9bf666 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df303b794f6f41169181abb96e07bd39 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a95e9a5ffc424f2a87cee85dae66f45d : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32c3891bdeca4f569cd1fe9ac85726b9 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703244a9153c4f25a82e24c92719d93c : SUCCESS in 3m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a7bed9f5ef4b408ebde162f323a437a6 : SUCCESS in 6m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6bdc944b5e0f4514a5677aef1ca5a78f : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/74f4f58c658c49ef958f34f722c32997 : SUCCESS in 1h 30m 32s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c04106e1ae1143309ebfa224ce341834 : SUCCESS in 1h 39m 33s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e2d29aa6d6b14cd8a668167bbaca7fa2 : SUCCESS in 59m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/299deccdc46f46be91b1df615f3158ba : SUCCESS in 1h 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"abe3770bc8108228301c8056e8d3433cdde50b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-15 09:35:57.000000000","message":"Uploaded patch set 37: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"7355231632f799e93a35c94d737faa9c4e793c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 11:23:33.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f5f53ef36b24a1d90d30f81132316cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0cea059682f41d7ba4d711aaf8e8117 : SUCCESS in 2m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe6e1523040f47249a7ff931368b3f3a : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4cd37ceb84d476ab64ed4bbead92ab4 : SUCCESS in 1h 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7940c40eb2345878a4e0b9f4fc74cb3 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9bc20a9f2d842c49c9f7b9b2c1470d3 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e761ec47784b26959170eb6f039d23 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ea8c8fadde945159746bcfe48906f8a : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212a3f2ddc93495e89f8540d6f79b4d8 : SUCCESS in 2m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6411774b73b24ec8ba0fb1a077f1d676 : SUCCESS in 4m 50s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5a4238ee9db842d2b2f74c184edb545f : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f9c78bf514f14108b87217f1bdef8d18 : SUCCESS in 1h 35m 17s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b45444f3ca7f4b90976986f378cfcfc9 : SUCCESS in 1h 12m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e2c94baa218c4343a69baa2b623deac9 : SUCCESS in 57m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9189adf5189847058e725e3c0133c8f3 : SUCCESS in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"b0cbf1d44b82a1bc88d2c136f4630fb2fb9757ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-19 14:48:53.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"05f215f6c5f709f655791487d918893244ff5f5d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-19 15:31:57.000000000","message":"Patch Set 38: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"be7f16901e4dc45da7e5061c66ed9ee3b4dcd949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 17:27:08.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d722e499714ab98cd54e1962e5d067\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d393c2dab8cf4a83b9532fa5d7f4f20c : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1659a51d29d4880b7eadf6a485a4ee3 : SUCCESS in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39d9bc68066a42329a9480748ab9a140 : SUCCESS in 58m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ede5cb906feb4806aa900949b9d1db07 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404db8c4dd6e46da909d835550605909 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4db21e63d7e48a396a2800357b7c179 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00193b741637433bbefe945487b10f26 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95eec1b277f457db6eda7323db021b1 : SUCCESS in 3m 59s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0092941fd6eb4bba89994504b79c64e7 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a81e674863664d738cfc605f3b20099c : FAILURE in 6m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a7bf5755aebb4b3baf2428e2075a7f61 : SUCCESS in 1h 22m 47s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/800a333c914a4dc3b9fd854971f35c73 : SUCCESS in 2h 08m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/76ddba1bbc2a4a268385d21e5016ce96 : SUCCESS in 49m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/755696d8627045a6a95748a663f401f0 : SUCCESS in 53m 57s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"1d41d937ee0c079b9af62ef0de57c8138bc94187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 07:45:37.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"a0ffe2a64804e7efa28c8fc2203b417d63d1cdf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:49:51.000000000","message":"Patch Set 39: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfe4af2b9574f199aa59720f79d09a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70115183af3946e791eb165ca2c8e96f : SUCCESS in 1m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/03456fd65ee14393bd256093f22345db : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/558d223f16f74938a69620d792a45e8c : SUCCESS in 59m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c97f260e6ff4518846b1b1293ab399a : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e612aa5c5f14d63acdd7e8ba16ae850 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0271eac4f6a4d3f87d270c0b4ea7dcb : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93aeef99defc48548ace0a9a33e0e49a : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a828721924ac4a53beef3b573f5a9bf7 : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0a0b2602ca054706a1b548d48279db99 : SUCCESS in 4m 46s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/53696d433fe646e0bae85141ad0afc56 : FAILURE in 5m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3aeb4d8340ac45e2989ec189e4f87d70 : SUCCESS in 1h 31m 09s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1e07bd078f82401e90ce8590acf87d19 : SUCCESS in 1h 56m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6e80d10036054fe5a1f9f71e4a6d3b87 : SUCCESS in 30m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4fda65a7f07c4bab8380296c61dd3763 : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"c53bc048776f2b413b5f10ccf650a79c980fb987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 11:23:17.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":40},{"id":"67d15b087ecb2fc9062c0edf2e001688abf2856c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:08:45.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788a55507abe4c50a8b91f4bdd07ce93\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af99b58f51ef4fabb24a351686bc41e7 : SUCCESS in 2m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01d463e7fe9648d7b3c15171e3dcf976 : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fed39688dce84301847cdc45883090e1 : SUCCESS in 1h 37m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/158b8dc909d947b6ae35df514f42f154 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967f3b0bab32421b8585147c9f6c0c63 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c21d652712540a6ab71ca7cf7d15c71 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f59d143b03fc4b3fb0b508e8613356cf : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ef02dcd24d4e4285cfd40a514720a1 : SUCCESS in 3m 13s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0d08304d43214c70b155a409c9981ad3 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7d8558f6418d44f18ed6ba0bc2e4e51f : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/846e1e010d7d437ab484b230e68587d6 : SUCCESS in 1h 42m 42s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c3f4cb64883a4139b7792cd8420d36f2 : SUCCESS in 1h 15m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c3f99ed06c154f60a732e799f74d6177 : SUCCESS in 34m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f8a3dca48af44858d9e50e0c323466e : SUCCESS in 1h 07m 59s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"bb381472d56b93824b7e02402b9929c05d599123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-26 08:24:40.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"dea59c5e942b28bd49c835653b680808562139ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 10:22:35.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ee7361f7f54bee8513e81c3d9d8011\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6884dfbbf68b487fb5bdf6d670d1cb03 : SUCCESS in 1m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/704f9b68a1f64bfd8a2c340afcf37201 : SUCCESS in 3m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/202698595d0c40048a85b9f068978c48 : SUCCESS in 54m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/100480b3a6d942c2a0951071755a2250 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f704f2ce09d4d20acc670d23657c6dd : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78118039de144d1aac5a5af9262f01ec : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d2b3b94a08b46a0804fea40b84d0697 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1901777136544230979ef7e7e349e9c9 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6bf7ecc4693843f0a82dcacb723f3a3b : SUCCESS in 4m 32s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a76ac6e9621a444aa1d1fda79546be07 : FAILURE in 7m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6d7f4fb9cf7143cf887baa98ff45a2fb : SUCCESS in 1h 43m 24s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f12b773a34d84dec92aff5e683b3d6ba : SUCCESS in 1h 51m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b423ee076e9a4d0ea697d99b06763d72 : SUCCESS in 58m 27s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b44dcde3893240b58571cac1d94a3e20 : SUCCESS in 1h 11m 55s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"fcc3e359a6b3b777fed0398df292d8e09127deaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"date":"2026-06-15 01:10:43.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":42},{"id":"28e3d3afc207795cce4b1c97b3341be575f034c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 03:01:29.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cedab0559b84022b115e31bec4578af\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a4a6a0d74ec4835953042d8e47cae92 : SUCCESS in 2m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/398d4a891e644d3b83c1436c6cbe8fa0 : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b048d547a184b3a9026f99c9497a214 : SUCCESS in 1h 37m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b25668c57cb7457e8731d19b44819946 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/222cabb32a184b2fa61a42355e9da42c : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c76ab38df06048c69ef3f0f4c11e7e78 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2159cb109d8347a688a4aab4369af9fd : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a95c675d9b60431e991c468354cb55fa : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8906427446184463acca8132fe8e33fe : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ccca1a111cd4b519135c13a2c1b9d32 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/c85296021a9e477e934ee7c36e22e286 : SUCCESS in 4m 29s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/7d7748c6f0d54256b910ed6150b1b167 : FAILURE in 4m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84e2b7a785934302b21f5ea089b8907d : SUCCESS in 1h 44m 31s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/666c9875182345f3bbff8acabfd2de5a : SUCCESS in 1h 07m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f9facb09988343a2859a7d2146986de0 : SUCCESS in 1h 04m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c505bca587af4d379f2978acd4154603 : SUCCESS in 1h 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"b34c0ab2fd4a483bd3470c811a104629d82e8710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 21:34:43.000000000","message":"Patch Set 43: Patch Set 42 was rebased on behalf of \u003cGERRIT_ACCOUNT_33451\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"_revision_number":43},{"id":"8755d29e15a79e395a11960b43cb3b3e5ffb538e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 23:21:50.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f815435bf51248a4a7a2abe7eaee14b9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10ed950a6c6a4cefbfb3e4bff2135759 : SUCCESS in 3m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f236ad8d036044b1be21cfdc08043c05 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55168f69bdae41478778234c4bb752a3 : SUCCESS in 1h 39m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a312cdf3af1a48f8963bc1cf1a6a7ce3 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5324895e5e4832a7963cc86ca1fa57 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97b3a9459d4b4aa493dd84aca0827b46 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29981518a0fc4e00a8eca086269b7829 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26d1d64750d44603800466985c52223c : SUCCESS in 6m 03s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/33fb1fdb116b424eabb02235d00510b1 : FAILURE in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5aba7fee66449b299c829e43f96a1ae : SUCCESS in 2m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/070694defac442ffa073af5a46366ff1 : SUCCESS in 3m 45s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/244ceefcd0874203a57f0b53ee9b8e3f : SUCCESS in 4m 22s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/5d45e17f12324931a211aaba5bc05193 : FAILURE in 4m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1e65d9d08c014bdd9f142facf33d4a27 : SUCCESS in 1h 30m 29s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/015dce07bceb4b559c8a483a61e080fb : SUCCESS in 1h 41m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ccb6c564e53a4051b215ce31e52aabfa : SUCCESS in 1h 01m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3d8abee52d247e3ba049105f7174352 : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"8835da7737ade6a46ea860b7637c0615cf5bd2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-17 05:36:29.000000000","message":"Patch Set 44: Patch Set 43 was rebased on behalf of \u003cGERRIT_ACCOUNT_33451\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"_revision_number":44},{"id":"1d07f473e2c68e8afd0987abcadf7893542702dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 07:22:47.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b851a916594486a8a2af59c93aa834\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2cb92254b5040ae89c9c3274f035bc3 : SUCCESS in 1m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fbf889b72f1a4903b723acfa14f3edc9 : SUCCESS in 1m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3710c52767e54d8a9b7f8daa36da7168 : SUCCESS in 1h 30m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/419b0159985c4c89a3ba62b8d56c71b4 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aacc698fa00443d2a940400de6db5b29 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d20586c74da43ccbddd499aca2b8d18 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2446a00b106541b1bf2b3c2a17e8ec78 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5df7d726e6f84f4caa7057b1a3c260ab : SUCCESS in 5m 54s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/508e44da55264865b6886dbe1b98c1c6 : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd63667a5d8b41d1ab48536ca5b9cd5c : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f1a1903437f4a34a3882f1a96589fee : SUCCESS in 4m 05s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/3edd445772114ebb966ed1f43d395378 : SUCCESS in 3m 40s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/4a6c5407c3554b0aac1bcc5c891d03a7 : FAILURE in 2m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9b98c8b87e4c4bb5ad058797c80b9a55 : SUCCESS in 1h 01m 31s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/853a2640a6084c6cbecb6b0909b1715f : SUCCESS in 1h 39m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5d176eaf9b9c4dd1a4d7999c42d8899c : SUCCESS in 1h 02m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/078ddcb2cf844337950356b0c76a9b6c : SUCCESS in 1h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"95586df45f2a99d728205c14de78a0e9b0671d2f","revisions":{"aea0a09e090d6f01929f8c9762a38608f0427797":{"kind":"REWORK","_number":1,"created":"2024-04-02 02:51:22.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/1"}}},"commit":{"parents":[{"commit":"2092c76255aac57d89f2248b78bbfa9636cdd54d","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2092c76255aac57d89f2248b78bbfa9636cdd54d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 02:51:19.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aea0a09e090d6f01929f8c9762a38608f0427797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aea0a09e090d6f01929f8c9762a38608f0427797"}]},"branch":"refs/heads/master"},"6b53535e364bbcdf620163b33ecdcfba4ada089e":{"kind":"REWORK","_number":2,"created":"2024-04-02 03:05:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/2"}}},"commit":{"parents":[{"commit":"7d0660b64cb1fdee33ac7168369837fd22bc58f1","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7d0660b64cb1fdee33ac7168369837fd22bc58f1"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 03:04:38.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6b53535e364bbcdf620163b33ecdcfba4ada089e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6b53535e364bbcdf620163b33ecdcfba4ada089e"}]},"branch":"refs/heads/master"},"a78c99a23c5a434b7a2d1266754cd7bd5e7b062a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-03 11:17:48.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/3"}}},"commit":{"parents":[{"commit":"d06ac93054d0d850e544f9f600915e26b072227d","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d06ac93054d0d850e544f9f600915e26b072227d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-03 11:11:21.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a78c99a23c5a434b7a2d1266754cd7bd5e7b062a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a78c99a23c5a434b7a2d1266754cd7bd5e7b062a"}]},"branch":"refs/heads/master"},"879c028b48c9bd4d30f0c24e758cd34a22bac057":{"kind":"REWORK","_number":4,"created":"2024-04-04 04:12:19.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/4"}}},"commit":{"parents":[{"commit":"d06ac93054d0d850e544f9f600915e26b072227d","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d06ac93054d0d850e544f9f600915e26b072227d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-04 04:09:22.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/879c028b48c9bd4d30f0c24e758cd34a22bac057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/879c028b48c9bd4d30f0c24e758cd34a22bac057"}]},"branch":"refs/heads/master"},"7bdb665d8498f2d3409db275d32b0ce84dd51fa5":{"kind":"REWORK","_number":5,"created":"2024-04-04 05:59:58.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/5"}}},"commit":{"parents":[{"commit":"d06ac93054d0d850e544f9f600915e26b072227d","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d06ac93054d0d850e544f9f600915e26b072227d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-04-04 05:59:42.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7bdb665d8498f2d3409db275d32b0ce84dd51fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7bdb665d8498f2d3409db275d32b0ce84dd51fa5"}]},"branch":"refs/heads/master"},"87c1b3e723d8f5e0c25fabcba7a30921a75761a8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-11-11 04:31:46.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/6"}}},"commit":{"parents":[{"commit":"4e9e4ab8a347dd73173385b74e208244f9fc8921","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e9e4ab8a347dd73173385b74e208244f9fc8921"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 04:31:13.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/87c1b3e723d8f5e0c25fabcba7a30921a75761a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/87c1b3e723d8f5e0c25fabcba7a30921a75761a8"}]},"branch":"refs/heads/master"},"c0fbc61d675dbeabd4d195ba6b05e084c265d6f1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-11-11 05:50:44.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/7"}}},"commit":{"parents":[{"commit":"926b9dd11d89e8659427afa551857907b54c0aea","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/926b9dd11d89e8659427afa551857907b54c0aea"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 05:50:05.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c0fbc61d675dbeabd4d195ba6b05e084c265d6f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c0fbc61d675dbeabd4d195ba6b05e084c265d6f1"}]},"branch":"refs/heads/master"},"d85259204deea6f9461c7bdd7eafe0ad88e1f166":{"kind":"REWORK","_number":8,"created":"2024-11-11 09:10:06.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/8"}}},"commit":{"parents":[{"commit":"fe64ba0bd5833fd859292673779ac90c09692d81","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe64ba0bd5833fd859292673779ac90c09692d81"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-11 09:09:19.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d85259204deea6f9461c7bdd7eafe0ad88e1f166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d85259204deea6f9461c7bdd7eafe0ad88e1f166"}]},"branch":"refs/heads/master"},"76bfdf6f6abc44b20d5bcbfa330dba07363fe6c3":{"kind":"REWORK","_number":9,"created":"2024-11-12 08:55:00.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/9"}}},"commit":{"parents":[{"commit":"fe64ba0bd5833fd859292673779ac90c09692d81","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe64ba0bd5833fd859292673779ac90c09692d81"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-12 08:52:30.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76bfdf6f6abc44b20d5bcbfa330dba07363fe6c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76bfdf6f6abc44b20d5bcbfa330dba07363fe6c3"}]},"branch":"refs/heads/master"},"232c00ef976f46e64c588ee69737ad32f4d96941":{"kind":"NO_CHANGE","_number":10,"created":"2024-11-12 09:01:54.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/10"}}},"commit":{"parents":[{"commit":"b453991c852d8816f18668883c57bd08fdcd2cf5","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b453991c852d8816f18668883c57bd08fdcd2cf5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-12 09:01:29.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/232c00ef976f46e64c588ee69737ad32f4d96941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/232c00ef976f46e64c588ee69737ad32f4d96941"}]},"branch":"refs/heads/master"},"c90ee68c3dc9c5ce63615d6692014868638f7284":{"kind":"REWORK","_number":11,"created":"2024-11-12 09:13:37.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/11"}}},"commit":{"parents":[{"commit":"b453991c852d8816f18668883c57bd08fdcd2cf5","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b453991c852d8816f18668883c57bd08fdcd2cf5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-12 09:11:31.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c90ee68c3dc9c5ce63615d6692014868638f7284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c90ee68c3dc9c5ce63615d6692014868638f7284"}]},"branch":"refs/heads/master"},"fc4a4d10cdc43cbf1a111fca5c56316aae240a9e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-11-13 03:06:03.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/12"}}},"commit":{"parents":[{"commit":"be13b041020e6d75c409b8421fc144fa0e867ca7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/be13b041020e6d75c409b8421fc144fa0e867ca7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-13 03:04:46.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fc4a4d10cdc43cbf1a111fca5c56316aae240a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fc4a4d10cdc43cbf1a111fca5c56316aae240a9e"}]},"branch":"refs/heads/master"},"103a8797984dbcdcdb2466181b77cfb341217e1a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-11-21 05:07:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/13"}}},"commit":{"parents":[{"commit":"8480a283131843bddc07489c06b16f4d5cb565c0","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8480a283131843bddc07489c06b16f4d5cb565c0"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-21 05:03:50.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/103a8797984dbcdcdb2466181b77cfb341217e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/103a8797984dbcdcdb2466181b77cfb341217e1a"}]},"branch":"refs/heads/master"},"e2ca4fed9b1f203d84cb5743c8503d04c18b3a54":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-11-25 07:12:09.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/14"}}},"commit":{"parents":[{"commit":"a31647e5ef2c1425b7b5b590037c36546650b90a","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a31647e5ef2c1425b7b5b590037c36546650b90a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-25 07:11:49.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e2ca4fed9b1f203d84cb5743c8503d04c18b3a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e2ca4fed9b1f203d84cb5743c8503d04c18b3a54"}]},"branch":"refs/heads/master"},"1422fc1f522ed2f7a2d9acc8e627128eb379d7b2":{"kind":"REWORK","_number":15,"created":"2024-11-25 10:10:34.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/15"}}},"commit":{"parents":[{"commit":"a31647e5ef2c1425b7b5b590037c36546650b90a","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a31647e5ef2c1425b7b5b590037c36546650b90a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-11-25 10:09:34.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1422fc1f522ed2f7a2d9acc8e627128eb379d7b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1422fc1f522ed2f7a2d9acc8e627128eb379d7b2"}]},"branch":"refs/heads/master"},"c5e61327ef10ba99ea0a98c0535f92eb12a5bdcb":{"kind":"REWORK","_number":16,"created":"2024-12-02 08:20:21.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/16"}}},"commit":{"parents":[{"commit":"43a6bbf778852590152bea5a1404da457cc6e2d4","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43a6bbf778852590152bea5a1404da457cc6e2d4"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-02 08:18:05.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5e61327ef10ba99ea0a98c0535f92eb12a5bdcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5e61327ef10ba99ea0a98c0535f92eb12a5bdcb"}]},"branch":"refs/heads/master"},"561b41e20e2948cd056a594dc019c47197bc0bb0":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-12-09 06:11:57.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/17"}}},"commit":{"parents":[{"commit":"6ad27d975a3ef5a666c6581751a4cf7fe62d6ab5","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6ad27d975a3ef5a666c6581751a4cf7fe62d6ab5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-02 10:12:22.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/561b41e20e2948cd056a594dc019c47197bc0bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/561b41e20e2948cd056a594dc019c47197bc0bb0"}]},"branch":"refs/heads/master"},"eb8769ea894df69b5948ee76db979b1767850717":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-12-19 05:58:24.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/18"}}},"commit":{"parents":[{"commit":"3128dde9e5a4e17b79755056e6bc926f19d530b6","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3128dde9e5a4e17b79755056e6bc926f19d530b6"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-19 05:57:05.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb8769ea894df69b5948ee76db979b1767850717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb8769ea894df69b5948ee76db979b1767850717"}]},"branch":"refs/heads/master"},"27ab0f68f268730090721199fbe75755898c72f6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":19,"created":"2024-12-27 11:06:22.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/19"}}},"commit":{"parents":[{"commit":"2e7bab65c4bc16288cff80efcba0e4493156a540","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e7bab65c4bc16288cff80efcba0e4493156a540"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-27 11:06:02.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/27ab0f68f268730090721199fbe75755898c72f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/27ab0f68f268730090721199fbe75755898c72f6"}]},"branch":"refs/heads/master"},"8f2c3d92bcbed5c1f59a4821211e09d3be358d11":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-12-27 14:53:30.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/20"}}},"commit":{"parents":[{"commit":"264366c4b218299cdfaa442368e3c82264b3a1b7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/264366c4b218299cdfaa442368e3c82264b3a1b7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-27 14:52:48.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f2c3d92bcbed5c1f59a4821211e09d3be358d11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f2c3d92bcbed5c1f59a4821211e09d3be358d11"}]},"branch":"refs/heads/master"},"62fe05f0bdfbd8b39f493875a7b5a5386492cddc":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-12-28 06:59:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/21"}}},"commit":{"parents":[{"commit":"6478d11fb0a9b24ba156f6aac2f90b7503c7eb11","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6478d11fb0a9b24ba156f6aac2f90b7503c7eb11"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2024-12-28 06:59:39.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/62fe05f0bdfbd8b39f493875a7b5a5386492cddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/62fe05f0bdfbd8b39f493875a7b5a5386492cddc"}]},"branch":"refs/heads/master"},"83fc5d39e250ef50751343fd2a4edfaf03c45460":{"kind":"NO_CHANGE","_number":22,"created":"2025-01-07 07:47:38.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/22"}}},"commit":{"parents":[{"commit":"ad86ff9dad926c7a22775d7b5db578346a59845e","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad86ff9dad926c7a22775d7b5db578346a59845e"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-01-07 07:47:22.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/83fc5d39e250ef50751343fd2a4edfaf03c45460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/83fc5d39e250ef50751343fd2a4edfaf03c45460"}]},"branch":"refs/heads/master"},"85e76b5704bb6606fe1aa0412b6932dfe59bdc3c":{"kind":"REWORK","_number":23,"created":"2025-03-05 05:19:19.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/23"}}},"commit":{"parents":[{"commit":"ad86ff9dad926c7a22775d7b5db578346a59845e","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad86ff9dad926c7a22775d7b5db578346a59845e"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-05 05:18:12.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/85e76b5704bb6606fe1aa0412b6932dfe59bdc3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/85e76b5704bb6606fe1aa0412b6932dfe59bdc3c"}]},"branch":"refs/heads/master"},"9add80aaa9af1725bcd80f2d2f4fab87780fda2c":{"kind":"REWORK","_number":24,"created":"2025-03-19 08:34:25.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/24"}}},"commit":{"parents":[{"commit":"d19e8d64de45aaddc5b62cdfe0aa0665f23eac3b","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d19e8d64de45aaddc5b62cdfe0aa0665f23eac3b"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 08:33:24.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9add80aaa9af1725bcd80f2d2f4fab87780fda2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9add80aaa9af1725bcd80f2d2f4fab87780fda2c"}]},"branch":"refs/heads/master"},"b19d671aa4194585f9e74b5995857195d6b28050":{"kind":"REWORK","_number":25,"created":"2025-03-21 11:03:04.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/25"}}},"commit":{"parents":[{"commit":"5359e6030822b0dd8e2110ce3c294be2ebec9f5a","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5359e6030822b0dd8e2110ce3c294be2ebec9f5a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-21 10:51:08.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b19d671aa4194585f9e74b5995857195d6b28050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b19d671aa4194585f9e74b5995857195d6b28050"}]},"branch":"refs/heads/master"},"48828c97c55af99ff0386c7d2d66f286286a630f":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-04-03 10:46:00.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/26"}}},"commit":{"parents":[{"commit":"e3b3ad8919faa2b63712eebae93080fcd89815e4","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3b3ad8919faa2b63712eebae93080fcd89815e4"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-03 10:44:12.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48828c97c55af99ff0386c7d2d66f286286a630f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/48828c97c55af99ff0386c7d2d66f286286a630f"}]},"branch":"refs/heads/master"},"45145b23b2fb95cf481766d4a1dc78389f105355":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-04-30 10:14:34.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/27"}}},"commit":{"parents":[{"commit":"40ec233e64f69b1354b4b9b31401f461b365a2ef","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/40ec233e64f69b1354b4b9b31401f461b365a2ef"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-30 10:11:42.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/45145b23b2fb95cf481766d4a1dc78389f105355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/45145b23b2fb95cf481766d4a1dc78389f105355"}]},"branch":"refs/heads/master"},"abffdcf555f9a2c3d04581e7e57b3036dbc32f1c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":28,"created":"2025-07-29 09:10:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/28"}}},"commit":{"parents":[{"commit":"ad98e52609d4817e45e21daf20ba2adc7209769e","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ad98e52609d4817e45e21daf20ba2adc7209769e"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-07-29 09:09:47.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abffdcf555f9a2c3d04581e7e57b3036dbc32f1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/abffdcf555f9a2c3d04581e7e57b3036dbc32f1c"}]},"branch":"refs/heads/master"},"f7a450e3a7796a5673df8708861eab5879a2b350":{"kind":"REWORK","_number":29,"created":"2025-09-18 04:49:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/29"}}},"commit":{"parents":[{"commit":"28e3dc1dada5a0b0763744388fb3e913c54a94e6","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/28e3dc1dada5a0b0763744388fb3e913c54a94e6"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 04:42:30.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f7a450e3a7796a5673df8708861eab5879a2b350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f7a450e3a7796a5673df8708861eab5879a2b350"}]},"branch":"refs/heads/master"},"8e89a4104e6d34d2e4c4699b117ac8e627d7207b":{"kind":"REWORK","_number":30,"created":"2025-09-18 09:23:14.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/30"}}},"commit":{"parents":[{"commit":"44db5e7a62a8621d7c0da0fb265ed8873114dd37","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44db5e7a62a8621d7c0da0fb265ed8873114dd37"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 08:43:49.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e89a4104e6d34d2e4c4699b117ac8e627d7207b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e89a4104e6d34d2e4c4699b117ac8e627d7207b"}]},"branch":"refs/heads/master"},"47e5936619b89a3105623d0d5a1ad489e19ee312":{"kind":"REWORK","_number":31,"created":"2025-09-19 01:37:23.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/31"}}},"commit":{"parents":[{"commit":"44db5e7a62a8621d7c0da0fb265ed8873114dd37","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44db5e7a62a8621d7c0da0fb265ed8873114dd37"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-19 01:23:20.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47e5936619b89a3105623d0d5a1ad489e19ee312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47e5936619b89a3105623d0d5a1ad489e19ee312"}]},"branch":"refs/heads/master"},"09af9a73a00b92dfdca493a7131a9c3175b3c077":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2025-12-26 08:39:20.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/32"}}},"commit":{"parents":[{"commit":"a573b63b1d9ec23602ef30576eb0ee37e798c600","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a573b63b1d9ec23602ef30576eb0ee37e798c600"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:25:42.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09af9a73a00b92dfdca493a7131a9c3175b3c077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09af9a73a00b92dfdca493a7131a9c3175b3c077"}]},"branch":"refs/heads/master"},"28a56aa1060a5b9d01d96185e7e0485335a37881":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2025-12-26 08:50:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/33"}}},"commit":{"parents":[{"commit":"d2204c5b79222858cdf20da850e8ea19291cadf7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d2204c5b79222858cdf20da850e8ea19291cadf7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:50:21.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/28a56aa1060a5b9d01d96185e7e0485335a37881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/28a56aa1060a5b9d01d96185e7e0485335a37881"}]},"branch":"refs/heads/master"},"911576632d6bef6782d8af00c26b2f712d569b17":{"kind":"REWORK","_number":34,"created":"2025-12-26 09:19:44.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/34"}}},"commit":{"parents":[{"commit":"d2204c5b79222858cdf20da850e8ea19291cadf7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d2204c5b79222858cdf20da850e8ea19291cadf7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 09:15:29.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/911576632d6bef6782d8af00c26b2f712d569b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/911576632d6bef6782d8af00c26b2f712d569b17"}]},"branch":"refs/heads/master"},"4fc15e141c452449221a0048585f2376014b4bcf":{"kind":"REWORK","_number":35,"created":"2026-01-05 09:09:45.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/35"}}},"commit":{"parents":[{"commit":"2c0d4c4aebb2a2538441d606c9eb43b6927a3f39","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2c0d4c4aebb2a2538441d606c9eb43b6927a3f39"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-05 08:49:27.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc15e141c452449221a0048585f2376014b4bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4fc15e141c452449221a0048585f2376014b4bcf"}]},"branch":"refs/heads/master"},"047036671e6cf46717c5871cafdc75f31fed0e7b":{"kind":"REWORK","_number":36,"created":"2026-01-06 08:30:42.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/36"}}},"commit":{"parents":[{"commit":"ed0dee904916fd0e6356384018a6d0a76c8a229a","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ed0dee904916fd0e6356384018a6d0a76c8a229a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-06 08:13:03.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/047036671e6cf46717c5871cafdc75f31fed0e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/047036671e6cf46717c5871cafdc75f31fed0e7b"}]},"branch":"refs/heads/master"},"f28803f4a3685859cad8fe787b4ab83bd22e70fa":{"kind":"NO_CODE_CHANGE","_number":37,"created":"2026-01-15 09:35:57.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/37"}}},"commit":{"parents":[{"commit":"c1949b0a73411765eddf5c050968519d3c9ce588","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1949b0a73411765eddf5c050968519d3c9ce588"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-15 09:31:46.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/5)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f28803f4a3685859cad8fe787b4ab83bd22e70fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f28803f4a3685859cad8fe787b4ab83bd22e70fa"}]},"branch":"refs/heads/master"},"c5dcf5f53928ea7af76c973e5c753fc23d676be4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":38,"created":"2026-01-19 14:48:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/38"}}},"commit":{"parents":[{"commit":"3ed02c36cfa5a5fe4db165f1f26dab25cefce924","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ed02c36cfa5a5fe4db165f1f26dab25cefce924"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-19 14:39:54.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5dcf5f53928ea7af76c973e5c753fc23d676be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5dcf5f53928ea7af76c973e5c753fc23d676be4"}]},"branch":"refs/heads/master"},"ce92bdb9b537e1124b1e739d39b27dbbdac5542a":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-01-20 07:45:37.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/39"}}},"commit":{"parents":[{"commit":"b458ee1b1bc9d3f114873c431a5f22ba3f854740","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b458ee1b1bc9d3f114873c431a5f22ba3f854740"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 07:43:26.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce92bdb9b537e1124b1e739d39b27dbbdac5542a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce92bdb9b537e1124b1e739d39b27dbbdac5542a"}]},"branch":"refs/heads/master"},"229b66074b4eea039ebbb24b2d91308bb17076e7":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2026-01-20 11:23:17.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/40"}}},"commit":{"parents":[{"commit":"024b618ae0fec4d80a7e4e1dc55e8672732d95c4","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/024b618ae0fec4d80a7e4e1dc55e8672732d95c4"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 10:44:28.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/229b66074b4eea039ebbb24b2d91308bb17076e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/229b66074b4eea039ebbb24b2d91308bb17076e7"}]},"branch":"refs/heads/master"},"e9cb06e1a3bcbb2f946e83b64557192aed55cfe5":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2026-01-26 08:24:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/40/914840/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/41"}}},"commit":{"parents":[{"commit":"43263a15f722fe01fbc2b941bba61fe684ac19c7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/43263a15f722fe01fbc2b941bba61fe684ac19c7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-26 06:46:25.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e9cb06e1a3bcbb2f946e83b64557192aed55cfe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e9cb06e1a3bcbb2f946e83b64557192aed55cfe5"}]},"branch":"refs/heads/master"},"3637e7eae18ba47b8d55eff806f1af9862f76ec0":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2026-06-15 01:10:43.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"ref":"refs/changes/40/914840/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/42"}}},"commit":{"parents":[{"commit":"09c8f167ca737eb6c0ff9a65af5274a32dae1b3a","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09c8f167ca737eb6c0ff9a65af5274a32dae1b3a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-06-15 01:09:35.000000000","tz":540},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3637e7eae18ba47b8d55eff806f1af9862f76ec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3637e7eae18ba47b8d55eff806f1af9862f76ec0"}]},"branch":"refs/heads/master"},"58706c39be9522fbc60763b2906b070a815c6cd5":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2026-07-11 21:34:43.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/40/914840/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/43"}}},"commit":{"parents":[{"commit":"3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-07-11 21:34:43.000000000","tz":0},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/58706c39be9522fbc60763b2906b070a815c6cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/58706c39be9522fbc60763b2906b070a815c6cd5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"09c8f167ca737eb6c0ff9a65af5274a32dae1b3a","ours":"3637e7eae18ba47b8d55eff806f1af9862f76ec0","theirs":"3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7","merge_strategy":"recursive","contains_conflicts":false}},"95586df45f2a99d728205c14de78a0e9b0671d2f":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2026-07-17 05:36:29.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/40/914840/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/914840/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/914840/44"}}},"commit":{"parents":[{"commit":"6862df61a9d5619379864939e75f0886af986cb5","subject":"Implement ConsulOperator as a service broker to support HTTP Driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6862df61a9d5619379864939e75f0886af986cb5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2024-04-02 01:56:43.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-07-17 05:36:29.000000000","tz":0},"subject":"Add HTTP driver to oslo.messaging.drivers","message":"Add HTTP driver to oslo.messaging.drivers\n\nThis commit initiates the implementation of the HTTP driver class.\nIt also sets up the interfaces for HTTPListener and HTTPClient,\nwhich will be implemented in subsequent stages.\n\nPartially-implements: blueprint oslo-http-driver (2/6)\nChange-Id: Ic6c01c7eafb20527e061a8c98225c6f6c9439788\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/95586df45f2a99d728205c14de78a0e9b0671d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/95586df45f2a99d728205c14de78a0e9b0671d2f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"3d1c4af30bb191d5013cc0f65fd4b1d52543d4d7","ours":"58706c39be9522fbc60763b2906b070a815c6cd5","theirs":"6862df61a9d5619379864939e75f0886af986cb5","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
