)]}'
{"id":"openstack%2Foslo.messaging~917809","triplet_id":"openstack%2Foslo.messaging~master~I323f2d387268950bb00dd25fc762f1f09d012c5e","project":"openstack/oslo.messaging","branch":"master","topic":"excludes-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-01 04:54:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I323f2d387268950bb00dd25fc762f1f09d012c5e","subject":"Remove old excludes","status":"MERGED","created":"2024-05-01 02:36:01.000000000","updated":"2024-05-01 08:27:31.000000000","submitted":"2024-05-01 08:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917809-excludes-cleanup","meta_rev_id":"22fbfea394e25068d5ee34f0ab80cad770370bd3","_number":917809,"virtual_id_number":917809,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-01 08:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-05-01 07:05:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-05-01 07:05:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 04:47:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1010afbfd7974f66c8a9d365985fbd535c57cd31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 02:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1bc76588a07ffe914842c6233bbc1f28cf8455c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 04:47:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1329291091504a14828b076c42c1c49f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/84a15b5e51fe4672916196f1d45daf46 : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/481a00552e1d40c18a91897a5958c1af : SUCCESS in 1h 46m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e66eeb104963448a9b114d7288c4f39d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11d136d0ea848a39533b3ad730805fe : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/657b3ccbd92a4edcb17b7cb35d4e7737 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39b5cc3e44ac4e5d9daafb1a4083e47a : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71a2606ca54a4ffeb4cdfdf6ec18fdea : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22f881f713034cb180a3549bd4abd76a : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6600f670394ccda5dfb0efbc1f2048 : SUCCESS in 3m 31s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/03bda295e863475d9beb7d46977b46d8 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/ea19f67b91de4e4388fc35bb52c20731 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/65897204894d4b288b529fc75c831c66 : FAILURE in 1h 31m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d2b66700fe5244128d27ad79600905e4 : FAILURE in 35m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8b40f952c5024d5fb4e984ba664caca6 : SUCCESS in 1h 52m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/38872d70aa9342b5a8fa87420fa1407c : SUCCESS in 1h 09m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cc40f2092f845b8afa434ec9fbe3d3e : SUCCESS in 1h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"486e8be215758dfab1eb2d166d5d7f922b5ae701","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 04:54:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"789cdef35b64edba986b3e670625f27af5dcb455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 06:47:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb04ae5ccef467f92447281c72b7929\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07869b727f5c45159d5c20f7ab81fd91 : SUCCESS in 1m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65ebf6cd3af34a99b4b73620cf9e1a72 : SUCCESS in 1h 20m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01357fca0df54506bf2d3d19fb9399ff : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb48734c150944a0b81168d470b50c71 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f2e5e1b2bb41e4a25cd377b733f507 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f955ec2b4154223977c43a9956af5c0 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/728a6d9e126c491a85b4eb5d415d0c92 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a46cb0e89cd433e8688386212bea071 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e8bf7bae0e4f9c8e7d8a8917c779ac : SUCCESS in 4m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/02253d100b174708b7c7f3119b281b51 : SUCCESS in 6m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5390546409b0440daf0d26f286f03880 : FAILURE in 6m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b89ded9ef6b3466f9bc89fd6c68e2438 : SUCCESS in 1h 36m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/00c7fe33025643c2abecdc8f95a1964d : FAILURE in 20m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9d471c82a9b347c595c8555b526ee7a8 : SUCCESS in 1h 47m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/39dd6800599640af99cdc3b19f07b8c9 : SUCCESS in 1h 07m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd6cb6f89610457abb707f46855f3eaa : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40cd2b554b6695c87a21195e6f76dc1c9ce7e12b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 07:05:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"922d5269ff9a64df2b5f931312b018e86295ef12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 07:05:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"530cf62d727528ccd462165ef57ca526ab081789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 08:26:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f116d94531d94a81a4fe2cf4de2199cb\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/73580b35e47b4f63afeab4f388fbeed5 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbaf23ddf8c94029a573283af269fd4f : SUCCESS in 1h 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bcbc5558ee24cb8afc03382acae9368 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9097e47ffe54907bdcb99c7a530ca25 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46094b5815254e33b28f5afcabe6f0d1 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecdae76c52b42799936da2b76393595 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d48ccb5b0d244bae9eaf282d7b475233 : SUCCESS in 2m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6ecba1a3a89b446cb319568793231342 : SUCCESS in 5m 17s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9e66baa2af88434cb5f435d5befb757f : SUCCESS in 1h 20m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0233b0e139937d491b6d1a4a97b429718bfc9d43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 08:26:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"22fbfea394e25068d5ee34f0ab80cad770370bd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 08:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37244162c21644bc959e3c1bdd2ab2e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091fd94703a64b0eaabdd9e08d804900 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7853212388a34c14bbed9e65f0ed552a : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5dbbcf323902b799bfe84e4ef854e7db167cb6c3","revisions":{"5dbbcf323902b799bfe84e4ef854e7db167cb6c3":{"kind":"REWORK","_number":1,"created":"2024-05-01 02:36:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/917809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/09/917809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/917809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/917809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/917809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/09/917809/1"}}},"commit":{"parents":[{"commit":"986cd4ab240ddeb5022513f6eccc4cb48e8d3de0","subject":"Merge \"Fix incorrect desc of rabbit_stream_fanout option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/986cd4ab240ddeb5022513f6eccc4cb48e8d3de0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 02:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 02:35:50.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nChange-Id: I323f2d387268950bb00dd25fc762f1f09d012c5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5dbbcf323902b799bfe84e4ef854e7db167cb6c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5dbbcf323902b799bfe84e4ef854e7db167cb6c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
