)]}'
{"id":"openstack%2Foslo.messaging~923337","triplet_id":"openstack%2Foslo.messaging~master~Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1847747","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-16 13:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2024-07-16 14:43:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5","subject":"Add RPC incoming and reply log","status":"MERGED","created":"2024-07-03 01:02:13.000000000","updated":"2024-07-16 14:44:20.000000000","submitted":"2024-07-16 14:43:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923337-bug/1847747","meta_rev_id":"b037d1d427845cdb6b36a3c0a0d81caa5c582334","_number":923337,"virtual_id_number":923337,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-16 14:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-07-16 13:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-07-12 11:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-07-16 13:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 02:58:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 13:33:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-07-12 11:32:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-07-16 13:04:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"c32b43bfab27c9b79cd8e0504c0af2ae4080f4fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-03 01:02:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e369063cf577da405edf33a3f58f153d3b0448ed","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-03 01:03:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47f44e31e7fc97d601c99b8960498bc5d35c6e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 02:58:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a444aa4ac2f540f8ae57cc3f9912aef3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d75bd1f10f0465a9fba45df5cc3ff50 : SUCCESS in 1h 41m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044b220590614d3f971560ead9423d81 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be82d2ae0984cd6957f2d9fa0f95a4f : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a921338705b4e5290bc38db54bd7959 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/090661b91cf94ef0b8207e409a584bc8 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18518bd8caad42a4b886d69a29dd0cad : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3453d5ff5f64733a3bcb08930b60836 : SUCCESS in 3m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/2ec0884d60574584a1841f616b95dd5c : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0ec45c451892474a9a124077e32c2149 : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e37484d322444f7f98e7eb4eefeaf3be : SUCCESS in 1h 34m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7a39794e16bc46b78322ac00bb619ae2 : FAILURE in 27m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/779ce3a95ef44712ad50912efe0f66df : SUCCESS in 1h 50m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/64c719f471fe441cbc2e7cf8d0f30407 : SUCCESS in 1h 01m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/068fcbb60dca4161884774ed89b80e21 : SUCCESS in 1h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bccb862700c73840e902f7b443d81b2fa7dbfbd","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-04 00:46:01.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1462c55774023a552a29e6c17c6dc26a2c6a1a6d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-04 13:33:21.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"977bc512b705b919c6f656d2aa982f87b35c3410","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-05 01:12:32.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8d6cc8c2e95f10a8f1f60cb9a9a0822853cf603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-05 01:14:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"64d1b0c18df707ad605de75a3a058ff2c7a78dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-05 01:36:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"013b88beab34d93ce6f4043cc62ff150ab12f721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 03:47:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72223cdecc5f4070b1d48104ec8f7f74\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1e5e439952d4d3b8bb628bf13eca73c : SUCCESS in 1h 39m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341a4318896945199bc93aa740900d0d : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df37a6a0e84c4443b35d2e9ebabe74ac : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4538f354e03545feb9de4f2636e9060a : FAILURE in 9m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2b67611488545988e02d42e01f0dc89 : FAILURE in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6b22d311d1349258d65404ba4fa4c33 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21bdb7063a945e9966b642ab337b8a3 : SUCCESS in 3m 35s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b03f0fdcd4b040bd9d40e2786de8e541 : FAILURE in 12m 20s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9f95bcd10b3744fcbc8a8f41c0b99cf1 : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5fa7f88e722145de83e4b96e0f3d2630 : SUCCESS in 1h 44m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5ccb68ea49b143689bb4512290d552b4 : FAILURE in 28m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/32217d32ffcf42a0946c3562840ba2e7 : SUCCESS in 1h 38m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bf4d5686f93840eab9afc796d74ee953 : SUCCESS in 1h 00m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdd61ab56402407da01728309be23445 : SUCCESS in 1h 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a50177ab6e0b6ebdf534b707ff4e706a5d650a15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-05 09:11:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f210c8f9616c9e0745584633e28d867efe328f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 11:00:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c19e02ef6d445a7a9c1c1b73319b64f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b8d43b2e1014eca999875d30a9e1880 : SUCCESS in 1h 47m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15bb8c2a7afa4e1b9b7afa3207c6abf1 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31c7c3d23604d37b6a89ec8af413ccf : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/334db25d030e467d9ad9e9a08afff393 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/341d668f5f344f7d9ffba8a4f05c47f5 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/348ffa65556941d38dd4e2eaece57448 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4030f0b17b114d2f9fde0ddad68227ca : SUCCESS in 3m 25s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e18d71317d7a439a8e3bf0320963826a : FAILURE in 5m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/abb0d6b5832a4553b15afdc9b84c1432 : FAILURE in 6m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8864739afd8e42eaba9a859f63dcc46f : SUCCESS in 1h 44m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3d34d7d3c7ad46c0aaaf73d9c3a4bfb5 : FAILURE in 26m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b5ebf22387744b1da2710c0824384c9d : SUCCESS in 1h 36m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f68b95bbb50a43458bf424d4ce9b8951 : SUCCESS in 1h 09m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/88035ff1f8ec420a929477d9d7554efe : SUCCESS in 1h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"12e7ee14a867ccd430f01a094ea6f49e2c94bd67","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-08 02:05:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8bb10887594f95bc78d8b19c6e9d01e600450b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 04:26:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84a270c730b24a55bce9abe7893c9068\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed9f660633944e7eb8bf959da78d8d06 : SUCCESS in 2h 05m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b01e7e04eb14608b0424a11284e3eeb : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72fdb8e0f3043958c57d24c3d75a5e1 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9c6a148a809451faa9d974416d0e274 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a86513492d4a4310b55350739719e912 : SUCCESS in 7m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1ec596f7e47457cac9675a19c04197f : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf54b84cd5b74d20bd803500623aa699 : SUCCESS in 3m 57s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8957ec6540a64c0fafbb42cf4924404a : FAILURE in 5m 38s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/805cb19459cf4e6689a5afeb2e7b171e : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/014f483e9d07491897d7c5648b8e0a51 : SUCCESS in 1h 34m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/68408b202f0d4ed584755c84c14d6a09 : FAILURE in 26m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8934ec75ecec4faa96a5923ad5ecf7f4 : SUCCESS in 1h 18m 13s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e7af131819654b539cde26ea95b5f0f3 : SUCCESS in 54m 23s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a335162660034075b2a63977894fd583 : SUCCESS in 1h 16m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1693b29d21e487eb8ce7039ab84f941c38d8473b","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-09 00:58:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"65363078500ddeea0a8c05058b0dff97dcca9812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 02:48:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e79702d5fb2346cfb443ddd5a3cd6e5f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f420d015574eceb3b66ca013d0c092 : SUCCESS in 1h 40m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18649f30c01e49f4a5ee573b491250f4 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da15daa905d54a8eba52acbc903c0edc : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91c4092771444f87975af17248901ecb : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5aa586dc4f334d1383aa08fecca43ca7 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6971f136bc814db69f866d2c01dcfc4e : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a4ff1f9fda4fc5803f670a3dc59793 : SUCCESS in 5m 57s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/5197da8961a741cbaec75461ce1ab25a : FAILURE in 8m 24s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/763bf39e6d394ad78db74f5ac9fc9f98 : FAILURE in 8m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c0d07bcdb64a4cb08b9decec43f174f8 : SUCCESS in 1h 33m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/13f0644e862f483da577ce1217fa2bfa : FAILURE in 15m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ea01fc46abfb40f9a7585c587f845a71 : SUCCESS in 1h 44m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5350ca0395ad49f88c86db2b99d27f92 : SUCCESS in 51m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/10d4ab6713f741e299fb3d628372b1dd : SUCCESS in 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99bad7a1432fcd04a4a564bb4ebd6462f8f4b498","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-09 06:57:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"dfc65c03cf25a3f33f6e59c79c87e55240314af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 08:47:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/056f556b17c44f8685ea5d8d1b3fa8d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/224ec34a71d14db699f3bc81fa14989c : SUCCESS in 1h 29m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/378aaf47091748beacf3ef1a867753e9 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952daa8730f44fb69c5b12a149c18b09 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899651d3eed7456bb5763d50e4f5fc06 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21a2cea12c8441ec8502120c25e08168 : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0200ca93b1e3409dbec1d081b529c669 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc315a54a24e460f9339c05b8a68cdba : SUCCESS in 4m 45s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/57d32d6927fa4502b5ebb336b7947322 : FAILURE in 7m 01s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8f4c7f61850e4687a59bb298068b9b7e : FAILURE in 9m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9ce6932460404fc78806103b51420bda : SUCCESS in 1h 37m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e6ab79e3b2d644cb942c2009c1ae72c5 : FAILURE in 28m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5b94af804b18451fb914609623e6460f : SUCCESS in 1h 34m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d32da3da870c47bbafbae2e48be04972 : SUCCESS in 1h 01m 12s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6a8388dee0e441e8b2c210541f10339 : SUCCESS in 1h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0c3afd4202add4ad3edb250fd23ff22ba43afd9","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-10 06:35:56.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f552a31e1f3095117006b9c5e659b01dcef2a2b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 07:27:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"16622b6d0fd13f0d368e029da657df301784d869","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-12 09:02:14.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"73f7ec9cd562de93710ab65eb612e91f71f4532e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 09:05:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bed006e16eca44d09227ebb943b15643\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca700ac10a5b4d40ab2d81ee3317e6ee : SUCCESS in 1h 25m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c5d0293ad1f4b92bf96402508df4a0f : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68c87c990304a29b35c8e9ff7022106 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b39d89eac80b4bf8974285a5daca98fc : FAILURE in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6e56f1113eb41edbf7455fa42cc0723 : FAILURE in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f65dc10219646e5ad075d141ad25351 : FAILURE in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a36daabe37f4de0a5a74ff8035d246c : SUCCESS in 4m 55s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/109d31f18cbe4ec08808a1182ff06f02 : SUCCESS in 5m 47s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/b60e3809ecd44ce58b3e1dfef58fc134 : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ce137225a1524e7c931799fa5e22c162 : SUCCESS in 1h 35m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/26b4d10390064eb68c42e419e57adb38 : FAILURE in 28m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/40f2da45025b476792a5447ffb954caa : SUCCESS in 1h 28m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c269b6c0c0fa4338bc9b3d9f7ffa4752 : SUCCESS in 1h 00m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/058ee1b1ae024f14a39dc140b40f8ea8 : SUCCESS in 1h 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2a9c672be967ce81b973efd88700a7968610e283","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:08:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"97ca8acd83beff10ac5768ec01b4ff7284df614b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:10:55.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"703a68fea61dc74b7bf5b1949b6ea7845d2adb95","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-12 09:14:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6eacd3f070b48bf10b3bed4474f9cea7061cc8b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:19:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1a56e57dc525e5a0348418df539581c0f091574d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:22:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"997a31d8b7a84bb902637f3ae97ee91f1907e184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:36:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ff63074fe8afa9ceab57747052ed3f13f2773dcb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:37:10.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9139137fa325174d50fa62597d2f164e92438afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 09:40:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1fa04a871ccd61ba3cf05f7a347b60a675bd69cb","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2024-07-12 11:07:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a133da0f93cef08aac45ecb3082970e9a6eb8c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 11:19:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ad4f96e9344052accdcfaed33b638a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4953ee290be4277a705fb88f259803e : SUCCESS in 1h 37m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c75b9b680684f0db30ac00feb3f5d83 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdeea6cd6e848af872df2ecf426218f : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8761733de3724cf591de725724fffefb : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56bee1eab02f490c819f7a84a7b89be5 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/294e64bd2d254ad7879ee7db3af15420 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b54a58206b24f2896b77ad47c54cd81 : SUCCESS in 4m 38s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bfdd6c740ee84758948918130e02708d : SUCCESS in 5m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/adab748590264a2db597fa405bc2ef66 : FAILURE in 6m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/35ef188cbc344f4f9fcad5667c5472e4 : SUCCESS in 1h 37m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f2d874ef265943c9869a823f290bb58a : FAILURE in 19m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1cc2969603fe4896b71d66c4341bf929 : FAILURE in 1h 18m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/40b65ddf079446f5828a64238614b407 : SUCCESS in 49m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/7afb99e6b92c42758bd7cdc3be936619 : SUCCESS in 1h 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0dcc8ccd6d1549ac4ff858404fca5b0addf3881d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-12 11:32:47.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ea49765283442535e18bb6a41a96c1abb2da1deb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-16 13:04:59.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"fff50872ae80b36a8a06db78c209517b718cbe5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 13:05:27.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a79f770f084d1681cea0d25b8f3c7109ee30ec8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:43:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835a0c3bb61741c7b8cbc8510115ed6f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3ddfacdef3f495fa10515a869cebf02 : SUCCESS in 1h 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93841d9ec4f4805b2e94d273ece8b66 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/786c9fee0e5c44569378d9c32bb9cb8a : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a834579b1cb04a7bae0831a5af96bc3d : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594c558b8050471c95da468482d221b7 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/8a628dab4a124eac967890f0f997826b : SUCCESS in 4m 47s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e54fa5315a7443c08f9c384f8b242eab : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":9},{"id":"fffd75334378d41c9b7c67c13d2568f149101922","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:43:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"b037d1d427845cdb6b36a3c0a0d81caa5c582334","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:44:20.000000000","message":"Patch Set 9:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c69f35ea35c42bbad1c9a15e51470c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070f18a5a46743a4a223ddfc242f35e0 : FAILURE in 57s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"104a63b9aa040dda567d356e598aad5f8734d3ec","revisions":{"46a66379847308233ad44b4dfffc760b1c2e4c2d":{"kind":"REWORK","_number":1,"created":"2024-07-03 01:02:13.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/37/923337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/1"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-03 01:00:03.000000000","tz":480},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the\nperformance, but give us more information about\nthe service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358\ndirectly revert the log due to the msg_id is not\naccessable for non-rabbit drivers. It\u0027s not the\nright way to make things better.\n\nSo add the log back with the attribute msg_id for base\nIncomingMessage class.\n\nCloses-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a66379847308233ad44b4dfffc760b1c2e4c2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a66379847308233ad44b4dfffc760b1c2e4c2d"}]},"branch":"refs/heads/master"},"2d6587f9c96152387f7844bab46c804ae0e3e21a":{"kind":"REWORK","_number":2,"created":"2024-07-05 01:14:25.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/37/923337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/2"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-05 01:12:05.000000000","tz":480},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the\nperformance, but give us more information about\nthe service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358\ndirectly revert the log due to the msg_id is not\naccessable for non-rabbit drivers. It\u0027s not the\nright way to make things better.\n\nSo add the log back with the attribute msg_id for base\nIncomingMessage class.\n\nCloses-Bug: 2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2d6587f9c96152387f7844bab46c804ae0e3e21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2d6587f9c96152387f7844bab46c804ae0e3e21a"}]},"branch":"refs/heads/master"},"f1ef8b48c72a8781848df04dfa3b4f77946b3526":{"kind":"REWORK","_number":3,"created":"2024-07-05 01:36:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/37/923337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/3"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-05 01:36:43.000000000","tz":480},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the\nperformance, but give us more information about\nthe service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358\ndirectly revert the log due to the msg_id is not\naccessable for non-rabbit drivers. It\u0027s not the\nright way to make things better.\n\nSo add the log back with the attribute msg_id for base\nIncomingMessage class.\n\nCloses-Bug: 2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1ef8b48c72a8781848df04dfa3b4f77946b3526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1ef8b48c72a8781848df04dfa3b4f77946b3526"}]},"branch":"refs/heads/master"},"3d28d72860a6c9c2353e643bd0c75de9a413a87a":{"kind":"REWORK","_number":4,"created":"2024-07-05 09:11:23.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/37/923337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/4"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-05 09:11:08.000000000","tz":480},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the\nperformance, but give us more information about\nthe service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358\ndirectly revert the log due to the msg_id is not\naccessable for non-rabbit drivers. It\u0027s not the\nright way to make things better.\n\nSo add the log back with the attribute msg_id for base\nIncomingMessage class.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3d28d72860a6c9c2353e643bd0c75de9a413a87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3d28d72860a6c9c2353e643bd0c75de9a413a87a"}]},"branch":"refs/heads/master"},"6b3268990f22db75360e8af2771160668c8c710a":{"kind":"REWORK","_number":5,"created":"2024-07-09 06:57:31.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/37/923337/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/5"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2024-07-09 01:14:36.000000000","tz":480},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the\nperformance, but give us more information about\nthe service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358\ndirectly revert the log due to the msg_id is not\naccessable for non-rabbit drivers. It\u0027s not the\nright way to make things better.\n\nSo add the log back with the attribute msg_id for base\nIncomingMessage class.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6b3268990f22db75360e8af2771160668c8c710a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6b3268990f22db75360e8af2771160668c8c710a"}]},"branch":"refs/heads/master"},"f62189ca932a6f04cde9fa2d5adc7ff4cea6324c":{"kind":"REWORK","_number":6,"created":"2024-07-12 07:27:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/923337/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/6"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 07:26:07.000000000","tz":540},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the performance, but give\nus more information about the service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358 directly revert\nthe log due to the msg_id is not accessable for non-rabbit drivers.\nIt\u0027s not the right way to make things better.\n\nSo add the log back with the attribute msg_id for base IncomingMessage\nclass. msg_id is set only when the messaging backend assigns id.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f62189ca932a6f04cde9fa2d5adc7ff4cea6324c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f62189ca932a6f04cde9fa2d5adc7ff4cea6324c"}]},"branch":"refs/heads/master"},"a47155ea736ede206745a85fb4cd7df620fee056":{"kind":"REWORK","_number":7,"created":"2024-07-12 09:08:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/923337/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/7"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 09:07:57.000000000","tz":540},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the performance, but give\nus more information about the service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358 directly revert\nthe log due to the msg_id is not accessable for non-rabbit drivers.\nIt\u0027s not the right way to make things better.\n\nSo add the log back with the attribute msg_id for base IncomingMessage\nclass. msg_id is set only when the messaging backend assigns id.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a47155ea736ede206745a85fb4cd7df620fee056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a47155ea736ede206745a85fb4cd7df620fee056"}]},"branch":"refs/heads/master"},"817ded9cb9a3a9f57014061470d56d7fad41bcc4":{"kind":"REWORK","_number":8,"created":"2024-07-12 09:36:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/923337/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/8"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 09:35:53.000000000","tz":540},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the performance, but give\nus more information about the service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358 directly revert\nthe log due to the msg_id is not accessable for non-rabbit drivers.\nIt\u0027s not the right way to make things better.\n\nSo add the log back with the attribute msg_id for base IncomingMessage\nclass.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/817ded9cb9a3a9f57014061470d56d7fad41bcc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/817ded9cb9a3a9f57014061470d56d7fad41bcc4"}]},"branch":"refs/heads/master"},"104a63b9aa040dda567d356e598aad5f8734d3ec":{"kind":"REWORK","_number":9,"created":"2024-07-12 09:40:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/923337/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/37/923337/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/37/923337/9"}}},"commit":{"parents":[{"commit":"0d3338fd1d0094150491f8195533605e63642f9d","subject":"Handle NotFound exception when declaring a queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0d3338fd1d0094150491f8195533605e63642f9d"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2019-10-11 15:22:23.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-12 09:40:11.000000000","tz":540},"subject":"Add RPC incoming and reply log","message":"Add RPC incoming and reply log\n\nTypically a simple log will not narrow down the performance, but give\nus more information about the service status.\n\nThe commit 859e0d4eaa45d75eb822f4a6be46c547187c6358 directly revert\nthe log due to the msg_id is not accessable for non-rabbit drivers.\nIt\u0027s not the right way to make things better.\n\nSo add the log back with the attribute msg_id for base IncomingMessage\nclass.\n\nCloses-Bug: #2072156\n\nRelated-Bug: #1847747\nRelated-Bug: #1855775\n\nChange-Id: Ib35c8fbb24d5c51d3b54e8ca63e663428318eca5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
