)]}'
{"id":"openstack%2Foslo.messaging~925778","triplet_id":"openstack%2Foslo.messaging~master~Ide8b22e2c66eae6639266950a39c4042d9a656fd","project":"openstack/oslo.messaging","branch":"master","topic":"depracted-rabbitmq-option","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-08-19 17:35:30.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-08-19 17:35:30.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-19 17:35:30.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-08-08 14:42:35.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-08-19 17:35:30.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-19 15:48:11.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"Ide8b22e2c66eae6639266950a39c4042d9a656fd","subject":"Deprecate the option heartbeat_in_pthread.","status":"MERGED","created":"2024-08-06 14:56:14.000000000","updated":"2024-08-19 17:36:31.000000000","submitted":"2024-08-19 17:35:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":70,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925778-depracted-rabbitmq-option","meta_rev_id":"18b2acceca0759d22f0231411743013925057d58","_number":925778,"virtual_id_number":925778,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-19 17:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-08-15 12:15:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-08-15 12:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-19 15:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-19 15:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 15:01:16.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-08-06 15:01:16.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-08-06 15:01:16.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-08-06 16:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-08 12:31:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-08-08 14:37:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-08-12 16:17:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-08-15 12:15:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-15 12:23:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"22d859a552d0ece7969ae5297329c0640c547383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-06 14:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"104fe9b8061468d27c2339f9fc2e7f347a880799","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-06 15:01:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49b9bb5ac1792d022c07833c35eb1306b690f952","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-06 15:39:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70e3b4d27e44799e9867bfa3e28999aa59b148fc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-06 15:46:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"81287d44dc5b3cbc605ae14b37e682efe71f9e8c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-06 15:53:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"da958011b79dc9bef2e90d3ae5a6ca28f26993f2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-06 15:55:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8100c1913a1a3aeb6564218e2598301fc0090dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 16:48:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd2e351c69af43988d561b7f04e02520\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d9642901d345548d05dd592d5ba053 : SUCCESS in 1h 24m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842b1ba8c83f49e383e143ab0b08ec61 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2185d7ffd4ef4d2cb26f91d1de9b173f : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06c737422ad24d359d73519f7e6d96d7 : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f4601a72ebc4761807b08c8eae088f8 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e46f0bf1df9437e90278a716d4b7f68 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25950c03f1d344aaa533c1438cb4678a : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/019cfcbcde624c0b966a64ad97e48fae : SUCCESS in 3m 25s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/705b1999a6ac45df9d58a55aa590aa61 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/1c3261d54aa5434b99091dc158f0a2e0 : FAILURE in 5m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/069341a651744334a0b796c68ba403d9 : SUCCESS in 1h 35m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/639d2de313da427b9a74425efb7aa6cb : FAILURE in 15m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/347366c6ed2b4c5b94c737cf2945dc5b : SUCCESS in 1h 45m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d4de0d90994840cd81638cdb93ef5c4d : SUCCESS in 53m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6cce982b7fe4f71a0c0e82847b63c28 : SUCCESS in 1h 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8c0f1cfe4e27962f6dd72a0f8271f25b7bd9dc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-07 13:53:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"82b0c558aa3e4d0aa2b6a59d35241724b94f3656","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-07 14:07:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b02e0eb3e9749731b1ab58803766fd972b874ae4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-07 14:07:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5766a298e767597d535ebe8be7262f38e9e281b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 15:41:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0e4e508ded4e10989056ab4cb49638\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4390ce473777424aab27dfcdfdc31719 : SUCCESS in 1h 23m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4518a04186b546cb974af5e5ba855623 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fbf6d788394f0782b2a8fcd1ffbe1f : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dae9aad03cc45e8bd925ba950dcb5fa : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e38d89907844374905a0c4c1b9dfd51 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9101fa59ed9a4a889bc0621699cea9b2 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22c568b7ba141cf85c5409823a839c1 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/609048b1166743b298afc228497e7000 : SUCCESS in 4m 10s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/81e56a10dec6454aaf5b423394fe8e18 : SUCCESS in 6m 02s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3bca6f973c764324ae668fe8899dc418 : FAILURE in 7m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/836d68c8645b410dbfa8c079f4370ab3 : SUCCESS in 1h 33m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b8df13cd8dba4a2995cf8b519ea6da4b : FAILURE in 28m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c34f779ebc8d42b9a5fac543362b24b2 : SUCCESS in 1h 40m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/88b60961e2294dd9b55311a9af971f54 : SUCCESS in 1h 02m 00s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/72de08d50a6446e3a8e1c853c9767f5c : SUCCESS in 1h 19m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4637d1ba626e54a2973d5a67eada50fc80a64cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 10:17:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0cc7e9f00529450afd85cc0538a627e8068acdb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 10:19:42.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2c161ba34a17a743211c0d2daf17545a345f66e3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-08 10:27:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f58c2a0714f47547e313f44ad6766b982c02a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 11:47:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffcb3b90b3764f6297fb00406eaaface\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65900bf08e444641997f4182827357ea : SUCCESS in 1h 26m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeddb923f1ef48c3935545898c6e0104 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488d67613a874fa09b952136495ad02e : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98ddd230ade04dc7a28b0f66f77a1492 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a067e620d21846b784701ec1d7fcf892 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/872cb6c21145408c8fa364e30955cecc : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a3dc523e1f4e50b68653e08ea81abf : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d7ff9b466404ca58ddc875ed8e85473 : SUCCESS in 2m 57s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ba8d18969d5b4bf48fc851f0237999bf : SUCCESS in 5m 41s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5a4c141149bc4812b9868a0e64724436 : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9b513408f7cf4a82a6ed1498dea7b056 : FAILURE in 1h 16m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b61e6b433b334848a50550ef3d53bf27 : FAILURE in 25m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e87e1b7ec6214adc95bc21f2cab279e8 : SUCCESS in 1h 22m 59s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fcd8b157164a4b808e21cf216c01885c : SUCCESS in 1h 00m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc1836a9caba434d9b1ae53686ed561a : SUCCESS in 1h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5f0553df266b5b8ffd8f381dc5c7a29d867d10b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 12:31:18.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"49c8135c1024813a09c1efec3e93d7572f4d2a5c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 13:19:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"05a3229214bfeaec9a15fb7c7de6b1d94f21b41b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 13:36:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf492968c4208ad2fb3d23d31cc3d2421639a554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 13:41:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a35923d023590f4a3d13c4d81f97252ac268d0b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 13:41:51.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"c5a87a8ea1e2d4943a734e9f107c0800d300f8e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 14:37:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"255c56555193373a3d36215ec7d8f0ca1934dfe1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 14:39:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"565a8177f39481952e98186a20f262935030c6b4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 14:41:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c92d8fd7321ad3fd41fe1c6d8fa0917611517c2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-08 14:42:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f6f16c29a8fafecd1833b7b0a7f53d955d7fd9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 15:34:39.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7442271f71284db397208cc3ac0c0e99\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b324033be10442ab8bc32bf9c975654a : SUCCESS in 1h 25m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e2b86e36aaf49a5be1e0e57ae72d35d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663d86f4b48c4b7faa5753253e45a253 : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6c416a0a4414b86bf628a10388d5923 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c404464b5de4430c9f817f11d185eb2f : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3b2b88fbebf4a84ac662ba25f3e77cb : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56357d40534f469f9646f010b4229461 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41243fb563040c2889690b86aeff4bb : SUCCESS in 2m 56s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/104a4a2944e647de808cd73922287281 : SUCCESS in 4m 48s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6dc537ff51664372aea5423211c7f82f : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e96cf9541c2d4cb9ae01f1c70c86119f : SUCCESS in 1h 23m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/45bb6d48655c46efb62d939948d1b7f5 : FAILURE in 29m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/12a7ff8f4b76491ca13fba0cb86ef243 : SUCCESS in 1h 44m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/93dd22b3a9aa450a9d9f0dc0064e465b : SUCCESS in 1h 07m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/519d34563bb44927928cd0c1ecd59de4 : SUCCESS in 1h 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1685cb8f00ee45ddbbeaf4c1da960166ef750709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 15:57:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b421442a46abb240b4867d2cc8937ace52b2c836","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 16:03:25.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0b90eecefb5cad13499a4100f91c4023066ed386","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 16:04:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1c383ae984c9ede88acde59f89c55ababe43dba5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 16:08:43.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ea07e790776badd6f4430817c90fc6360088dcf0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 16:17:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"46cb989a03b3c487bb58a6f4bf6c34b318d9c989","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 16:18:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"adad2b309300853d16ba5f99cb5ecd5eaba26e0a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-08 16:28:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3ef455236ae48dcf8e64266e23e3b0c64a36ef49","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-08 16:29:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"be8a5ea88dc5425432564057e1951ef8426856e4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 16:38:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f19e992538aa8b0b9c33e31589b0425b1b636ac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-08 16:42:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fb31eb94c6204f03e7faf01391c9f997fbb87fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 18:21:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/500758b712444c689d5fb560a17625e4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c02e301f42441d1b6ff8d19c15f72ef : SUCCESS in 1h 37m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213ef7412110465d96b61e5b4d1aa92e : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/432a1fb932984b0b9285c6f6076893ee : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f7bb38c077b4c1089732b9fd13fd40a : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bee5c94bada40f78fd5a1546015aa39 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c36af9e8f4a46bdae0a9f2ad3fec2ea : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c65d598af088487cbaf7e258c5340c47 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de893c5874ff4eb4b7cfbd4c3a59199c : SUCCESS in 3m 39s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/74464dd1f0ca45158c36f5f935eb232a : SUCCESS in 4m 38s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/dedeed3ddd4f409fadb5afc091510db3 : FAILURE in 5m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a3c208563f904087af629d870f097677 : FAILURE in 1h 29m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/de95706ce6a6419c8982e6fcc0410341 : FAILURE in 27m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d896e8961fdb412d8c814c81a5c528e6 : SUCCESS in 1h 32m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6da9d94e29474b9199ce7fdc79facd3a : SUCCESS in 1h 04m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/232aaf0f6ca34ee7b3d130d254e03089 : SUCCESS in 1h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3c0d37f2a23cc82fc2314dfa7430da27abc9fcc6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-08 18:54:16.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3fd27463b0ff4c74cd11782e8b3f032afee0abd","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-09 13:23:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"45dc13334d4bce01fe2bbc3307d5fbff9cd1a89f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-09 13:23:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2cc41fa2636a41c9c03f0d780b3fe0114aac13de","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-09 13:25:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"568430bfae09aca01e0cd6832ad71b7b46b0c073","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-09 14:04:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2d58dc8ac1860ec4e280b1d72fd0489e5aa4073f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-09 14:04:36.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d341727b9ed9b01e71ee6775aec2c2fd252c0b4d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-09 14:12:44.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1e7887de29566331878b2501cae826f779a053f0","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-09 14:14:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"47cb7f7c6d44e82ef4e313806d2889202fc46641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 15:06:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cd82049ed54d34be23cc183ffb9d01\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/391031bd60a1487ab1da5ff5babfbb38 : SUCCESS in 1h 41m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0daa4b4fd814f2aa80e94c06d302b9e : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e968033c98be480ba401fe981323b58f : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06d1704534d749cf86a3f4ed6b4f2dc4 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aca81e5714b4838b8ec5021afa7f2a1 : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4757cfb1a1624c6fa08e053c2f070ebd : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8efb574f745401e8a381ff7851dabeb : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/071329deef0945a69408fa70fc22b9be : SUCCESS in 3m 13s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/35a82254b5874f81ae584b0b2f45a9d9 : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9a1b7f1fae084951b41527c83cdba04d : FAILURE in 6m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/89a6ed61474d469582a33458f9c4a0fc : SUCCESS in 1h 24m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a8b85113ade84da7bb38cf1d5f554c10 : FAILURE in 15m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/66bfd4e48f4940c195ad5e93e2de4628 : SUCCESS in 1h 27m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d5d8d86ea7eb4c778db98dab64f5a0fa : SUCCESS in 1h 00m 57s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0f9a2a7377a4d8981eca103e88fd122 : SUCCESS in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"057cb642f3dd9f119de48d261cbe3c4d0170331a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-09 15:08:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fad1f311954097174fcd874b3350ef79e40005a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-12 16:17:18.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1b8785deaf3ca04c25af89dc3b9185397dd6be0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-15 09:06:47.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"5ac63dbd5b2e305d70d03c39f9315f08a174e2ca","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-15 09:07:16.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d68d34c56ccf4da8d1242821ef7fe1133ffde5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 09:59:52.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"becbd90f55026cb2a3ed76c58386e0b0b962642e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 10:02:49.000000000","message":"Patch Set 10:\n\n(12 comments)","accounts_in_message":[],"_revision_number":10},{"id":"292b6418325915b7a29ca77e7ef4f00af5b1ab34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 10:03:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"814456df74daa158750fd5cf3ffb59a6a727f091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 11:40:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d30fed797a241cba0ac55360a4b8dc6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b7029e7065742e3be22dc645e9a6364 : SUCCESS in 1h 36m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a44fcf42004ee08ab57f702818d370 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed38e77084a5433fbfd6b314d46f8e3b : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e0e2f4b333345ed823d87668dd108e0 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c34bfdc9a62c4f679a4a20b4cce9300a : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/250353dd3be14377af139809a79c5a21 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9718caa4b63d46e2a1837c8ba214df55 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ab7e7eead7a41ea802ada850b2ba18b : SUCCESS in 3m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/881ce11456694d93a0b4ce6eb7b2f1b0 : SUCCESS in 5m 46s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/470170973a06488686a7c05808b37f3e : FAILURE in 6m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f24316ba00c54c34be961711d65d5414 : SUCCESS in 1h 19m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/376e12503883438ab451d42280da9622 : FAILURE in 26m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/25ab8dfb87f44cb783b8d10c513fc705 : SUCCESS in 1h 20m 23s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2c9d9ad239454427a6290885f073389e : SUCCESS in 1h 13m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/04f5b27694384d6b98cc8f2d10c230d1 : SUCCESS in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b3a575a36b52790b7fc081e74b44958440615220","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 12:15:10.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7f59d7c906a42ac0900c790e2434afa8047298ae","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-15 12:23:48.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"5af0583c62f6c8903a95976efbfacdd73fbe6080","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 15:48:11.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"64ef08a5ce77a1dacb16083eca267cfc8f80fc08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:48:40.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"c6d9cd9139a5e4618d90964782d2cb4acf7ef379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 17:35:29.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a6b41a2014149f390af776a24a621ea\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3294dd1553748568205acd0031c0888 : SUCCESS in 1h 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7913a7b61e477e844384f6671f1ac9 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defeb818e96f43de93f31c4be0a5715f : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/057166559ac649c496e637620fa5714e : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a9f44bd8ff4ff3b71a47c3ccd58a82 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0867454fc53145d8963a750c79cc765c : SUCCESS in 2m 59s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/d368c71e0c974c7587eb1cd47a9f9a27 : SUCCESS in 5m 01s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/144fe3e6323747af9f0e1538b125cad0 : SUCCESS in 1h 44m 10s","accounts_in_message":[],"_revision_number":11},{"id":"b08608278a39403b25b947d5356af904e1a4b63c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 17:35:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"18b2acceca0759d22f0231411743013925057d58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 17:36:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe6219660c34a1ab8300a6bfbed8d2d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b958045e064dc2b8cb972d574a0a22 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0092d7ed3fa4a739fa4557ddbfb33c9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"133ca65fa4dd394e1468085635311c80423eef27","revisions":{"c77a1ab95ff029b699716fdbeaacbf27325bbb5d":{"kind":"REWORK","_number":1,"created":"2024-08-06 14:56:14.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/1"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet and totally useless in term of efficiency.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c77a1ab95ff029b699716fdbeaacbf27325bbb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c77a1ab95ff029b699716fdbeaacbf27325bbb5d"}]},"branch":"refs/heads/master"},"991c01fd04c9c6c640bf7de8e94131f108bdd57c":{"kind":"REWORK","_number":2,"created":"2024-08-07 13:53:59.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/2"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-07 13:53:50.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet and totally useless in term of efficiency.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/991c01fd04c9c6c640bf7de8e94131f108bdd57c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/991c01fd04c9c6c640bf7de8e94131f108bdd57c"}]},"branch":"refs/heads/master"},"c35cc24c4b20ba66553d64ba8ba8523bddeee6ba":{"kind":"REWORK","_number":3,"created":"2024-08-08 10:17:45.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/3"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-07 14:06:26.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet and totally useless in term of efficiency.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c35cc24c4b20ba66553d64ba8ba8523bddeee6ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c35cc24c4b20ba66553d64ba8ba8523bddeee6ba"}]},"branch":"refs/heads/master"},"4cb9f9566f6f2e02331a5abec86d44de3c213917":{"kind":"NO_CHANGE","_number":4,"created":"2024-08-08 10:19:42.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/4"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-08 10:19:05.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet and totally useless in term of efficiency.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4cb9f9566f6f2e02331a5abec86d44de3c213917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4cb9f9566f6f2e02331a5abec86d44de3c213917"}]},"branch":"refs/heads/master"},"6e4928b864fc7702f38375e6afb0a6ea3e8ba643":{"kind":"REWORK","_number":5,"created":"2024-08-08 13:41:16.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/5"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-08 13:41:06.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed and the option have never worked as excepted\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e4928b864fc7702f38375e6afb0a6ea3e8ba643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e4928b864fc7702f38375e6afb0a6ea3e8ba643"}]},"branch":"refs/heads/master"},"4d93b2be8c48819ae1fc77919b8f5720fe78b607":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-08-08 13:41:51.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/6"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-08 13:41:44.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed and the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d93b2be8c48819ae1fc77919b8f5720fe78b607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d93b2be8c48819ae1fc77919b8f5720fe78b607"}]},"branch":"refs/heads/master"},"e15373b03fd7db69209087ded1c394abec63338a":{"kind":"REWORK","_number":7,"created":"2024-08-08 15:57:10.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/7"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-08 15:56:59.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed and the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e15373b03fd7db69209087ded1c394abec63338a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e15373b03fd7db69209087ded1c394abec63338a"}]},"branch":"refs/heads/master"},"22f5de69e4f07b16bf41badac710ab679f9962a0":{"kind":"REWORK","_number":8,"created":"2024-08-08 16:42:33.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/8"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-08 16:42:19.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed and the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/22f5de69e4f07b16bf41badac710ab679f9962a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/22f5de69e4f07b16bf41badac710ab679f9962a0"}]},"branch":"refs/heads/master"},"2206958e35f4e6c3f84d6922834853f1b62e1742":{"kind":"REWORK","_number":9,"created":"2024-08-15 09:06:47.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/78/925778/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/9"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-15 09:06:36.000000000","tz":120},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed and the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2206958e35f4e6c3f84d6922834853f1b62e1742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2206958e35f4e6c3f84d6922834853f1b62e1742"}]},"branch":"refs/heads/master"},"020e7a5f3f70b5421b5ac51f2ad275d01dd19fb2":{"kind":"REWORK","_number":10,"created":"2024-08-15 09:59:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/925778/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/10"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-15 09:59:09.000000000","tz":540},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed\nand the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/020e7a5f3f70b5421b5ac51f2ad275d01dd19fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/020e7a5f3f70b5421b5ac51f2ad275d01dd19fb2"}]},"branch":"refs/heads/master"},"133ca65fa4dd394e1468085635311c80423eef27":{"kind":"REWORK","_number":11,"created":"2024-08-15 10:03:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/925778/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/78/925778/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/78/925778/11"}}},"commit":{"parents":[{"commit":"104a63b9aa040dda567d356e598aad5f8734d3ec","subject":"Add RPC incoming and reply log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/104a63b9aa040dda567d356e598aad5f8734d3ec"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-06 14:56:04.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-15 10:03:16.000000000","tz":540},"subject":"Deprecate the option heartbeat_in_pthread.","message":"Deprecate the option heartbeat_in_pthread.\n\nThe option is strongly related to Eventlet. Eventlet will be removed\nand the option have never worked as excepted.\n\nChange-Id: Ide8b22e2c66eae6639266950a39c4042d9a656fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/133ca65fa4dd394e1468085635311c80423eef27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/133ca65fa4dd394e1468085635311c80423eef27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
