)]}'
{"id":"openstack%2Foslo.messaging~926704","triplet_id":"openstack%2Foslo.messaging~master~Ide5eb73164632c4e7be8582ba7bc2acde8287c78","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-08-23 05:23:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide5eb73164632c4e7be8582ba7bc2acde8287c78","subject":"Reduce log level to DEBUG for new messages","status":"MERGED","created":"2024-08-21 06:27:18.000000000","updated":"2024-09-27 05:08:40.000000000","submitted":"2024-08-23 05:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926704","meta_rev_id":"a713c9d4a2c9890862daf60e408e595306c7aacc","_number":926704,"virtual_id_number":926704,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-23 05:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-08-21 11:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-21 12:23:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-08-22 02:50:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-08-23 03:28:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-21 12:23:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-09-27 05:08:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 08:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-21 11:43:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-21 12:23:28.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-08-22 02:50:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"66c0749d1ce0a459f5a40cf769a6849a063bda78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-21 06:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1b6ecdf81e6675e7927f062eeaf1802fc9d24c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 08:16:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1551003ea1454b10b384631eee6f23c7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4302417339d42d89f777a268d798cab : SUCCESS in 1h 26m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/184160ec349441ca902611a30c0aee35 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cbeac160bf433ebf45076631c0d83e : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1b630d7d70841639d7d7c532e2d3461 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbb628bbbb084c0ab5d77436a1d4b101 : SUCCESS in 7m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/878466ce8a474664b6136cb931c38ff0 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14866816bcb4d848d2695d93f8f78d1 : SUCCESS in 4m 20s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c85fc53a50ed43569446e6880c1d68fb : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/463f1ce5c8834059bc06b1422798db9f : FAILURE in 5m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4f1e31c7cd9247ac8c0880dcd368064d : SUCCESS in 1h 39m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/207adb62e45843e88a9eeeafce541e9d : FAILURE in 27m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3e59491a9c194cfa9e737d60d44f0b2f : SUCCESS in 1h 46m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/68b19e20756b4f8fa7c545bda0650ff6 : SUCCESS in 1h 00m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/42089197819940418cecf665eb684be6 : SUCCESS in 1h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e9572fcc4c05f52b05b031e14ed2471b7372526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-21 10:21:06.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a43173964b8cf48e90d1c3968c06d06ca67bf2f3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 11:43:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2298b75d7c3fe9643880d61ce5dd35cbe686d167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 12:13:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d06219673b6e4fd1808b76f506111df8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef5a42de4df24aedb663ed41143b5430 : SUCCESS in 1h 41m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ec9b48a51f4e2ebd88d6f5754ddf13 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b6c4946e304a189f007d88c0b2b309 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6509f1b599d2453a9ef30d6d1cc8a4f8 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79f6744752d546f2bd6e6b6c76903ed6 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e45a6a208b64445ba6dfe01ba0295204 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d516f5aed74698927313081a441f49 : SUCCESS in 3m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f053e6d29cde4ba7803a63fe5fa8aa9e : SUCCESS in 4m 27s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7c85eb2efc53472eaeeb4874c45a7bbd : FAILURE in 5m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bc1702479b6b4e1db222c3556a94a55f : SUCCESS in 1h 51m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/92a3fd9fc4d54910a3300b2f9bb1e5f4 : FAILURE in 20m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d4fd2bb87daf445d9989369fcea9b390 : SUCCESS in 1h 47m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/93f5b71a5eb846a797b69565741af6d1 : SUCCESS in 59m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7a7bb0d42dd4f35a8dc63d005875fe4 : SUCCESS in 1h 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a61a2a89e0b49f57fe559d3afb75b45d59616ebc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-21 12:23:28.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fbdce983d0195d2de4f62c13f14dabfae54acae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-22 02:50:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ae6b92c52571df30a271551d77345893cdaa1fa3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-23 03:28:34.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"15d094d9beab7e5425be3a2736e16671d383438c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 03:29:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef2ed479b9b1f6dec26684f9400c445385559b64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 05:23:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b343a0466ce4407bb808b61f144c958\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6027100288ed48c6989dc04ae1992ef6 : SUCCESS in 1h 25m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165eb333004f41a0ac4e88120cfd663d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31d25258073149458867e15e40a63136 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/615ed258916943e69fc0d1424366fba2 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f08b08e3c64cea88774a997aaed16c : SUCCESS in 3m 32s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f771e3ddec72428abaacf98a5f92f810 : SUCCESS in 5m 04s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/32b31aaf031c4453a469a1d4a249e024 : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"56afb920959646bf70919e0965fd06d738654d54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 05:23:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1fa1536a477606dccb22ce0ccb28de08f9a9c0bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 05:24:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d498ac78b24f009bc2e23fad25b376\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333f38da63e542ccaf0eb36b3f1c02a3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2},{"id":"a713c9d4a2c9890862daf60e408e595306c7aacc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-27 05:08:40.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d601f7aae0b775a7203441c83f3cd1d53b68990c","revisions":{"bd89850074b8641b2dce44a7cd1cc77ce9d4ed41":{"kind":"REWORK","_number":1,"created":"2024-08-21 06:27:18.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/04/926704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/04/926704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/1"}}},"commit":{"parents":[{"commit":"133ca65fa4dd394e1468085635311c80423eef27","subject":"Deprecate the option heartbeat_in_pthread.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/133ca65fa4dd394e1468085635311c80423eef27"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 06:25:39.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 06:25:39.000000000","tz":120},"subject":"Reduce log level to DEBUG for new messages","message":"Reduce log level to DEBUG for new messages\n\nThe change [1] introduced new logging that\nuses INFO loglevel instead of DEBUG with\nmessages that doesn\u0027t help and instead\nwhen there is a lot of messages fills\nup the log file.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/889425\n\nChange-Id: Ide5eb73164632c4e7be8582ba7bc2acde8287c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd89850074b8641b2dce44a7cd1cc77ce9d4ed41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd89850074b8641b2dce44a7cd1cc77ce9d4ed41"}]},"branch":"refs/heads/master"},"d601f7aae0b775a7203441c83f3cd1d53b68990c":{"kind":"REWORK","_number":2,"created":"2024-08-21 10:21:06.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/04/926704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/04/926704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/04/926704/2"}}},"commit":{"parents":[{"commit":"133ca65fa4dd394e1468085635311c80423eef27","subject":"Deprecate the option heartbeat_in_pthread.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/133ca65fa4dd394e1468085635311c80423eef27"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 06:25:39.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-21 10:21:06.000000000","tz":0},"subject":"Reduce log level to DEBUG for new messages","message":"Reduce log level to DEBUG for new messages\n\nThe change [1] introduced new logging that\nuses INFO loglevel instead of DEBUG with\nmessages that doesn\u0027t help and instead\nwhen there is a lot of messages fills\nup the log file.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/889425\n\nChange-Id: Ide5eb73164632c4e7be8582ba7bc2acde8287c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
