)]}'
{"id":"openstack%2Foslo.messaging~927574","triplet_id":"openstack%2Foslo.messaging~master~I915d5396b0ac62b73246e2e5271109e979f0aab8","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-09-13 18:28:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I915d5396b0ac62b73246e2e5271109e979f0aab8","subject":"Remove comment reference to removed function","status":"MERGED","created":"2024-08-30 08:37:11.000000000","updated":"2024-09-13 18:29:15.000000000","submitted":"2024-09-13 18:28:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927574","meta_rev_id":"2a5484a8ab06e6d618c2f6ab1e38f994f350b7c5","_number":927574,"virtual_id_number":927574,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-13 18:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-09-13 12:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-09-13 12:57:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-30 10:36:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-13 12:55:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"439b1d280aa6a55aa4b3919fa36ec5e0853653be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-30 08:37:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"353c84d6575681dc738b2eed73f50aad65e5fb00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9c09182a23c473ca6ff982c07d61b71\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea956c7b9b884bd384ab913401a957ac : SUCCESS in 1h 46m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa50156037264ad99f89f9328212aa33 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e870024bb9ce4cf88d8e2eb8d7301ff5 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02a5a49e182948bb9dbb6a6ea7227dbd : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df518953b0324b22957cd4ac24bea92f : SUCCESS in 6m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45c86d87864f4590a1182e3fcc0310e4 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d48511d02104de9ac3fe762b266b048 : SUCCESS in 3m 13s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/39220d3d16354d3a8f332398b0dcc669 : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/364b45d8776e4277b0f5f8e7d290ff68 : FAILURE in 5m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5c71a6c169f2489a95f77ca048ac51ba : SUCCESS in 1h 18m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d5698bb6cd6e466bba7e4626f5c08178 : FAILURE in 27m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b175f8afedad49e2803669ff9e5a3ee4 : SUCCESS in 1h 53m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/713ae900c20345dfb0fe98f9fd6905e6 : SUCCESS in 51m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/35820ed7408840a6a75360c06482501b : SUCCESS in 1h 00m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b06438bc87af7192eb5c34bc37e552a13a86756d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-13 12:55:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6150db07c7b84ba094122689463b7d57c9f66607","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-13 12:57:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf6a1440a063d77541a449a0fcfc714092caa6c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 12:58:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"571d903705a09a0f1f143583f1e7e431ecf91a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 14:47:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e4b601b08564b1a8f7cbc15e10cf034\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/707f8f91800a4a5b828420d6d0465d2d : SUCCESS in 1h 43m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e18ed17c7045da943d21097f5a2b6d : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74d0eec2cd0a45d7b18bbcf5033822b3 : SUCCESS in 6m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/145db706cf854659a189153adde10933 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090b350e0a9e4e4fbd58f0f9453487f1 : SUCCESS in 3m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c7905ca5e8934ae2a9ae3d07be60075a : SUCCESS in 5m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d8c9887b92ab4947afeb9e9cbf641dca : FAILURE in 1h 08m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f56cf2a98b2cf954213e04d46046b92b55255f81","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-13 14:52:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dea29ac3cb8aa8a76cc1bffd5d5b4152e7cd46bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 17:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c7e09ea76f14a3b85b7e2fee7e1cc07\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f4f272fe4bf47668a5a3f97d81aad36 : SUCCESS in 1h 34m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a024108bbb04e338d5b92450be60734 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9f1782c34d4363b484d755b586be02 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/177a0b3f3eb745e1b571e18e530482ff : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf9b2a21b96477a84601ada9a740603 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2abe8c326bf147c8a0cc87ef1a0c7f95 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8762230573464a0b8a878821d580d61c : SUCCESS in 3m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/702de5c48ad143a3a9fcd3c4245c5a0b : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/04f0f47f239a488c9362873e6a067271 : FAILURE in 7m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5453b1f4da2649579a7d6188f448f311 : SUCCESS in 1h 37m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cc4d6ea43a1d46bd8718bbb71c4a6e7a : FAILURE in 21m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/faa3c2af3a7241469168574afbb0562d : SUCCESS in 2h 07m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/aae2aeccc2a742df8089518e3c9fc15b : SUCCESS in 39m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a8cdeb087af4448a1a4f3d2fc613009 : SUCCESS in 1h 21m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2776db75e00554f356796c3d1ab89ea82f17f83b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 17:01:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dfaeef2f42393b8887f842a4f81dd3fecf6dac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 18:28:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc45be5834e43d089386ba96f84bbbb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ea860ccd5574aec84777df4e689e5b0 : SUCCESS in 1h 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a431666049d54ee9b58b487b6f3fc8e3 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c2e5cc83f748c5a92fa45cd0537d09 : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/976ee9ae613643de83e3fd294aca42d4 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c89527ff29a45de948da00683364360 : SUCCESS in 6m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e13fa1cadc414496aa5e05391938f2a5 : SUCCESS in 7m 35s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d234624d95d14dfa8d3492f94198e4ec : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"597300d263c3b7f1b48c423d616b79c773edc65b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 18:28:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a5484a8ab06e6d618c2f6ab1e38f994f350b7c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 18:29:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d2522971a74c02a3033ef8d10f5068\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ddfdd87d63418fbe8957c51ee887f7 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1440d5606e88a2fcef9131a83cb9e8f39ffc39b","revisions":{"e1440d5606e88a2fcef9131a83cb9e8f39ffc39b":{"kind":"REWORK","_number":1,"created":"2024-08-30 08:37:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/74/927574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/927574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/927574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/927574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/927574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/927574/1"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-30 08:35:44.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-08-30 08:36:49.000000000","tz":120},"subject":"Remove comment reference to removed function","message":"Remove comment reference to removed function\n\nThis comment references the wrap_socket()\nfunction in the ssl module that is removed\nin Python 3.12\n\nChange-Id: I915d5396b0ac62b73246e2e5271109e979f0aab8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e1440d5606e88a2fcef9131a83cb9e8f39ffc39b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e1440d5606e88a2fcef9131a83cb9e8f39ffc39b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
