)]}'
{"id":"openstack%2Foslo.messaging~928034","triplet_id":"openstack%2Foslo.messaging~master~I9e22433ec039ad6783593d9cb7fbe22c9090534e","project":"openstack/oslo.messaging","branch":"master","topic":"fix-queue-manager","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-30 11:56:21.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-30 10:09:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-09-30 11:56:21.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-30 11:56:21.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-09-30 07:55:06.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-09-30 06:52:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I9e22433ec039ad6783593d9cb7fbe22c9090534e","subject":"Fix queue_manager in a containerized environment","status":"MERGED","created":"2024-09-04 17:06:21.000000000","updated":"2024-09-30 11:57:17.000000000","submitted":"2024-09-30 11:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928034-fix-queue-manager","meta_rev_id":"b817307f46858f538e933960bbcb4a7dcab20cf0","_number":928034,"virtual_id_number":928034,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-30 11:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-09-17 13:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-09-20 07:13:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-30 07:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2024-09-30 06:52:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-09-30 10:09:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 17:06:59.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-09-04 18:10:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-09-04 18:10:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-04 20:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-05 09:28:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-09-05 11:35:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-09-05 21:29:14.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2024-09-16 12:21:52.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-20 07:13:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0f256e690f9de9be28068ed3dec7b53cb551bd0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 17:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8eb8c0c1fd7d6063a854e1f9553c2536c300734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 17:06:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6161cd848ee03cf70c864d7aecf7b0feb1449984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 17:10:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"477e27d5249a6745b3f72ab56c9312c7e306d67c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 18:23:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"346925b0273c52f0ebc8493958eeccd325477d7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 18:27:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1e7bdbb0237c23db2aef5969083362a7ab8f3c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 20:17:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58841a5d71c24edc9ce4a8af945780a1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd63548747d34f608a7b13458ae3cfb2 : SUCCESS in 1h 44m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30a818c2ac624c24a09d61482894aa23 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b55ebf5589413aabb866a3f75a602a : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3768b0c3130046c8a4cf4e404215ca26 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe03372122e3497fb5ccea0b4dec0677 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/894d3d4440db4ac0a24d2febd1166dec : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1197918c11484201aa96843b1e0240f4 : SUCCESS in 3m 15s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ca15193ced4d4888b70a3e5c1fa0aaea : SUCCESS in 5m 27s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/1f8994029ec742b6871b4f8abcb4c5a6 : FAILURE in 5m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d502a1bfb8324658b2be01ecbde9835d : SUCCESS in 1h 20m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/65876463976c402b80304b9d3a961971 : FAILURE in 18m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dc279c58360f403dac40704768824605 : FAILURE in 1h 22m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/28de43785d4843ee82fde0fa7567f227 : SUCCESS in 57m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/eed90ba6fc744b78a01149917c102e3c : SUCCESS in 1h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fdd065ab35bafa8e5b25d21948b390961b6d1aef","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-04 21:04:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34c8e8ba926ffcc768b925609003a61c0604eb1d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-05 09:09:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d73d1abc3523adef5ea9c1c52a4bbbd0e77b777e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-05 21:29:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"770e42aa685c1f066d479c4ad22b475617058f33","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-09 15:30:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e851b44249e11dd7af5984111855b30caebc7b5b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-10 13:45:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d0faf537f3951e3086cdd36436c181db1363eac3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-11 02:49:48.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8823e9bf076dddfbf878b893015efe2b743df983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-11 18:57:03.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"501ee8a98bde5bc928dab3de0573e1d908967af9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-11 18:57:48.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"64ecbf6a336757bb517669084563df7c80978fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 20:35:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8e4c83927da449b8acc98bad2f4e7e8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e5dee3f55d74c2d8cff68263c07ab70 : SUCCESS in 1h 11m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d97b1b936bd14081a5709906f82176f2 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a469677f23494325ac27c789c1171ea5 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b44c92b85765434a96af310229cfe92d : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0953b0c2328142edb4a4f8d06535efae : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e960ded2fda64fa1bd9d5491b2933928 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39acba5e33d64796baf5d229de0f20ae : SUCCESS in 3m 32s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0c71d81d6b7e4997b98d3066e89c2ad7 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9dcb5ea3a21741cb94ac9e84b15ddbdf : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4adfa98b46a949649920475f65136a26 : FAILURE in 1h 20m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/414cb460caaf4cfbb16bd433c4349c61 : FAILURE in 28m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a690a096a26341c3915b9955bbe5379e : SUCCESS in 1h 37m 26s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ce30ed1c5d6f4749b3080c730d7ee44a : SUCCESS in 1h 06m 02s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/64da8a06cc444402bc0b28bd6e89a1c6 : SUCCESS in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3020ff4a6e024837bb2c0eda16ea09a4f115899c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-13 21:20:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a0ed085995ca62e00a67349c7664e04fddec5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 22:59:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5494f14ea36149dea90dc432e2272de3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7943253e13724891864f24988257c497 : SUCCESS in 59m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db1b8ec819b1435cb76c75ce4647ef84 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823eb9653f744ca093184799366957c5 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61e6a33317684277948f72376732b93c : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6d82dcd421749e19c6d969e71194a24 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/658f0e6537d540d0a2c4165762c18504 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35bb53352980443386f6cb740d7f6889 : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/101252892823403cbed7d667747e1171 : SUCCESS in 5m 49s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7c074d1a23c4488098447630250b2e80 : FAILURE in 7m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5524a1b1e18e49a4b6799eb50f3301bf : SUCCESS in 1h 27m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/04d6fa6cb51d441ea2d5e7eb9ba1423c : FAILURE in 19m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/df0d25b58fcd457c83e07940df8ef1a6 : SUCCESS in 1h 37m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ce577b10a6734f14bb9fb41af5c341b0 : SUCCESS in 1h 08m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb5e240e3497473fb7bd34d9e58e9e7b : FAILURE in 1h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"19454cec1e06a7870e09817a30484ef982729bee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-16 08:49:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7d245c38ded026ff9e3da3275cec16ab4046ab36","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-16 12:21:52.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1666edf657fd36c6d57cda068ea72258423089de","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-17 05:47:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"408e00e4d12e14f29f9fe16734b4d99a86c71f2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-17 13:30:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7bcfb1cf9a506283df2d6533587020fa6fdd1ebc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-17 13:31:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a59087673ac526cce19649984dae92fe7e8e2a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-17 13:31:39.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e8d6bac57178feed5801651b4149a13a9db0d3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:11:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e09e6b78e9041baac35f65ff58c77b0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97e4e3366fec40e2b23317308c9f21dc : SUCCESS in 1h 08m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a7691bf0da6411e85e2f9e316d92566 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f31d23327e44a9bccfb76fa4bed1bf : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00a4f42ba25d4e159c7631fe0c52ea9c : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68558408dd74457b96b85552f76c2078 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acb88bf5fab743aba759d3d79fe0e6f8 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17375e9b597b47f79cdeb6bfc15a6970 : SUCCESS in 3m 24s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b2c6e76eca124265a16a9f92d1c3596b : SUCCESS in 5m 00s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9d53dc8908a7456bbde1c48f5cd737e9 : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e34fb6d49d874ee9a671bc05a965e943 : SUCCESS in 1h 36m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a608de956dae46dfac21632dbb3ac59a : FAILURE in 17m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/350a1c97ca46410dba6fb237e543859c : FAILURE in 1h 29m 43s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/599dcb3e00b4422585502faf8588cae6 : SUCCESS in 50m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d97d83590109426ababafd27c93257c6 : SUCCESS in 1h 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f9ad3ea44bd7ac116e23ebd9f989630abb2f7de8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-20 07:13:34.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7dce3ffd359dcab4a5c3eb1156b6c3e26bf3bafb","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-09-30 06:52:08.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"dc82d5aac358904b8067f6cb2ed5eeec2549c3f6","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-09-30 07:55:06.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0249b1fffd622ca159b1d31936be8f480691465e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 10:09:20.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cd1c94f8df7a157ed363d81f76ee7c90b2487362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 10:09:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"bd8ebc35a57da4369a2305f8ce8d1ede41b0b278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:56:21.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b868ab68c0414b88a483aaa35715d915\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8892d39e5ad64a2cbf7780489843442d : SUCCESS in 1h 43m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a726699e4a41ddaba0a38ab79c4153 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f47c0b9493834c9a8b5ed3687608e779 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1613bd96a88546cc8ef25024e3fb559a : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b3930f0f794bb092a3b6167b7eafe4 : SUCCESS in 3m 22s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/15df41ef93584005a5e6361b5c657f0a : SUCCESS in 4m 26s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6c641e4022454d129092575036063072 : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":7},{"id":"9e53c1a68cc232b169f5f8d69ceb3c9938efc753","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:56:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b817307f46858f538e933960bbcb4a7dcab20cf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 11:57:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b0894ff9ed4fb6b8bdde4de9b97a5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbc0fa1277843cc9a7ea00a70ad6407 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6790f702fa5a56c9d25a4becba670bf1810a4162","revisions":{"2e01dddaf14cea004a6f3b04f1ccd7033988b162":{"kind":"REWORK","_number":1,"created":"2024-09-04 17:06:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/1"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:06:19.000000000","tz":120},"subject":"Fix for queue_manager in a containerized environment","message":"Fix for queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e01dddaf14cea004a6f3b04f1ccd7033988b162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e01dddaf14cea004a6f3b04f1ccd7033988b162"}]},"branch":"refs/heads/master"},"14edf01e69e75cfb3e4af7a88b0ef6eb8e164eef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-04 17:06:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/2"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:06:44.000000000","tz":120},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14edf01e69e75cfb3e4af7a88b0ef6eb8e164eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/14edf01e69e75cfb3e4af7a88b0ef6eb8e164eef"}]},"branch":"refs/heads/master"},"6dbd24a615522ff01a458400b33ecec94233a87d":{"kind":"REWORK","_number":3,"created":"2024-09-04 17:10:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/3"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:09:57.000000000","tz":120},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6dbd24a615522ff01a458400b33ecec94233a87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6dbd24a615522ff01a458400b33ecec94233a87d"}]},"branch":"refs/heads/master"},"fdb6f780cc6fbdc491984c5a14695a6434b3c0ca":{"kind":"REWORK","_number":4,"created":"2024-09-04 18:23:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/4"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 18:22:08.000000000","tz":120},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fdb6f780cc6fbdc491984c5a14695a6434b3c0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fdb6f780cc6fbdc491984c5a14695a6434b3c0ca"}]},"branch":"refs/heads/master"},"aab853a62cc185ccf82a50e92859b22a0728cf5f":{"kind":"REWORK","_number":5,"created":"2024-09-04 18:27:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/5"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 18:27:07.000000000","tz":120},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aab853a62cc185ccf82a50e92859b22a0728cf5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aab853a62cc185ccf82a50e92859b22a0728cf5f"}]},"branch":"refs/heads/master"},"0ad5c678f17ed9c688dc9d7485a5b81b5455096f":{"kind":"REWORK","_number":6,"created":"2024-09-11 18:57:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/34/928034/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/6"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-11 18:56:03.000000000","tz":120},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ad5c678f17ed9c688dc9d7485a5b81b5455096f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ad5c678f17ed9c688dc9d7485a5b81b5455096f"}]},"branch":"refs/heads/master"},"6790f702fa5a56c9d25a4becba670bf1810a4162":{"kind":"REWORK","_number":7,"created":"2024-09-17 13:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/928034/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/928034/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/928034/7"}}},"commit":{"parents":[{"commit":"d601f7aae0b775a7203441c83f3cd1d53b68990c","subject":"Reduce log level to DEBUG for new messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d601f7aae0b775a7203441c83f3cd1d53b68990c"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-04 17:01:56.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-17 13:30:21.000000000","tz":540},"subject":"Fix queue_manager in a containerized environment","message":"Fix queue_manager in a containerized environment\n\nThis patch fixes the operation of queue_manager in a\ncontainerized environment by adding an additional check\non the start_time in ticks since boot.\nThis way, we can detect a restart even when the PID remains\nunchanged as it is ussual in containers, but the start_time is\ndifferent.\n\n[1] https://www.man7.org/linux/man-pages//man5/proc_pid_stat.5.html\n\nFrom man page above:\n\n(22) starttime  %llu\n    The time the process started after system boot.\n    Before Linux 2.6, this value was expressed in\n    jiffies.  Since Linux 2.6, the value is expressed\n    in clock ticks (divide by sysconf(_SC_CLK_TCK)).\n\nCloses-Bug: #2078935\nChange-Id: I9e22433ec039ad6783593d9cb7fbe22c9090534e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6790f702fa5a56c9d25a4becba670bf1810a4162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6790f702fa5a56c9d25a4becba670bf1810a4162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
