)]}'
{"id":"openstack%2Foslo.messaging~933399","triplet_id":"openstack%2Foslo.messaging~master~I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7","project":"openstack/oslo.messaging","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-20 09:30:35.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-12-20 09:30:35.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-12-20 09:30:35.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-12-20 08:02:33.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-12-19 18:35:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7","subject":"Deprecate the Eventlet executor","status":"MERGED","created":"2024-10-25 09:00:15.000000000","updated":"2024-12-20 09:31:32.000000000","submitted":"2024-12-20 09:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"933399-eventlet-removal","meta_rev_id":"9c05ec37108d2936ec18747677f8ef0e06ff056e","_number":933399,"virtual_id_number":933399,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-20 09:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-12-19 16:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-20 08:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2024-12-19 18:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-20 08:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 09:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-26 16:25:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-10-29 13:57:30.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2024-10-31 06:22:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-12-15 09:45:43.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-12-17 19:06:52.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-12-19 14:06:21.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"e01accd4751964c5b93777ac2cf34a17c1918421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-25 09:00:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d47b90ebfa9ae253f06a1c9d2d7b7c8c9fe571f5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-25 09:37:26.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"68a97ac8893518c77af704bef735e635abe138ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:50:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6e216c38db14ed384d860906084fdae\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14824a1fc8c145af97d94fc9e08f0c6f : FAILURE in 15m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1114ecac77ad4e4b82b1ecf86e48c0ca : FAILURE in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a27e8775504d36bb03173f6569713c : FAILURE in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/642dc7bdc2f940c09f6ad75a3813ad66 : FAILURE in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e72e40a0841e4dc3bdf78830ec1c8a7b : FAILURE in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/546d96d5f65247de9166c91ecf447263 : FAILURE in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30c1cc70aaf845079b6f972a34a72378 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/25b1b9be6cff4aeaabaedef5808d2b9d : FAILURE in 3m 23s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/113de68b0a6141c6a26dc7da08ceceef : FAILURE in 3m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e9f76477e2884321a8fb1d0b75014f3e : FAILURE in 20m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/41368ef8449f4841a0a3f21c3644d8ae : FAILURE in 18m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a95710d445dd4bd0901794144aa63f56 : FAILURE in 12m 07s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d49a5e81a48d4c2b9f5da3478bc64fd3 : FAILURE in 45m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fe3e25934164e61813dc1eb93260982 : FAILURE in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f04744c9b5e85eee0abaac0837d52d5117592bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-25 14:01:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e72332b9e0a4f4c04f617a84b66a3dd7d682bba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 15:53:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dabce27c3bca4926b59e8443c636bc9d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbba3b89b75241b591027248423ab208 : SUCCESS in 1h 46m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e054232116543b1be44f8bcbfc9ebc1 : FAILURE in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a9996e0e1ac434d98b1d3851d1b226f : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/766fce123f1842cba00b17b99f386071 : FAILURE in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84861e1725564771a9b11041dc3d782f : FAILURE in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c19c6cfe5ee498ea084f6cd05507fe0 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7988312fe121495bbb53a6f8b4adf885 : SUCCESS in 3m 00s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e56c2efc6dbe432187e5ae638c22e57b : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/490efec6431c4a2c8678328c1848f1ed : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f9c538f55c7f4e65a6133fb12b9d21f8 : SUCCESS in 1h 42m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a8b27bf90f5a40ce8a53075875576665 : FAILURE in 30m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8a5333b161874ef286893a378c4c2f30 : SUCCESS in 1h 36m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/18324f4fd88344ae83fac54f6af12d1d : SUCCESS in 57m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/da9185dabb4e49b2bcf6907e4b8718e6 : SUCCESS in 1h 06m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6f2cdee63d054e0e6a56e579687ed227a744837","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 16:25:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"977529eb08e747924fe4f6e33bab4b9b0728113a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-28 12:56:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c1a099e3a8960d22096445d1f97cf330b173bbf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-28 13:08:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a3c987b0f46797fd2fa04e96457e886967dd969a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-28 13:09:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e619a49293d49f1f2977a900eb82e085e7b59d43","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-28 13:50:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"93e4dfcbd056f978e4f1294060b14bf052c3ab56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 15:02:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a336819274f94a30adeb7af227ec33e3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca143afb13f940638219af4ba179d04e : SUCCESS in 1h 36m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73f43035d11c4228a55152cdeb0016fe : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c9d00433b1d4659aebc9df622f9711d : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6051348be20e493ea2473ab0b9794e3c : FAILURE in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4288d4c0f2054c6da3e908dfcb17492a : FAILURE in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e8b3b34aba45c995e538a4d6aab3b2 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46feb4d006bd4f7e8645daa422f43fc2 : SUCCESS in 2m 30s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/dd9656675a734303968a9079705afee2 : SUCCESS in 4m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5b661280bdac4f46a107aeb2715c44f1 : FAILURE in 5m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9c7f219e195d4ff78731f557cefbaf54 : SUCCESS in 1h 48m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e8c5186db0ee4e5cade6cd880bb5443d : FAILURE in 31m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/399baf63b6424b7ea44c90d2694666f8 : SUCCESS in 1h 27m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/16bbd6b8fa9b443c8491e6c1c95cdc0c : SUCCESS in 40m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/936c2c8fc4434745b541384a68c6caa0 : SUCCESS in 53m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54b7e0e885ef011e60f4d1a09abb57e89ed74d72","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-28 15:16:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"299466b61aa278152bca4694d2bd9cd4ce923b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:48:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/feee743595c84a84af62f3af1f183ea9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e7093ff18e042b7a220d6fa6c56f87e : SUCCESS in 1h 04m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ce952c3b334b489b6353f402187d07 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad2b95e94fa45f7a5b105534aee9957 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdcd0448833647a9be04fb9367e54dd0 : FAILURE in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c086d8731269443cb7f6cbe24ba32389 : FAILURE in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2232ce6d044ff3ab5405b52dc96f67 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b51d3c4d9f54bf0ac9eafaa200f86fb : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/454272f744b8422b9dd99baab60fc3bd : SUCCESS in 4m 11s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/fb676ec8468e4adea8b4ebd12e875875 : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/28d9154a24334d8580d20bcded91196f : SUCCESS in 1h 21m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8871e07d14b64f78ad765aeaaf0eaa0f : FAILURE in 19m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/852aa797305946a1b383c62eb02d52b5 : SUCCESS in 1h 31m 29s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f548e76e8bee4fa48c8beff6dd90f20d : SUCCESS in 55m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a29fbb5ea9d4356ab857f4aa10c2265 : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9dc8eb45e37ed436e3b5f73832adb4513c94817","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 13:13:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f1fe0ed0f41cf02f72beb1ed2274c8f626c3e91c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-29 13:57:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0a917506c4c168679c4c29aef8953ed05655478","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-29 13:57:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39211ecbbd2c4818120523bfbec42d8b394b6059","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 09:28:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dfcd3d3825bad40c9e43df612df00caf25b4358e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 09:31:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"649418ccd3d71c06670360a134c5e870455c469a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 10:55:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe1e391e265cf1d0a2a1936949fa740ee7b9b2fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 11:01:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf5838238d2927b95872263b060e9a8ac203f8ce","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 11:18:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c2ad9a82b2b57b957811fb4c09f17f53e4bc90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 11:31:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48a09babd8ce46d2946cafd145090da8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa8cfebcfc7a49b3bcc376f4e60b6834 : SUCCESS in 1h 54m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3070bab5e0434714985a26ad6d8b3234 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00180162570b46cea10a98b8fa097601 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3126de32a8c4f2894d492a4478138a6 : FAILURE in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3038250284314646b9df39e8b9db5a61 : FAILURE in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd8977b2f8b4b95a3ccf8a6ca38a6ad : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c06b0be5ee443f90f00e6bc6e47e4f : SUCCESS in 2m 47s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b8b648037ced47a5aa2c6683830f405a : SUCCESS in 4m 50s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/b494371500b64c7796d0df1b383a766d : FAILURE in 5m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/32b51d1416cd4112bbf1c4aa4158d128 : SUCCESS in 1h 45m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6eda37bc927e4a1989363097172728d4 : FAILURE in 27m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2f36b7ad584447b6a68965baf7025c5b : SUCCESS in 1h 43m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/460f5e05732d474fbb76880bd5b0de4f : SUCCESS in 1h 03m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7d859a746b64e31bd0241df1883aa19 : SUCCESS in 1h 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5ad91949a307cd55e51921f1697c5630b83f92c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-30 13:04:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"76c0df743bba53e1ae70155269074c860ab04a1a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-09 13:18:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40e2efdab438f86ec1dd64cb4c0215593bc56b84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-09 14:13:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"eda3359eda2fc1308134b8943f66c7099e332c74","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-09 14:14:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6dd551f172dddd457db2ae1fd95c680baba054fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 16:50:12.000000000","message":"Patch Set 5: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/530e139bd69b46ea8df5cde96a656bb8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e996ef2349144288bf0b091c0eb674cb : SUCCESS in 1h 52m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50f695072446403e98506c2ba95e8780 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc24d9537ab4cf5b80d80b2200522f4 : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e3c402c9cfd4ae5a2366556a2de6da5 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2fdc840a69a94f33b1f5487822dd3b1e : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fe08d638414e76aa5cb2836e1c867e : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a5d143183bb4a4fac6bdd8322ec7a5a : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/990f5375a2154e7b9685a9b5c8fbc9e7 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/94152757d812496bb076c55fb4ed0c0c : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/370f10e57d1842e090f26440447ac035 : SUCCESS in 2h 00m 56s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e73b6e8c2c794d31b022cf24bd2e2dca : SUCCESS in 1h 42m 16s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d914522c706444dd8358880e5c3127c3 : SUCCESS in 54m 44s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4a37b574d764b5ca760264561d3b1b2 : SUCCESS in 1h 02m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee170c81d41316a9d95101501d2fffa5945fc154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-16 13:44:32.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"b931875afc4d598714706eb5c08dc75a6a64c32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 15:47:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a40da91c33e4900b02d8f4627bc6ebc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/834688b5f40245859302d2aa372beb95 : SUCCESS in 1h 38m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0266163649e447aaa7c741fcf7cf3d67 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60ce6b61676466e8e1b4151c904e21f : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbfe7208018941a4a77b59f8c5df5816 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63b7fb26c3094ab1a1d83ebc650485c7 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2154c6678c0424c80155906bd8e2729 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/747dfffc0e274f0ea5ea5d546fac7e0b : SUCCESS in 3m 02s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/04ddf7ecdeea4bbf83641466f5b6f31d : SUCCESS in 4m 28s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9a10a8f7c80f41e38d86fa920a810007 : FAILURE in 4m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4bba780173bd42ce9a9bcf0fcd9b1535 : SUCCESS in 1h 44m 50s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/52c58a2ef75349c6b3690efe1b3ad933 : SUCCESS in 1h 55m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/16c0361c05bb477a8e9e370728f7dd79 : FAILURE in 40m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c31928c9ea8048e58749de5bbd7d1c1b : SUCCESS in 1h 17m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a09604e3fc6e944ea4af795cde46f469773f4c94","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-17 19:06:52.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9048581ad16a6cb520b348d6aea9d66e738bfce4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-12-19 14:06:21.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f4a6aeff3c698f0a1c5f1abe1de85fed64b0d2e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-19 15:38:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b100a50f52212818dfdefd67a376cdc5d30fa3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-19 15:50:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9e37635c5960f44d62c3041a748b5e82c437cdb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-19 16:00:52.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2108ae843db85272cbb65110577b102de8feaa2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-19 16:01:28.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5e8a6361e15edc59552528bb416e18033dddaf54","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-19 16:01:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"2eeb5de04abc9ad6102baeda3bd6a7a24da11db3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-19 16:02:59.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ccc74e3c20c9208332c4311e757feb0da650b455","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-19 16:04:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"474b1ae478c49f21e8c8ee5584260fc32668ce1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 17:33:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b455e293700448193e3606f512dc3b0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be8cd5fd8de048b88153d377fb5c8b93 : SUCCESS in 1h 28m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30084421e2134921950494d7738cde55 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c0b6d2a13f497d9ca73208f085841f : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2780d4bf00634866b5e9d794485627be : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efda7807e618477ba6ad9461449067b8 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c4304d07f24554bdf03321412e2030 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef78ecc353724401906827d7a628ec2c : SUCCESS in 4m 03s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/36f795d1370741a7a0168f1c46f14994 : SUCCESS in 5m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d8f1fbf62cba49248f9b222a93a38fb8 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/457539a73ee645d2813fa89ec3c95682 : SUCCESS in 1h 28m 27s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/267cd34004cf4ed7bd87b76c922b817c : SUCCESS in 1h 30m 56s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7c82175d1a554427990db0206b212291 : SUCCESS in 1h 00m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa92ef7ac8f049efa0dde911c728d15d : SUCCESS in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"eefeaad203da0eec10eea1117ce4f31f11d6f72b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-19 18:35:00.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"81a3f1bde331dbabff4432025b82dfdc80c443d5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-12-20 08:02:33.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0692cac4735374064cc4f96135d3df5bb079da71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 08:03:05.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"2b1c5bc90a0eb831ed054bc776f8bf12f983b192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 09:30:34.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfa8d23b25a44ddac1809c2e5cde1a1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a83d438f3f9447dbb9411f2e95cf7926 : SUCCESS in 1h 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2064d6f3c6f743e9871bb32d75d845a4 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a0779cf345b42afbb5a2da38b2243f3 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c09029442a9548378d41a705eb8181c6 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5987b8db6243a18f13a7f8760e7eba : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92137da23325442fb8d814c0dc7f1d5a : SUCCESS in 2m 36s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bac7b2a54de34970a398c19b1e48b223 : SUCCESS in 4m 00s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ded8dd725515478f85118ee810164e4e : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":9},{"id":"0aed36bf201d008cdcf18ecdf3ffbe8504bd6a90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 09:30:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"9c05ec37108d2936ec18747677f8ef0e06ff056e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 09:31:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdadaf9c723940d2adf7f4377bbe3c92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc76194d82948fb84eca9e5504d85bc : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15f5187447614351a1b71b091687dfaa : SUCCESS in 39s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"27d833e374947cfb5489ad7e6b32fd0fbaa2f250","revisions":{"dfe996d1d20561565ddd3da0c7be06456cab3b0a":{"kind":"REWORK","_number":1,"created":"2024-10-25 09:00:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/1"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 09:00:13.000000000","tz":120},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfe996d1d20561565ddd3da0c7be06456cab3b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dfe996d1d20561565ddd3da0c7be06456cab3b0a"}]},"branch":"refs/heads/master"},"c5c3eba686f816b9a1e32a6c5f7201967ae95be0":{"kind":"REWORK","_number":2,"created":"2024-10-25 14:01:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/2"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 14:01:43.000000000","tz":120},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5c3eba686f816b9a1e32a6c5f7201967ae95be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c5c3eba686f816b9a1e32a6c5f7201967ae95be0"}]},"branch":"refs/heads/master"},"6bb3acd9cfdeae1b0ef9701868207a8c76dff41b":{"kind":"REWORK","_number":3,"created":"2024-10-28 13:08:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/3"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-28 13:08:47.000000000","tz":60},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bb3acd9cfdeae1b0ef9701868207a8c76dff41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bb3acd9cfdeae1b0ef9701868207a8c76dff41b"}]},"branch":"refs/heads/master"},"5cd72c206940e5f88f95212f62b588eecfc2990e":{"kind":"REWORK","_number":4,"created":"2024-10-30 09:28:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/4"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-30 09:28:15.000000000","tz":60},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cd72c206940e5f88f95212f62b588eecfc2990e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cd72c206940e5f88f95212f62b588eecfc2990e"}]},"branch":"refs/heads/master"},"4f4aa9d186a9ef74355a2ec9aa330e2924d41f2a":{"kind":"REWORK","_number":5,"created":"2024-12-09 14:13:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/5"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-12-09 14:11:54.000000000","tz":60},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\nThe executor parameter is now planned for removal.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f4aa9d186a9ef74355a2ec9aa330e2924d41f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f4aa9d186a9ef74355a2ec9aa330e2924d41f2a"}]},"branch":"refs/heads/master"},"8f043cd3bea082d131eaad0c25b4c6fa8ef70d6c":{"kind":"REWORK","_number":6,"created":"2024-12-16 13:44:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/6"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-12-16 13:44:22.000000000","tz":60},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\nThe executor parameter is now planned for removal.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f043cd3bea082d131eaad0c25b4c6fa8ef70d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f043cd3bea082d131eaad0c25b4c6fa8ef70d6c"}]},"branch":"refs/heads/master"},"2f9aa44eea7b270aa9d6c6fe4266798ccf007efa":{"kind":"REWORK","_number":7,"created":"2024-12-19 15:50:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/99/933399/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/7"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-12-19 15:50:11.000000000","tz":60},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\nThe executor parameter is now planned for removal.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f9aa44eea7b270aa9d6c6fe4266798ccf007efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f9aa44eea7b270aa9d6c6fe4266798ccf007efa"}]},"branch":"refs/heads/master"},"f05075c7e295ebbe5b159f963a7c8243b3dcdf96":{"kind":"REWORK","_number":8,"created":"2024-12-19 16:00:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/933399/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/8"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-19 16:00:51.000000000","tz":0},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\nThe executor parameter is now planned for removal.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f05075c7e295ebbe5b159f963a7c8243b3dcdf96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f05075c7e295ebbe5b159f963a7c8243b3dcdf96"}]},"branch":"refs/heads/master"},"27d833e374947cfb5489ad7e6b32fd0fbaa2f250":{"kind":"REWORK","_number":9,"created":"2024-12-19 16:01:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/933399/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/933399/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/933399/9"}}},"commit":{"parents":[{"commit":"5994f3ca95c2dae3704614594ec799333741e20b","subject":"Merge \"Skip grenade job on doc updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5994f3ca95c2dae3704614594ec799333741e20b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-10-25 08:57:40.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-19 16:01:28.000000000","tz":0},"subject":"Deprecate the Eventlet executor","message":"Deprecate the Eventlet executor\n\nThe removal of Eventlet is planned[1].\nServices should start migrating there RPC server\nto the threading executor.\nThe Eventlet executor will be removed in G.\n\nThe executor parameter is now planned for removal.\n\n[1] https://governance.openstack.org/tc//goals/proposed/remove-eventlet.html\n\nChange-Id: I9dc3fe42931c84ad2aef35cb663ba019e3bf26b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/27d833e374947cfb5489ad7e6b32fd0fbaa2f250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/27d833e374947cfb5489ad7e6b32fd0fbaa2f250"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
