)]}'
{"id":"openstack%2Foslo.messaging~934116","triplet_id":"openstack%2Foslo.messaging~master~Iaeb52791008a7a6736c99459f66d2bdbb2dea17b","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-16 17:35:53.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-11-16 17:35:53.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-11-16 17:35:53.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-11-16 15:51:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Iaeb52791008a7a6736c99459f66d2bdbb2dea17b","subject":"Remove deprecated AMQP1 driver","status":"MERGED","created":"2024-11-05 11:20:09.000000000","updated":"2024-11-27 15:52:00.000000000","submitted":"2024-11-16 17:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934116","meta_rev_id":"b5e5a8c76980e30903eb3a8fea68a90e11c047ae","_number":934116,"virtual_id_number":934116,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-16 17:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-16 14:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-16 15:51:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-16 14:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2024-11-16 14:00:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 13:07:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 15:24:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-11-05 16:55:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-07 13:25:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-11-08 07:19:03.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-11-27 15:41:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"ee9f6248d0d5bb03906bc9c76e1583eb6cec9ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-05 11:20:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e15b0da8985c66dee320f0e9850c60c99acdfa9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 13:07:03.000000000","message":"Patch Set 1: Verified-1\n\n(45 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81ef2beb15f9425a882fe3a25d4d5df1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba7fb7d08db44471b4ebe2a2416b6a9f : SUCCESS in 1h 28m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36685c31b35545209ed5d7af29654dd5 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0996495c573b473cb29dabcfccc7ade6 : FAILURE in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a35a43e740544f59c397c13fc2b9fe1 : FAILURE in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d35f02116b814588ba88175191a08908 : FAILURE in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8295c725e95043a1ba94e00c2807eabf : FAILURE in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/225a858c834e4ac0824b33ab11b98308 : SUCCESS in 3m 13s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/edf6033c5344424781eba8b4f7cfcd72 : FAILURE in 3m 28s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6d0b9d4a275140489dc2b4ae6fd16226 : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b9bb9853d2e049118c0e5c1af2421ac6 : FAILURE in 1h 29m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/370899fec43f4fd183da36631d18a66c : FAILURE in 33m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5038ebfb4c0a4d68ab5f54ff95fc6caa : SUCCESS in 1h 41m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c59938336d0746708f1e97703b1d838a : SUCCESS in 37m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/de6f90f954d7487f89f09a6b0a9c940f : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c81ad0e62e795634a0a767d0e90bad0da568f39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-05 14:02:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"088bf512142c40d571d89785548740036e7d59b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-05 15:24:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"139a07000bf5288bea4d222b9cfd1910fcfefab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 15:58:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/737db05f902e4caf98916b542579333b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/952ecae8ee1c4a01b80945308ed6e703 : SUCCESS in 1h 05m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d8248bdab1545e5a9ab5f5c13b1445f : FAILURE in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/272d5f2aec264be0b827bb4b747e95cd : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1340539add874329b12f1d4c4792db6e : FAILURE in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a3768d53d054c2aaf07a526ca2c72c0 : FAILURE in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8962061e752c42f4967a385c3d76fb9f : FAILURE in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a70074852874ff082111408c9784d91 : SUCCESS in 3m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1c76ff7f3222432884278eed96ca2487 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/bc23bd36c00541ee963cbba9e900dfb3 : FAILURE in 7m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c51eb3fe64044446944f26b19002ffcd : SUCCESS in 1h 26m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f2342968914a4b2894027f9faed85bf4 : FAILURE in 16m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c775b212d5f84eb3bd78efaf0734d784 : SUCCESS in 1h 49m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0b43e9ef2d994596a3a566d9b1b7618c : SUCCESS in 58m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/11e515193e794e468612831edb90f8f8 : SUCCESS in 1h 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e613c433b2de794863737ab71288b02cfca0f1e9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-05 16:55:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"331fe666fb426c5010ccf44667142552f3f832e8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-06 09:05:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1dcd79d82c031bb43f13dd42422a2954c9b03426","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-06 09:12:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f613e6a129c87a6b8a4c3a6f26693558c219c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-06 09:13:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"50f5c6e5d51fb7faaaf7ac022f89b0ed26b1f6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-06 09:37:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e7298fd58a6a0306f5c24be8951f25eeb7827afb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-06 09:38:35.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a438e0364100aeb4882018852c09875695fc55d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-06 10:47:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e956453aeb4be87b9f6cab94e1a3b2bb9846fb77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 11:25:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/702c3f6270de42ee913bbebfe5e35d9a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e5644c53ad8470a98bcc173e28ff814 : SUCCESS in 1m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b35efa857b44d1bbf35161dffe0f3bb : SUCCESS in 1h 46m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d6e02a22d340acbb79121c47224ed1 : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50bc846854a414a9d300a6fbc3820b7 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9620141bcce94b4e80ab53687c536b74 : FAILURE in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb206a3ee4c142b0b9f1f03f74130e44 : FAILURE in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76443f7b6f64861adbfec75fbddecfe : FAILURE in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac84bf68584d451682287b65d4b964f7 : FAILURE in 3m 48s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/fd1dca34c1b8495b8222081fcb6c127b : SUCCESS in 5m 06s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/354e86d6f7494e7f9867f2a293aad0f0 : FAILURE in 7m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2babee1cb2c041ed88b95751927c93e4 : SUCCESS in 1h 46m 17s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/aabdbee02513405f876080e886ce0a03 : FAILURE in 26m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/273c72d0d4554043bd7ed8015e3901eb : SUCCESS in 1h 35m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/020520897422482681ac8b007cf48d1b : SUCCESS in 55m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/daf3a080de5040f3ab8bbc1d25ecb3d4 : SUCCESS in 1h 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd7b8e8068517f02b93a6ae48474c9e73b567713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 09:40:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b019081b2f22f2166b1fd26e593a8f1b6a588f3e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 09:43:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"36c29ad0cf46013268d6505003aeb12400186afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 09:50:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"054d563634d3a2b5491d7705b559cd1c0b9a0b80","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 09:51:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2811a7a50806eb89ed837a11dcfcabc1f1708f3e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 09:51:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ebc88d95d32b93a1cd869ab7823d65dee402e7c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 10:02:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"916910f11a6d5bf89724c6aeb303d017c1af91d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-07 10:06:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7b1983473eebec01f5eeed540c244299493c508e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 10:52:28.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"945a8c98904bd40b616dca3eb99c0b94642398b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 10:56:48.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"cf1b05f5b182f130689f868e3db3a0225b611b94","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-07 10:58:58.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8f37105c656acc7131dbed0d0223c6162a218e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 12:45:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86c4ad86908346e59e812e190fb388fa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a0df80130aed4c5fb2602fb145863ef8 : SUCCESS in 1m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13a7e91c5ef746b6b8e977c1aaf4f82f : SUCCESS in 1h 23m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c9fe977e7fe411f83d6847b2f83e47d : FAILURE in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e21b423ad33424d86d108490047423e : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b47f34926a04633a6b93965d65f6c37 : FAILURE in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/310f91b7e8524dac913489ea2374fe08 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a257a7047fe46fca54842d7e38d8393 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0943ea6fa6394b8eb53992de74026148 : FAILURE in 3m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/637c2436fd374e4c8cc2e19f430febac : SUCCESS in 5m 51s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/b70d3526987048418fbb333b924f9293 : FAILURE in 5m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e3289a34b4cf459ca32d821036ec7f57 : SUCCESS in 1h 42m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0ecc64ca22434ab389afdc3cf905cac7 : FAILURE in 27m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dc567c19d527482abc7180159eb0290f : SUCCESS in 1h 14m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/aa054501c91f4804be2d0ab0641dcada : SUCCESS in 1h 05m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/949c751af5e04c9f819a988d05ecd612 : SUCCESS in 1h 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cb261a3fd1579632e70aa2a5589a30c0d220dc4b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-07 13:11:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fee37b8e48842346904a54ab31adc033072b9e52","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-07 13:25:42.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"adefea2d4c2f4c7bf52931b8623b97b493bd8a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-11 09:31:36.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"757d9afb65068562d2bab2ab6f84b81c24bcb5d1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-11 09:32:39.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"743f643b92425dfb1f3b7085ce3718246ac9425f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-11 09:39:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"32f30b071ea2913381772750e38054241f2e1584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 11:21:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5296209ad404b26be758cb95b231aae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d0fa03c0c6340bca4e7169fc6230a6f : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877a10e174e34ff185b444f7872cd9e5 : SUCCESS in 1h 18m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee2a4ff26bae492480086148644941fd : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ca728f55ce42e48020d5eb6d40b56d : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6860a2a978e5411baa2a3518942df7ba : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2dc275ae41a43ec8c76294d241b34d0 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1e465410a5429d8fe2e9d1bf3dba7f : FAILURE in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb9100d9c56e491a8ec72c945599c6cd : FAILURE in 3m 03s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/741ff210bc4f487cb16076d0aed67dff : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/dba0494504d14a1883bfd8b5a585c639 : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c6507878ed604917b2eec81cd1cfb246 : SUCCESS in 1h 17m 45s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9b136c58eb0344b785ab372fa093a6ed : FAILURE in 26m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6f6877fd949a474d97c947ed1d633c49 : SUCCESS in 1h 16m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3bcf6a8d90194146a8050fa79ccb3a87 : SUCCESS in 1h 08m 34s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/299299823273473f862de5a3ef01f013 : SUCCESS in 1h 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6695fb94be784bef2296600e344dd4f1c3af9ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-12 07:58:48.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"fb2bdfde46bab80f53bebfdfd100e841d62e05ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 09:37:38.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7376833832bb431bacf5c195f74eb1e8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/773de736c94747638c36450a66f0bdc0 : POST_FAILURE in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b83d6692813b4f96a6c7d3a284270f5c : SUCCESS in 1h 14m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/602b53bd693f4644b1be2e19329d6935 : POST_FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de73ffe5bab241049c621a705cd9abf7 : POST_FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e3efebe1bb6474b9ad6a583884ba4b9 : POST_FAILURE in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e83d6ef39d394128af588b0fd7673b22 : POST_FAILURE in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10fd92abcc594dd19c1e895d2dedd2c0 : POST_FAILURE in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91498d46771641ff82fb6eb6c8908fdb : POST_FAILURE in 3m 02s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/05cb898a60d44423aba8e48f16563637 : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e233f443c36e49438ec068d07939f8ae : FAILURE in 6m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cee066105d5a4a91abfb02169c15b5b7 : SUCCESS in 1h 37m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/87929d978cab4eb5af60fbbbdef810e6 : FAILURE in 17m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6b19ea3277154b818f490c58a756a45e : SUCCESS in 1h 17m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/08bdc28c996c4f06be0e5bee3c862413 : SUCCESS in 55m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b56b60db47b45568c574ea9f647940e : SUCCESS in 1h 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4474d367c4479f3f4a12c299d3d61d6f38243465","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-12 10:31:47.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2ac270da12aaf2e9a03cd362dc272730de1561a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:26:04.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1780517345e4017b48673cb85de711b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a3de234143264b66a4df7a4c8fc33468 : SUCCESS in 2m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21605bd4bc3a477aa9e7828bec4efb47 : SUCCESS in 1h 36m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4deb3ec32fe04666b2d08b96aae7eb15 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5711c18b418422a9d3511d062ee7c2c : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8cc63c4cde54a779d3afc80a36c77c1 : SUCCESS in 6m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aea49a1f515745d79b59cb9733050a7e : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4330acc328b43a0a4656ec2b49306a7 : FAILURE in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/086dbb9948b846f3b579e770f7006f03 : SUCCESS in 3m 44s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b501c3b035cd4e87a574b68d9920e821 : SUCCESS in 5m 59s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/96fb041377ec46e997d89e2183c13bf4 : FAILURE in 6m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/69e24695bba1495082f8d21aca335aa5 : FAILURE in 1h 44m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/49eaa66943c742f78e737951999a80e2 : FAILURE in 16m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/590eb4c4f0ee4b9fbfbcc4593ef85c8a : SUCCESS in 1h 26m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0fb9a4c44eca422d8f0574f787cce808 : SUCCESS in 52m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/68d57ef7f839476f8ab41397adccb7e1 : SUCCESS in 56m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"80381744d86d67a6f3ce01a0c1d0f49d14498f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-12 13:36:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"63f88143b969623abd77b5f1fdcf1c212928b406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 15:20:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9017929fae8b49e781a1a83a660bc0ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7b6a52c7a98427680b79be765e2b497 : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1211610c26f2408b86a19c25430255ae : SUCCESS in 1h 41m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ac3c86fb38847f681f0f029afce5ad0 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79382831780a47d4b1a030ab559f1b97 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f9b363823144cf4ba9b5cba3c4187ae : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/371839e92fb94219974a2520f22f0369 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1995b2b871aa42c38f8069e1f89a41a1 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ce004f60fef4bf6bb9d7bdc735232bd : SUCCESS in 3m 41s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/cedac430430a418189aba40d1050705e : SUCCESS in 4m 28s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c1b12bfdeb534a618fd7f5d0ca7cd940 : FAILURE in 6m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/85084ef0f39e48b79ed74772842bd342 : FAILURE in 1h 20m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c6357c0aafc948c7b5200660e91340bd : FAILURE in 18m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4cba03dd0ba045fe9e64b8efd0f78247 : SUCCESS in 1h 41m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2a79541f89084959a14ec1d92063011e : SUCCESS in 57m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f68483f6e584f6db18e047321eab1fc : SUCCESS in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ad5f978e13c1327e9dcaad88628b85264f555ae6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-13 09:13:30.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"e5256ab7a39d3cc3171cf30d2d921ec14d931017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-13 13:49:48.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"b4b80e85e67fd5bed73ddfcf83e4c738b1328b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-13 13:53:06.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"0ebf68b90ba0f5461dc597119d60973449cfb748","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-13 14:06:23.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"082b6dbba1347cac5852fb3f36fe7f84eb186b64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-13 14:13:23.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e189bd67e6fba4feb0106a8b3b6f91403ca31521","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-13 14:14:35.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"21b73a4ae4a73481cc1e45da087697752acad7e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 16:06:20.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f674cd2967c4472eac7bcb5bf3360eff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/762d796d2e4a44d8b76e7453f0a643f9 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e114f57e937d493fbb8d0335508c4abd : SUCCESS in 1h 46m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9f1a047504c44f0beb47f43d8656064 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af27f82cd974e9caeceb92547e12f9c : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcb89143269d419d9e26592f8020aa17 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dddcbc24ce3f40158986a2dcf729c46f : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463f5c908208401790bd674782ee6542 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f2a637f19904f81a47d18bc7be1204c : SUCCESS in 2m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3020ea714ec842d58c4ecb7962889887 : SUCCESS in 5m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f23c9a7619984677a40bfa533204b250 : FAILURE in 7m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2117035fbafc40779b016d52efb43dd6 : SUCCESS in 1h 43m 58s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c6f35c7bec1746229056a1b170e3e73b : SUCCESS in 1h 40m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b2321eb2b6d649f889beddfb2ee8bae2 : SUCCESS in 1h 09m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/064842fd38734a4cadcd431529d43692 : SUCCESS in 1h 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"0a8a36fe71e14a808f806dcf5b4bfbe051d12b82","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-14 09:51:25.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"bf5140772d28abd49d73c776f4024be420fd81b4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-11-14 18:53:13.000000000","message":"Patch Set 16: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"bd708e8980771cdb8ad1463f5f3495b3d5096202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 14:00:02.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"cabab3ea8f777e7287df3929768eff434b91c2a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 14:00:05.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"60a04555add9dba8b2df35ab3b700f310e15366f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 14:00:32.000000000","message":"Patch Set 17: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"1fe555e5510c48d2737ae84a82a9662d3dfdac5f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-11-16 15:51:39.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"5502d12a6473ab1715afa2b484c22f9e2c8607af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 15:53:20.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23a5eef849334105a599249fdc278c12\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97adf14896634b2f9987074284ec1f5d : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3217e1fb9e1943a6a06b435229110403 : SUCCESS in 1h 04m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a04f694c71e7403980be960c9b197583 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f4c24094974be88697fd0958dfd501 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f377d3b14f7439a893ea8af6bce943f : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cfc74a591594f268ba13474307309e0 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d635e13e380545859157ffba6cc66c29 : SUCCESS in 2m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dbfe6c756234513b2bb11fa2a21cbc4 : SUCCESS in 3m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/623d5605dff5441f9cd2aa1fb072445a : SUCCESS in 4m 36s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/34f2d225ffe44cd99d0c7fdd494afa8c : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c3381f8f5d7f47b1aeaca7b114960781 : SUCCESS in 1h 45m 12s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/af1128120268436baae432363ac09239 : SUCCESS in 1h 42m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fd46d27807bf4713827822295b9f2d09 : SUCCESS in 1h 00m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1443bfa643143f98a7f836c60364840 : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9412b84d73cc7fc5c121274b878b06d0b1cbf90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 15:53:56.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"25727889d0202a20b712ab4e34a47b8712dfc2f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:35:53.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fa6f51703f54219b84534eebea11f4b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af544ba7bca24c21958bf8d96a92dd67 : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24ad5058c0084d88bf3687ae4edacfb8 : SUCCESS in 1h 35m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6672d43d5654daa8c61feed881fab6a : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56c62084b504d0c8cad576756aebff6 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4112214a4e34344b55bba4e44fe28ca : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e02a8b21e64beabe033fff06a78019 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d450986c653543a8886bf88700cd3d2b : SUCCESS in 2m 33s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/20d1d95e87fe459389991d6c7bedd850 : SUCCESS in 4m 11s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a59e47e083884a5d97ea14a8051c3522 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":17},{"id":"eee462966e19682e4aa041510b873e543a4dbc5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:35:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"25a657bfa00dffb736ab9a65dbf08143c94cf2b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 17:38:16.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3be1aece87441beb0755d68cab6cfb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05bb1976edc4f418d5369d9e256e4c2 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2ae57bc96894c5095c9566d090434e6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":17},{"id":"41631f6df9234539c14c81b3cefeab8f653af502","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-27 15:41:49.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b5e5a8c76980e30903eb3a8fea68a90e11c047ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 15:52:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"7997a199acada382b0c61dc84aa956e4c65de986","revisions":{"12d8833876c4c9cb8df4bba014a7a87b226a4d0a":{"kind":"REWORK","_number":1,"created":"2024-11-05 11:20:09.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/1"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:19:40.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/12d8833876c4c9cb8df4bba014a7a87b226a4d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/12d8833876c4c9cb8df4bba014a7a87b226a4d0a"}]},"branch":"refs/heads/master"},"6c95282c1ee367aaf306fff09c40535271452ae8":{"kind":"REWORK","_number":2,"created":"2024-11-05 14:02:04.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/2"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 14:01:55.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c95282c1ee367aaf306fff09c40535271452ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c95282c1ee367aaf306fff09c40535271452ae8"}]},"branch":"refs/heads/master"},"1acbad2c1bb07550b73e66356619569c01328197":{"kind":"REWORK","_number":3,"created":"2024-11-06 09:13:18.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/3"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-06 09:13:03.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1acbad2c1bb07550b73e66356619569c01328197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1acbad2c1bb07550b73e66356619569c01328197"}]},"branch":"refs/heads/master"},"60ffc9b1c4ac2bc89a230bb75ef07430e0d1e7c5":{"kind":"REWORK","_number":4,"created":"2024-11-06 09:37:50.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/4"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-06 09:37:42.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60ffc9b1c4ac2bc89a230bb75ef07430e0d1e7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/60ffc9b1c4ac2bc89a230bb75ef07430e0d1e7c5"}]},"branch":"refs/heads/master"},"845e051d050f2ea7e17c9cb719afe5d64dd112c0":{"kind":"REWORK","_number":5,"created":"2024-11-07 09:40:49.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/5"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-07 09:40:39.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/845e051d050f2ea7e17c9cb719afe5d64dd112c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/845e051d050f2ea7e17c9cb719afe5d64dd112c0"}]},"branch":"refs/heads/master"},"6e52106b1f768229354311472fd6d93eb33e057a":{"kind":"REWORK","_number":6,"created":"2024-11-07 09:50:55.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/6"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-07 09:50:43.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e52106b1f768229354311472fd6d93eb33e057a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e52106b1f768229354311472fd6d93eb33e057a"}]},"branch":"refs/heads/master"},"99e07bec1618bf2abdd3d28aff3e8c9ee04adcf3":{"kind":"REWORK","_number":7,"created":"2024-11-07 10:02:37.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/7"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-07 10:02:30.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99e07bec1618bf2abdd3d28aff3e8c9ee04adcf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99e07bec1618bf2abdd3d28aff3e8c9ee04adcf3"}]},"branch":"refs/heads/master"},"4a917103b4539b45da0fe14a7392a01fd58dae9b":{"kind":"NO_CHANGE","_number":8,"created":"2024-11-07 10:52:28.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/8"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-07 10:52:20.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years and is no longer maintained or widely used. This commit removes the AMQP1 driver from the code base.\n\nReferences:\n- Original deprecation commit: https://github.com/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4a917103b4539b45da0fe14a7392a01fd58dae9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4a917103b4539b45da0fe14a7392a01fd58dae9b"}]},"branch":"refs/heads/master"},"c3938a52189d48b6cdeffeb3038e82b4f694bff3":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-11-07 10:56:48.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/9"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-07 10:56:30.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c3938a52189d48b6cdeffeb3038e82b4f694bff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c3938a52189d48b6cdeffeb3038e82b4f694bff3"}]},"branch":"refs/heads/master"},"6456a35327e8666c7a311d9a51e462b301ac2b1a":{"kind":"REWORK","_number":10,"created":"2024-11-11 09:31:36.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/10"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-11 09:31:28.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6456a35327e8666c7a311d9a51e462b301ac2b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6456a35327e8666c7a311d9a51e462b301ac2b1a"}]},"branch":"refs/heads/master"},"f051e8adc23d86a3477fc1c253efc41cae2c7bcb":{"kind":"REWORK","_number":11,"created":"2024-11-11 09:39:23.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/11"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-11 09:39:13.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f051e8adc23d86a3477fc1c253efc41cae2c7bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f051e8adc23d86a3477fc1c253efc41cae2c7bcb"}]},"branch":"refs/heads/master"},"086df8f00304adfc52d9cf14780ef18b6b3a9083":{"kind":"REWORK","_number":12,"created":"2024-11-12 07:58:48.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/12"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-12 07:58:35.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/086df8f00304adfc52d9cf14780ef18b6b3a9083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/086df8f00304adfc52d9cf14780ef18b6b3a9083"}]},"branch":"refs/heads/master"},"4e28b23f69ebe4a951f345a05192816fa1288050":{"kind":"REWORK","_number":13,"created":"2024-11-12 13:36:08.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/13"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-12 13:36:01.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e28b23f69ebe4a951f345a05192816fa1288050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e28b23f69ebe4a951f345a05192816fa1288050"}]},"branch":"refs/heads/master"},"dcee8826242af308dc632201349cdafed2caa564":{"kind":"REWORK","_number":14,"created":"2024-11-13 13:49:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/934116/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/14"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-13 13:49:37.000000000","tz":540},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dcee8826242af308dc632201349cdafed2caa564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dcee8826242af308dc632201349cdafed2caa564"}]},"branch":"refs/heads/master"},"6bbc4788ff1606d1be896018e725892553e15d4f":{"kind":"REWORK","_number":15,"created":"2024-11-13 13:53:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/934116/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/15"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-13 13:52:56.000000000","tz":540},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bbc4788ff1606d1be896018e725892553e15d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6bbc4788ff1606d1be896018e725892553e15d4f"}]},"branch":"refs/heads/master"},"4d400fc72d26f76116f753ef92b24a802084bcea":{"kind":"REWORK","_number":16,"created":"2024-11-13 14:13:23.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/16/934116/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/16"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-13 14:13:06.000000000","tz":60},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d400fc72d26f76116f753ef92b24a802084bcea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4d400fc72d26f76116f753ef92b24a802084bcea"}]},"branch":"refs/heads/master"},"7997a199acada382b0c61dc84aa956e4c65de986":{"kind":"REWORK","_number":17,"created":"2024-11-16 14:00:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/16/934116/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/934116/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/934116/17"}}},"commit":{"parents":[{"commit":"a9c03bdf1e7d5fc049efc7053185fc11cc292d04","subject":"Merge \"Run pyupgrade to clean up Python 2 syntaxes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9c03bdf1e7d5fc049efc7053185fc11cc292d04"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-11-05 11:14:17.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-16 13:59:49.000000000","tz":540},"subject":"Remove deprecated AMQP1 driver","message":"Remove deprecated AMQP1 driver\n\nThe AMQP1 driver has been deprecated for several years, since 14.4.0 and\nis no longer maintained or widely used. This commit removes the AMQP1\ndriver from the code base.\n\nReferences:\n- Original deprecation commit: https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a\n\nChange-Id: Iaeb52791008a7a6736c99459f66d2bdbb2dea17b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7997a199acada382b0c61dc84aa956e4c65de986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7997a199acada382b0c61dc84aa956e4c65de986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
