)]}'
{"id":"openstack%2Foslo.messaging~934138","triplet_id":"openstack%2Foslo.messaging~master~I353778c9a063cc8361a7e86591e2050ecfdea79b","project":"openstack/oslo.messaging","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-12-09 14:13:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I353778c9a063cc8361a7e86591e2050ecfdea79b","subject":"deprecate the executor parameter","status":"ABANDONED","created":"2024-11-05 15:06:01.000000000","updated":"2024-12-09 14:13:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a1a81f51bc5b9d8da64d6987b12a005d533c18cb","_number":934138,"virtual_id_number":934138,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-05 16:42:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 16:42:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a1b6450ef526af4a45b38108ee95f34018b6625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-05 15:06:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8047bb7acc9f7bfc68d0be132baf7408fc1e01c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-05 15:06:45.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"4813b4b01a2d2deb77eaa258fb78579ff8f5ba86","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-05 15:18:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca1d6871873f0a4e07f1cf74fc379f7be4f4efaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:42:17.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/855f19c43bb04fc78d443d3488d6632b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8a40e31dd134ec69ba6f71522066147 : SUCCESS in 1h 30m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc72229dbe14ae0806e0ee50e85171c : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb2fa9b4c8d430f9ac8851983afa50c : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e795bf4e215244f8b5deb1083410fbe9 : FAILURE in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4af4404563524263afda84873305bbae : FAILURE in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85b00652e084ae0b69df4fd5a9b1ca8 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06dc229e867b4e1b9013143a601d256a : SUCCESS in 4m 58s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/65b760d56dcf4c3bac2d6ee9c20adfb1 : SUCCESS in 5m 56s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d1263484f7f34b5184b77952238eec16 : FAILURE in 7m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0cdaeb87ac524ca2b739bbf428c104dc : FAILURE in 1h 29m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d68069ba1a3248e089c649f8fa9332b5 : FAILURE in 20m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/581ef41d5b4242518aa51b0a33becd08 : FAILURE in 1h 06m 37s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d094ec4ba3b348f09a41cca009a9ceed : SUCCESS in 57m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf7eba09350f43dda2c77b1f98f1501c : SUCCESS in 57m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1a81f51bc5b9d8da64d6987b12a005d533c18cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-09 14:13:29.000000000","message":"Abandoned\n\nmerged with the parent patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82f367dce4424169c5f88ea683f88bad23ae925b","revisions":{"82f367dce4424169c5f88ea683f88bad23ae925b":{"kind":"REWORK","_number":1,"created":"2024-11-05 15:06:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/934138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/934138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/934138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/934138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/934138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/934138/1"}}},"commit":{"parents":[{"commit":"5cd72c206940e5f88f95212f62b588eecfc2990e","subject":"Deprecate the Eventlet executor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5cd72c206940e5f88f95212f62b588eecfc2990e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-11-05 15:04:16.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2024-11-05 15:05:28.000000000","tz":60},"subject":"deprecate the executor parameter","message":"deprecate the executor parameter\n\nthe executor parameter is now planned for removal\n\nChange-Id: I353778c9a063cc8361a7e86591e2050ecfdea79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82f367dce4424169c5f88ea683f88bad23ae925b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82f367dce4424169c5f88ea683f88bad23ae925b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
