)]}'
{"id":"openstack%2Foslo.messaging~936847","triplet_id":"openstack%2Foslo.messaging~master~Ib04346eef7e88d01e6c6bda79e10708cac553321","project":"openstack/oslo.messaging","branch":"master","topic":"bug/2077152","attention_set":{},"removed_from_attention_set":{"36083":{"account":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"last_update":"2026-09-16 15:56:06.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-15 21:03:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-16 13:41:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-16 15:56:06.000000000","reason":"Change was submitted"},"38329":{"account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"last_update":"2026-09-16 15:56:06.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-09-16 15:56:06.000000000","reason":"Change was submitted"},"38493":{"account":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"last_update":"2026-09-16 15:56:06.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-09-15 18:29:01.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ib04346eef7e88d01e6c6bda79e10708cac553321","subject":"Fix rabbit consumer reconnection loop","status":"MERGED","created":"2024-12-02 15:33:17.000000000","updated":"2026-09-16 15:57:26.000000000","submitted":"2026-09-16 15:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"936847-bug/2077152","meta_rev_id":"8a5fd4f0e5ff03c43289fb520cc0ee9bb27b445f","_number":936847,"virtual_id_number":936847,"owner":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 15:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 21:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-16 13:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-15 18:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-16 13:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 17:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-09 08:26:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-12-09 10:31:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-03-27 12:51:28.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-04-28 12:12:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-04 09:01:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-05-04 09:01:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-05-09 16:02:56.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-05-22 15:44:42.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"CC"},{"updated":"2026-05-22 15:44:56.000000000","updated_by":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"REVIEWER"},{"updated":"2026-09-09 10:55:13.000000000","updated_by":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"reviewer":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"state":"CC"}],"messages":[{"id":"28a21f75f0473b4ee067c7857e1714023e149641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"date":"2024-12-02 15:33:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fabe401b699c79d6f647193f10ff0895ca8a4c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:48:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/443618e6c1c049f7b7e9bd4862587162\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95d1215250c14e6aad3da8306c81f9ca : SUCCESS in 1h 51m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/735dc7792bff40d58306c8b17879c669 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39a655658a74ddf9db8dfecd5aa3f49 : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce944f78dcac49e8a01f340163d4e27d : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e66298efa094d26a50ef583ff301dc0 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07a91bfd1bc4da382b421b345e87cb7 : SUCCESS in 3m 22s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1530c3654d1b48679efc94407a308d7d : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/dd1251bb8cc047f9a8dd44fa8a7c93c4 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e3cb72cde6f94c119d7103b813aeee9f : SUCCESS in 1h 53m 58s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ba1a5c95e8e3497da5f3343bc643401a : SUCCESS in 1h 58m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2f3c30ee3dc1488fae1e4f1ec6ea01d0 : SUCCESS in 1h 08m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c84a26dc4d824ecd85feff20eae01057 : SUCCESS in 1h 05m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f34950ca3da2937e0f61dd8363efa85400b83e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"date":"2024-12-03 08:43:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdde5a713d0239a43b2dddc0166f27d28886bb15","author":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"date":"2024-12-03 10:21:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02d80b1f374682dfceb2fdf974090c02613ffa40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 10:47:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f1c7cf8542443d4930c618ac8554eab\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc856fcb6cdb47308807e7a0f8b48e0a : FAILURE in 19m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/601f34ac1d3149ebb3d073417db361c9 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1811308bbb4e51943332fa7b9fc4cf : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e453ca37460b4589adc096f3f390776d : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d6e845135654daabb3d3848698f612e : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a5e5a7fa064a9f94efdce0ee20f150 : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/fd4af7e00c5b4a50ba40e64d7316be7f : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/834dea3e88014db0a6fa5fd6b9aadf39 : FAILURE in 6m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/74cbf28a3eab4badb7a3e8c994e50a6f : SUCCESS in 1h 50m 40s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2c5db75ccb84440b89d02cd08e775c7e : SUCCESS in 1h 53m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/eb429c147fe84fea9fd912ee95bb6fae : SUCCESS in 49m 12s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6500e74dbd45424aa98afb2566f9bb4d : SUCCESS in 1h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e11aaa3fd65267a6474ffa1cc8be7e949ca73ee5","author":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"date":"2024-12-05 12:47:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2adfaa41959795e5d1a3de0fde6bc16c513c8f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-09 08:26:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f2cdf08094e36a00ef6dda835b01561c2bc7082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 10:22:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a12d13351204d5bb5e7bf1697f0a029\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d72b345c4d742dd88e0150f62ec1a3c : SUCCESS in 1h 19m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06a5e7964cd94b70aa6f54969e127072 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9be0a1712e84066b56eff367c1e041e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0d21c1d814a4e0297de6705089722f7 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/546a430d969c4604a3323e8743f763da : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3372f73d38cf4d9bab65798b53592204 : SUCCESS in 2m 57s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/4f4f9d403d4a47ca95accfba44600186 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/245cec9818b547e0ab2e4c189a251434 : FAILURE in 5m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c457ab69ee03430996a4216d4ba1268b : SUCCESS in 1h 44m 00s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/939c78f6e3334b9fa822b5ce24a0734c : SUCCESS in 1h 49m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b9a83797ce9243b6858c3f0d8d40061f : SUCCESS in 59m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/952cf43851bb4ecdad6f1fc5d554bee1 : SUCCESS in 1h 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b30011dee346b118897242965a1dd4b61345dd82","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-12-09 10:31:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9c4cdd13ec1f4f2642e1d083f3ca96e7dd1058c","author":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"date":"2024-12-09 12:41:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5e757930ff17e14c0d4ce41807561a24ce24018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-28 12:12:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d40dba4de267331b19324d8ead9e4b51c84d3805","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-28 12:17:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"035dcf54979d507320fdcee2294a5b7f89ee243a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:54:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85d45b6314d048098a941425ae4c60e1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8984ea8e51142c888bf21e0521f81cf : SUCCESS in 1h 19m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ddd2b5ed5364e878ce53732cffe607d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0c0ab677514f77a7e0e6307fa8085c : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef426931aa48453192624ee206d59e34 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e29a5ce261524bb29ad19aceba7f2454 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e4d1174d9ef484eafab99d04b7359fe : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e548dae75684001a0108e616f5ea09b : SUCCESS in 3m 30s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ae100d357e5f461ca0a420ab72b5263f : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/22c667e1ac584e87bd281276e65f5e9a : FAILURE in 16m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b6242daa752c47d695de1809411a53b2 : FAILURE in 1h 27m 06s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c2fadab866bf43c08db7611f071b935d : SUCCESS in 1h 39m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/086e6dfa32bf4eabbea47031587260aa : SUCCESS in 47m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c82898f7ce874c0ab842469276025c4a : SUCCESS in 1h 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2e4eac7590d6bddce903cf1b6a42350e2b7de096","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-29 11:23:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"32f0c53badd235c4036accbac89238e514d6bee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 13:35:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddf65ea87cb8442ca8f47eeaecd02a48\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a3ccf28a33c4e1e86aaccdb0bcbb635 : SUCCESS in 1h 55m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aed2bd417c844a9a97afb1441ebc24a : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ec745b10b248708370bf7b99d4aefd : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf03ab694b5c44b8bf15e90290e311cc : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61eac3c235da48a9af22f4a65834c069 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d7fcefbab7f4880863e3d2f06cd5713 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e36dd8aed4424e8ef78e549e9a070e : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e3343cf1ea414378a0e5509b6616d440 : SUCCESS in 6m 12s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/016bda15972841108de101e324afd878 : FAILURE in 8m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ec48b3d4f06f41ffbc35abfd5e2dcd28 : SUCCESS in 1h 45m 58s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9f6e74bd45194331a19436777472d1cc : SUCCESS in 2h 06m 29s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/81d1ae37adcc4746ae84f94adb108a4c : SUCCESS in 47m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc86cca51f7243e8bcb7275d1db75ffc : SUCCESS in 1h 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b1152955cdfa1b0186163275d7769d7d3daa3f32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-30 11:49:22.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f395056f2b2d38a81bb4e99f25265e6ef874b2c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-30 11:50:59.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"99ff6680bb05291bb28664757012e2c64a0c9e6a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-30 11:51:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1cb27fef91d522b587e4d1e7771c63d676254e46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:03:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab27b243f3c494dae764537f43799cc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/656b7289f2b042ecbff96ff23faf820e : SUCCESS in 1h 06m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b755346351c4e8dbabe32a80665f644 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cb2019a22b4c9faf3b63d42af1a1a1 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8dff0846b2c6416b93c144d900aa848d : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33cfd38d9c4a42e1831461b695621426 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c87ef061d71a4499bb5b90fc52431b4f : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27c4cb21a2145029dc0d4eca72b7aa1 : SUCCESS in 5m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/19980a01b77c4f0eac068ce75092c021 : SUCCESS in 4m 42s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/369592527cb149088ba75efacd8e7241 : FAILURE in 7m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c1783843e36345a7b1ac53122f09b690 : SUCCESS in 2h 06m 54s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d9ed9003452c4a1e9cfb34adf0c1f653 : SUCCESS in 2h 00m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2cd0000457524e74a09d4e44757d49ca : SUCCESS in 1h 14m 12s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/170039eae20047049d53841ecfeabffc : SUCCESS in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9a1f5a353d835357c321b2a6be413163edd234af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-05-03 16:56:49.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6173cdf91152e758a65e352066cf1c14a5ddbd3b","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-09 16:02:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c463d2728a17403ff47d0836a9fb68ec3457336a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-13 08:30:28.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"11957eea79e12fca1109df47e64e06f63ed7622a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 10:56:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e8f73551654834b153344b7a47b6db\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3bc4ab16b3148c7890fafbfa6154200 : SUCCESS in 1h 50m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ac97ec5abb64e33a38a345b25344eae : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77186ed7fbf4c09b49090e6e59c3f6a : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16182974687749cf9dd220104ba6f816 : SUCCESS in 7m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfd9ad3c5b854410af1cef3e834b882f : SUCCESS in 7m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd48d874e74c4d0fb30cf169f7676a14 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f1927aba5e42158563fb83b585dee9 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7f955291cf4696b4017b15cd0c99c6 : SUCCESS in 5m 11s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/d89bf7a4fe864d98942b564fa22b70e7 : SUCCESS in 5m 00s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/72173317685f462789f188995bd34ca1 : FAILURE in 17m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/39f803d907e34c0cad75c86e9f768638 : SUCCESS in 2h 06m 56s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c6c76a4ccfe044d79138c57f8b725b9d : SUCCESS in 1h 43m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4f869b6e7f374fd380276927d06ca7ac : SUCCESS in 59m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b508db34d3d042608285fd21062adcaf : SUCCESS in 1h 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d7eb870c8af728cbf6880b87b8f00d585f6edd54","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-05-13 16:57:43.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e469b0d8d4d8d898c31a0559325521362464b93d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 17:43:28.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"34cc161fdde85b9774a463f9447ceb1a75545a55","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-22 14:59:15.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cf0af716d322f46b745bf1b32ff455b0153a8295","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-04 15:32:17.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"57182b7466ebe83f744bab26e485ef4d8a12f080","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-05 11:38:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8685ed2094efec5feb92130d14fed855f5e486db","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-15 17:04:27.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"afe5f8cae8ec5eac9995fe7712ebcfda3b95c678","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-15 17:05:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"12599bf33c2fd71e526ddb75484849ff2b8e285f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-16 11:48:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2162a0ef15580a1038bcfafd17f75d2d83f00a94","author":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"date":"2026-09-09 10:55:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"78d2009085d4de9ca48d6a585a03013646b72b03","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-09 16:56:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1599715e6c1bafacab5472194288a91a758097f4","author":{"_account_id":38493,"name":"Dmitrii Shevchenko","display_name":"Dmitrii Shevchenko","email":"dsvkrab@gmail.com","username":"dsvkrab"},"date":"2026-09-10 06:52:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9fc3c59011a84afaad344338cde470e59e4c8a51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-14 12:07:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"ba22a9f7806d2329a84434bbffbe53e9c4b187ad","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-14 12:09:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ba4b93fbb97aeb5e20f5e49c9cc61b65fcba772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:54:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc70f89ec39d4e3db23142b328ef35ad\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d2e7584ae2244d692d1ad195652e9a7 : SUCCESS in 1h 30m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bca9921dbea4fa38dc4c79792a2a2cd : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8cbda0c1ae4da4b284d5560b4e24df : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88901c058123434da81c5b33e377c050 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11d16673af7c4e089e9d661bbe09bf3b : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8feda354e924410187b43918af32894c : SUCCESS in 8m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0d6b73ce81c8471a922b1c0da514c8fe : FAILURE in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be82d101da684f309b0c4a2f8a54c46b : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af46f93bfcc048da849aeec71a11f91e : SUCCESS in 3m 22s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/9c56bf541e834d70963f9a8825dae64a : SUCCESS in 4m 55s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/29954075231b44e7b03f411b4bda6695 : FAILURE in 7m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9f7e2f95cac54767974518a1649dedeb : SUCCESS in 1h 41m 14s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7b6dbcc6d2b3477da96b97fce8741556 : SUCCESS in 1h 18m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/79a55a4f40e94c508e1f09d2092f5fbe : SUCCESS in 1h 06m 27s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0369d7ff369741fbb8f19f39f430904b : SUCCESS in 1h 03m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"395b162179f6945c661f1418818b392035929df3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-09-15 18:29:01.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef9cc12a2b3f9d44ed9601067e9477e690aedd7a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:03:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"f1c8eb5c7607bc2710878778603571f84396645f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 13:41:57.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"381a878a5da3ba9c99918df1a6df7a982b2ab12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:57:48.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"671eacfb4713a12e86633ddaddcbd7679e3e6adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:56:05.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b220a69861149f1bddecff8d46b055c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed83b0f45dc54d3fae46ac3d4fb1dbb3 : SUCCESS in 1h 43m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43b8ab59990b451f82221039d2856e1b : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05efac49ac834cf39ad67a055eea2ec3 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/098bf00c34ec4bf7b3cb7609015b2066 : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb002d75c3814fc28d0b1931c5045594 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b5103bd7504a58b386f8650277a72a : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec268d045fd4021af5e5a228dce376c : SUCCESS in 4m 28s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/218d5964231748a2bda6177618fb39c8 : SUCCESS in 5m 58s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/038e6be132114417ab141d33885f0e44 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":7},{"id":"4c27fa14cf8ec0c69930ad0f630e1c2961a5accb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:56:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8a5fd4f0e5ff03c43289fb520cc0ee9bb27b445f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:57:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e910bc2d2344901aeeacfba421c7040\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6ddf70f68b4e239090ceaf4bb97cac : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b53730d307783ec435a49ce2cb753a6b4f384005","revisions":{"84c89ae6a18934b27bccbbb4137f0a34fa91a3a3":{"kind":"REWORK","_number":1,"created":"2024-12-02 15:33:17.000000000","uploader":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"ref":"refs/changes/47/936847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/1"}}},"commit":{"parents":[{"commit":"0deb6828683db276e7e3fadff0994b4fb04a347d","subject":"Merge \"Revert \"Remove explicit pbr dependency\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0deb6828683db276e7e3fadff0994b4fb04a347d"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:32:52.000000000","tz":180},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/84c89ae6a18934b27bccbbb4137f0a34fa91a3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/84c89ae6a18934b27bccbbb4137f0a34fa91a3a3"}]},"branch":"refs/heads/master"},"b434adad782963460852c24cb6a031f2b00127c1":{"kind":"REWORK","_number":2,"created":"2024-12-03 08:43:14.000000000","uploader":{"_account_id":36083,"name":"Nikita Kushnarenko","display_name":"Nikita Kushnarenko","email":"NKushnarenko@itkey.com","username":"NickKush"},"ref":"refs/changes/47/936847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/2"}}},"commit":{"parents":[{"commit":"0deb6828683db276e7e3fadff0994b4fb04a347d","subject":"Merge \"Revert \"Remove explicit pbr dependency\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0deb6828683db276e7e3fadff0994b4fb04a347d"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-03 08:43:06.000000000","tz":180},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b434adad782963460852c24cb6a031f2b00127c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b434adad782963460852c24cb6a031f2b00127c1"}]},"branch":"refs/heads/master"},"8d12df2ea60caf60bd2896a2f67652587cbc4292":{"kind":"REWORK","_number":3,"created":"2026-04-28 12:12:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/936847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/3"}}},"commit":{"parents":[{"commit":"624f8572485676e2d17b63580948b71fb5ab2b02","subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-28 12:12:24.000000000","tz":120},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nSigned-off-by: Nikita Kushnarenko \u003cNKushnarenko@itkey.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d12df2ea60caf60bd2896a2f67652587cbc4292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d12df2ea60caf60bd2896a2f67652587cbc4292"}]},"branch":"refs/heads/master"},"7f4b6c6f973a8f5f3e843c2807b104f9561d174f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-04-30 11:49:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/936847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/4"}}},"commit":{"parents":[{"commit":"624f8572485676e2d17b63580948b71fb5ab2b02","subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-04-30 11:49:22.000000000","tz":0},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nAssisted-By: claude-opus-4.6\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nSigned-off-by: Nikita Kushnarenko \u003cNKushnarenko@itkey.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f4b6c6f973a8f5f3e843c2807b104f9561d174f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f4b6c6f973a8f5f3e843c2807b104f9561d174f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c8a8b05818a3244516a9f33f7c3b2186c660e411":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-30 11:50:59.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/936847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/5"}}},"commit":{"parents":[{"commit":"624f8572485676e2d17b63580948b71fb5ab2b02","subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-04-30 11:50:59.000000000","tz":0},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nAssisted-By: claude-opus-4.6\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8a8b05818a3244516a9f33f7c3b2186c660e411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c8a8b05818a3244516a9f33f7c3b2186c660e411"}]},"branch":"refs/heads/master","description":"Edit commit message"},"11e949b861794af4a31cd7829ebe6bc39b25391a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-05-13 08:30:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/936847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/6"}}},"commit":{"parents":[{"commit":"624f8572485676e2d17b63580948b71fb5ab2b02","subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-05-13 08:30:27.000000000","tz":0},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy`.\n\nAssisted-By: claude-opus-4.6\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Nikita Kushnarenko \u003cNKushnarenko@itkey.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/11e949b861794af4a31cd7829ebe6bc39b25391a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/11e949b861794af4a31cd7829ebe6bc39b25391a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b53730d307783ec435a49ce2cb753a6b4f384005":{"kind":"REWORK","_number":7,"created":"2026-09-14 12:07:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/47/936847/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/936847/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/936847/7"}}},"commit":{"parents":[{"commit":"c59264fcef24c1d192788f813c5079fb73d26386","subject":"Merge \"Resolve deprecation warnings from setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c59264fcef24c1d192788f813c5079fb73d26386"}]}],"author":{"name":"Kushnarenko Nikita","email":"NKushnarenko@itkey.com","date":"2024-12-02 15:00:57.000000000","tz":180},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-09-14 12:05:54.000000000","tz":120},"subject":"Fix rabbit consumer reconnection loop","message":"Fix rabbit consumer reconnection loop\n\nThis patch fixes the issue of infinite reconnection in the event of\nrabbit host failure.\n\nBy adding a connection check before consumer call we ensure channel\navailability. In the event of multiple rabbit hosts, we initiates a\nfailover to a live host using `kombu_failover_strategy` and\ncalling self.connection.maybe_switch_next().\n\nAssisted-By: claude-opus-4.6\nCloses-Bug: #2077152\nChange-Id: Ib04346eef7e88d01e6c6bda79e10708cac553321\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nCo-Authored-By: Nikita Kushnarenko \u003cNKushnarenko@itkey.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b53730d307783ec435a49ce2cb753a6b4f384005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b53730d307783ec435a49ce2cb753a6b4f384005"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
