)]}'
{"id":"openstack%2Foslo.messaging~938536","triplet_id":"openstack%2Foslo.messaging~master~I9d67deff03993829c3a7aacbde5e01a897870b26","project":"openstack/oslo.messaging","branch":"master","topic":"bp/oslo-http-driver","attention_set":{"36129":{"account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"last_update":"2026-09-10 23:33:43.000000000","reason":"Someone else replied on the change"},"10115":{"account":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"last_update":"2025-04-04 01:41:03.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-09-11 08:48:18.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-26 08:39:20.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-04-11 05:29:13.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}},"33451":{"account":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"last_update":"2026-09-10 23:33:43.000000000","reason":"Someone else replied on the change"},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2025-04-04 01:41:03.000000000","reason":"\u003cGERRIT_ACCOUNT_36129\u003e replied on the change","reason_account":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-09-27 11:02:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9d67deff03993829c3a7aacbde5e01a897870b26","subject":"Add Broadcaster for RPC Fanout","status":"NEW","created":"2025-01-07 07:47:38.000000000","updated":"2026-09-10 23:33:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":30,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b4e2165b55466b51f4b37794776ca598c0dd8183","_number":938536,"virtual_id_number":938536,"owner":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-17 07:24:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 07:48:28.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"state":"REVIEWER"},{"updated":"2025-01-07 07:48:28.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"state":"REVIEWER"},{"updated":"2025-01-07 07:48:28.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2025-01-07 07:48:28.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-07 09:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 07:52:14.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-03-05 09:30:04.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REMOVED"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REMOVED"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"state":"REMOVED"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"state":"REMOVED"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REMOVED"},{"updated":"2025-04-04 01:38:44.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REMOVED"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"state":"REVIEWER"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"state":"REVIEWER"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-04-04 01:41:03.000000000","updated_by":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2025-04-09 09:29:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-08-15 22:34:34.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2026-09-10 23:33:43.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"CC"}],"messages":[{"id":"895985274d4d424bf2fe500d55ae01406e109cbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-07 07:47:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"327b53096fd7c5e8c97d83534ec69e1f9b866bab","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-07 07:58:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae3e6b8e2e232682c048fa521272f0fa52dad535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 09:48:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ef2a8ac29a347e089bb6cf00d893544\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f5f52bf3cf904e67ba3c9ac8938e91b3 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de22126a49414e60bb44505d6c539488 : SUCCESS in 1h 37m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d173f044b40f40308bd18b97642fb81e : FAILURE in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09597ffcd390487f9b7704d2fdd9c0f0 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12abad119d104d4eafb8ed6edc29c3dc : FAILURE in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e506c512ed64121bb21f3ce4444ad03 : FAILURE in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c908fedd2a41406385bbf417780e55f7 : FAILURE in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94749bdf02ae4f7eb965558d6869cf40 : SUCCESS in 2m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1f4f7016328040e88445d7a045d32b17 : FAILURE in 3m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/2831ef7e9049457f9dcb0be8429bf442 : FAILURE in 2m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ae17d29704a24afd9927c2e6294bc73f : SUCCESS in 1h 36m 39s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/29baa42b46ee4c88b568c03ed1fbcb6a : SUCCESS in 1h 45m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b3126d3f1796478cbafedea4261de88b : SUCCESS in 1h 00m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1eaeced6ba1a41dfbc70f3224de45ea0 : SUCCESS in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47399a2b433bd38faaed55624aa9737421125fd9","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-01-14 00:58:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"808949bb11f3cfe47943a8d73b0f497026b6d188","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-02-10 03:09:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0bc4296c5189bc463ffe234ce77b04ed335ab20","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-11 07:52:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2aa726bc3721d8fd6f8af90d0bf8763ccd90762a","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-02-14 08:45:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dcf2647b84a9fa61fb69305bfb53fe93b8a88fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-05 05:19:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"97dea05704a576478b3dbb01e1def7d862f225c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 07:14:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54a3e504d504487ab6aacc6871c97e2f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b683b79998034930a1217f177d8942ae : SUCCESS in 2m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d7ac455b86a4448a672f385df978076 : SUCCESS in 1h 29m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62518500dd75409d922e992a181f0e80 : FAILURE in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbaf13d848e84f4baf9c0b2104e47246 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fea2f3f78daa414983e390f232193e53 : FAILURE in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55dd48054f014a0ea3f630b562e02423 : FAILURE in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e0eb1041364061b06decc477c0e3b1 : FAILURE in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c766d8bd01a428fac6454b9b28443c0 : SUCCESS in 2m 58s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3731b3daa5bc401fa88b2517e12e7ee2 : FAILURE in 2m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5b08dfdaa14c4873af9ba1640d2857e7 : FAILURE in 2m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e064529d5b6548fe82e615621fa3d371 : SUCCESS in 1h 33m 13s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3d0b337533fb4ec1bd9aa665d34e35ee : SUCCESS in 1h 47m 20s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6c1d3861500747dda0b1cf119e04b826 : SUCCESS in 58m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/017deadd4f904bde8575d1740bfbc52d : SUCCESS in 1h 08m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abfc158b7f86418fe52775bd7d04ccd2a6c3fb0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-19 08:34:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3214e02640f5b9583b28ee8e0fa5b97554c7d69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-19 08:40:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ac35cab4a9230dd8ed3d9a06a37afd65e81f224d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 11:20:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a65c08bee0a048ee8fdeb16a666e18bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4ec47a2c0a9f42bcb0b6f361a0ac8dc9 : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b99a8bb2c7a4b7ca29a5e3a64ad2cb3 : TIMED_OUT in 2h 34m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05ae33f70c3a4ebdaf9d2e0d8b364d6a : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0a3368516c4d5fbc55dfd0d55ab7d5 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7c028ce80b54df696333fd39b144734 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e3d3e0ddb5f4f1da3013b265d2ac6b1 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3dfd8f48d849bdb37632649cc251fa : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8326af81715b4357a4fd57efcbc73bb6 : SUCCESS in 3m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/76906eb1a3b74193b0312eb95cbbcbbc : SUCCESS in 8m 28s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4e47b71f1ac244129bbd6ed89bbada3d : FAILURE in 7m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d5b27929d963482d83d13833f0b35bc1 : SUCCESS in 1h 28m 01s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d2750504c34e4bebbfe28632e2331197 : SUCCESS in 1h 32m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f57bddf9346445e590fda22b4cf0adda : SUCCESS in 51m 15s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c3ae037a34543bcb72fbfbd213c99f5 : SUCCESS in 1h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d9c55e68491c2174c3a6fa99b83e07359874b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-21 11:03:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7a59af3f143cdb9b65a4d3c31e985fc5e8007956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:37:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b4716e769e48889b79966e19f468d1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85354dbe114e418aa61fd02cdb4c74af : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/caaacda778aa453abf3502e3027fc5b3 : SUCCESS in 59m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e219466ddf4c3aba46850522766925 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea8cae1aa2a4cc0924b1f5f5c97e41a : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6989c3c3a4c643dd8169fb2dd2e7129f : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5cc754ab6b3645568f31ae838696b85f : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c95eb86b0e44484aa5580b4295a585c : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd616d7dc51b429f84af976abab146ed : SUCCESS in 1m 48s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e3f861c049c24e7eac37dd63e7e6c98a : SUCCESS in 4m 46s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f5dbc80090fb48de84942572a4418f22 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/85525e1fffe24a189dd1c09e39165588 : SUCCESS in 1h 32m 57s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6d055f9faffd41e78976890036765708 : SUCCESS in 1h 26m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8b55999922ac4817b01477b0a068d0df : SUCCESS in 42m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/43dbc3db28684c438647e19cdd563093 : SUCCESS in 1h 05m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d7cc9dcac701d9824a4c6396d1738983c8b5341","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-24 06:29:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd0dce5f64db578424533e4a22e2d52f149d499b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-28 12:42:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"472d160d7c5ba1fbbeee06aa1c84a94962dd9d21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:46:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96538e0faefc47779b536e6692f3ac30\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5a2f5019795f45a8affcf72d32a91878 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/385f5653e4cd4a56950e6136e3a05109 : SUCCESS in 1h 13m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cecf252d80549c3a286fde44589e96d : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9177e8f5f0f4cf69836f91754e0c30a : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/425807ecc50e4b6a84a17b09cb94ec82 : SUCCESS in 8m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ad6fa2c1d8f4e4b9a87c044cf102f98 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e26f007019c4b09b0ef5f92ee9896e9 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0fef0d5d684559a533133c4cc79639 : SUCCESS in 3m 45s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/44c33b6c97e847d5abe5aacb51cd619b : SUCCESS in 5m 08s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/64938d748a0e4640979df798577371d0 : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9b7b9ec24e904fd7b31184d29b27ad7a : SUCCESS in 1h 59m 09s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8236761f72604e85ac8881d3ad806fe8 : SUCCESS in 1h 40m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7822df642e564174a109ef9b673866cf : SUCCESS in 54m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/02a766bb4cba4444951fa1fd1a6b3a18 : SUCCESS in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8774e99d7ef121ccc26b9b9a08e953386cd85b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-03-31 08:19:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e2c1f23ac5abbd8884cc278579af875393f6892a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 10:08:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4d0418646a244d993dece8b5f61e8f3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e18b15decf6147dea58f9f1d5922dd98 : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aa30992537443b4a8507d093a59bc0f : SUCCESS in 1h 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16d5df1a33804636a953c1390095ab93 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4d0d99b1c274fb2834a3f4d19ec99a0 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b178f422b64ddbaadd85714558bf37 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ea98e3423f349b8b93c91d650135440 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68804d91ce0e49eaa834651cd6d93a18 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/81e1fa063bb14e99adba553de08663b3 : SUCCESS in 4m 27s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0c20c6bc9498488ea422d416d4a0a5d1 : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/554c88ea825140b9ac9d9248075fce14 : SUCCESS in 1h 30m 04s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c03818cf35dc41e688bd8039ac792e7a : SUCCESS in 1h 17m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/13f25a830e194c4dbbb8bae590733a82 : SUCCESS in 43m 54s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8173a2c60b3b43ca9ba630a58550da2c : SUCCESS in 1h 03m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ae250b760a0c803f926caadb732681fd3143b49d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-03 10:46:00.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"3976316bfcf7c1b7048f085e215c2e59b64b5763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 13:37:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85bca9ca1e1d4cb6978743cbd0831707\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56a90c517f8e4d16b2bf634bc302b09c : SUCCESS in 2m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c14b76eb5a8a4818ad64bd81477f9610 : TIMED_OUT in 2h 38m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a25518bdb0248aeba68dd5ac6d622e5 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3871268cc8b44642b63c0bfb09226477 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb0007ad686e4ef3a9551d14934e3816 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02b8d784e7604d8c9a43fca88a63bb4f : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5ddf86161ef4c51977de139213e6339 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4736f30c9fee44f2b4d356098d58fc31 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c52954aa239447b4847963180a7f7128 : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d31f085b9b4f47158e02fdeb5132e212 : FAILURE in 5m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f4a5dc93f4314344ac3ba8a4e25e25fc : SUCCESS in 1h 35m 27s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/041982577be74b48aaecac6b0568a6c7 : SUCCESS in 1h 17m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/97ff752254a34df0a0b89a4110279e29 : SUCCESS in 54m 15s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1ef545a49a74db2b8542fbca504790f : SUCCESS in 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7f3599fe9e364071c51e02af960d066aa66acc12","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-04 01:38:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a89eb493d14c82f7308c56b53192736555dd71ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 03:23:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c0e51851da4ae493d8f8734464981a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3242bfe84c3477798d30a00e70184ee : SUCCESS in 2m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3aa2ff30aa75436790e220f222d6e3ae : SUCCESS in 1h 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d9530522984c0e8106f0ea7c1e79e6 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d63ba94a2af444eb30be30c1a96cf51 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf246b23be974321823c2eeae1984045 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f570acdf748472d9e7f9ea21da17b6a : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e9326e6c4a34c3791e3068f28cb93d1 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed81d710a2c47c2bfc49530031c4df8 : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/14ff8aad846243bea5f800a6db51438f : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f53e141db88f4ab782dbd23b63bcc243 : FAILURE in 4m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b702622aeae442e2b13efaec5978522b : SUCCESS in 1h 43m 56s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2946015e195842feb1ebff58d73ef649 : SUCCESS in 1h 24m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/85aeeaedef9847b5af58a20d29c584c2 : SUCCESS in 46m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/081f79ef2ff44b08b6935a59256c8706 : SUCCESS in 54m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"062492bfe01a4259d99f2b0ac306ad5afa194bba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-04-09 09:29:08.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b1a5e8181cc9839e7dcba5400fb0b401c5bd5768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-09 16:13:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0299d50a461836737c5b4984e549d21e6bc22c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 18:12:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4ab3f768d404ae2a3e8f587049be340\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27c8d3f50f14431dbb6679fd160fb572 : SUCCESS in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42ceb7a1a7b542c99c1b8a81f6d57212 : SUCCESS in 1h 33m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43736da5ebd148f9bbfb644e690f9528 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a966a134a1c487996fdbd603a2102d1 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3b3cde59de74911a131da1a1e7c2190 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b010e6d8ae0f438bbf35f6a27e3d1055 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/feed742dd7bb4e11b250c3722409c455 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8706e9b434fc451faf54b64c549de556 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/469d7bc253774f9ebef14753d0f0d3da : SUCCESS in 4m 11s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4e254f6203ea4877a7b3bf3a5a6af946 : FAILURE in 5m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0676356dde76473ab43746868548bb49 : SUCCESS in 1h 56m 54s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ebdf6826816240b4bbbacf30ec484e98 : SUCCESS in 1h 23m 15s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0a65dc7f16c74845a50c5ba3ee020ae1 : SUCCESS in 56m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/469a0f5069da42e3961a157661c2c0d6 : SUCCESS in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3d51d6aa74f2e81de9716e7272afb4aa49861f81","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-11 05:29:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7da32b7471100d29942773c5d9eede5fe54c6122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-04-30 10:14:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"18a0bcccd05cab05bb9badc5677cef0300e82006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:17:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d46423acd48439eae7b10ecd2b29dd6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/caf579f9211041438cd8e4b16e995c59 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77689bb1e1f14febb9aabcd595fe162f : SUCCESS in 1h 35m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdb525add27a4d0f9610b5aefe7885be : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02efc05d13a9443697ac293dd1b3b68e : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acb35740244a44df84c8cef6cef5a615 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2b77a0a2dc84b9cac43d84e8393b4b9 : SUCCESS in 9m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/803250303c1e4e3e8f75d0a1a8698e82 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05387eb2366432ca75d763b32cf5995 : SUCCESS in 3m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0a40f7fefabb4c4780d0b2e92b86a2dd : SUCCESS in 5m 37s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3877452d19a942ca8db68a0ae0c133f9 : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e8b978d5d9e94cfc82faea146ef84e89 : SUCCESS in 1h 31m 45s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e57d308652194a23a31e4ed571206da7 : SUCCESS in 1h 42m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ea9f826687154bb4b86fc887034d0a9c : SUCCESS in 55m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/16d172878ca44cabb0602df3eeb97611 : SUCCESS in 1h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4815ecf86d86a5612e71ca4563b6ae38d9841d59","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-05-29 09:01:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6ea8c7d4033900a9132146afaeb63e051e5531bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-07-29 09:10:49.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"fc6ffd6875a174b3941c453c5c6a3ab895b2b624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 11:18:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e560a4199b0a4dc88f57ac02b4fedd84\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/338b86ea15fd4d3abc80418c548b6e8d : SUCCESS in 3m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4443364b9cec4682878fcd28dfd43ec1 : SUCCESS in 1h 42m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2bff736245e4160880a83c16038c208 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80dfaf0e7a1d40c68202abd803bb31a2 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/917f13395be04e06bee8f0bd95567afe : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de24d7b6181a4c6a9436f2e0aa85a039 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b647187a0a147539c0ec3b6fc8f3b31 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30920e8a09044a0a8a2cebd491252955 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/749db2d330144b75897a273562de4fba : SUCCESS in 4m 38s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e7129e43caf14ccabb026637c53a14ec : FAILURE in 7m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/30778a7c4cad4d2ca97c8aa4d980b314 : SUCCESS in 1h 49m 05s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/daee76f7d64b4ab88911da01ea81e904 : SUCCESS in 1h 49m 15s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/aef8db3022594691896f244ca11b1215 : SUCCESS in 58m 35s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ca20cc3bba744579928473d63e54512 : SUCCESS in 1h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"22a605a492bb476c55a93b79b96dd1b8efb1bbaa","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-08-15 22:34:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"97c99c28d41d5b2f714b0265aca65b7dd7dfe940","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-11 08:48:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"39757877b53871f24c9ec0b65ec4116f8d82a689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 04:49:40.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"d15abc736291d78c2039c29024b80343a3e1b2f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 06:00:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c0cd51f81f442e58318439d355b2fad\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b66f8d60576d4e70b7822cc8b9b96f42 : FAILURE in 21m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e203672c75a4244a11a98da2adb33fd : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51c8644df204f4e82472a556f0ec7dd : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21137a3a2e0641d481990eb68ccf592b : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abce015dd3b84423988921d9680657ff : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbb8c7581a1c4ef784c997cd7f2aca38 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4892d8927f2f4c2d869142b13e703a75 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/794c5cdac9dc4491b3d3dc7463b4d17d : SUCCESS in 5m 18s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d19bc3e5ecf54f118ba0cc0e6e7f0d69 : FAILURE in 10m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/086f2a5e37d64f3bbed65eb8e65c5d57 : FAILURE in 13m 13s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e829c4f589d34af98d1e6c77a71500cb : FAILURE in 17m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fbe99b1df7b1434a9547ce9589b93e44 : SUCCESS in 55m 58s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5190cb2e9ba44491a5a8dd4191f4c161 : SUCCESS in 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ffe34145bc3fa5fa3c428966e2b33bbb352acc3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 09:23:14.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"94dd0b9ccdeb344b57356305343159bfcebb8485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 09:32:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"67b29e686fe1514e0a59b3c60c749953a1ef176a","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-18 09:34:58.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"152bf2f56ca99637824c790ed1fc406736531fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 11:24:19.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14026d2b03b7462abddcd5158c3bef82\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3eca0e61cb6b4c4a97700399402e1ada : SUCCESS in 1h 35m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35707d0d7c3a47dcaa4bbf54944cc2f4 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151f8eba5a3c4692801ad502a300c1e0 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/310c29c1dbb14fe08e4beb224f7c777c : FAILURE in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/121ed1c4b9084a8b923ba8c430900278 : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45e30375829d4459bc471ea3ad0a8c37 : FAILURE in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de0284312844ff7a87cad8b4c17ba1b : SUCCESS in 3m 53s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/4b627b9d9e0d4d26b4ea03aed91e702c : SUCCESS in 5m 05s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a5d53425b87f418f8136bf4c35ed300d : FAILURE in 16m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6e21a049ca8943d68874d63e15f6c45e : SUCCESS in 1h 51m 15s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b04f691e12464a29982b8ba73c1c2c89 : SUCCESS in 1h 21m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/35a29a118f0644da837a3326d5ae0a99 : SUCCESS in 32m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c6da4581c844bc3ba8e2d8363343a36 : SUCCESS in 1h 06m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"03309bf9ad6da109d8f6d4c5d2cd8f1d47b7f6cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-19 01:37:23.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7e1163ec68b560a050bbe81f030f23ce45d5f8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 03:30:03.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64703e3af4dd49508574616e9cdaf8f2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbea24a0abb14dcc8af2a25c8d36ba60 : SUCCESS in 1h 44m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d11fa0064a09495ba3c0f5e938ed08fb : FAILURE in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0326289534bb41b8ab0d55c20253ff3d : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0af265224234e1dba546519073dc8cc : FAILURE in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb034112f67942f880ece2c82239d586 : FAILURE in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6432a92f2c4244bb00b1f5af8d9002 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef932220fa294a37bd70f33a0855c1a7 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6120a50292524d95932f271d86df2af8 : SUCCESS in 5m 13s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/a090f7601187468c91d19d99d76f6cea : FAILURE in 7m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2db62d30b2ca4240a3fd8cd75b626cc9 : FAILURE in 1h 16m 12s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/23aad80660c94103af7f5a8dd4813c3e : SUCCESS in 1h 46m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/14926b730a714e45b0ae4c505a8c6589 : SUCCESS in 47m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/04ee8baf8b9743b19f5b8a77dd8319a8 : SUCCESS in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"626905ccb4e2defa646d8075134c476dbf4e119e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-09-24 04:31:01.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"df10dcca7b81a28b76da6045ff3d32b39bef63c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 06:15:34.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53430a420eb3493e9259508d5c94ae00\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3e80d06db144bb8b91db86ee13a876a : SUCCESS in 55m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b448d5c8cf1541a092dabf5a3b15a4dd : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87048a926fd9414585dd06299e676cca : SUCCESS in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa181813b16d47e9b545f0a624209a94 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c99221b0aa914c54a50445f4418d69ab : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbebb911c63d44a4bc310cff87852bb9 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fed1c429874449380673110107a18b5 : SUCCESS in 3m 08s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f2f2e2b9e27c41ce8482f531178c8c50 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f97d0c64c9a94aa6aa3d9993d626cd37 : FAILURE in 14m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ffd62138376940c2bca5883e065a58d8 : SUCCESS in 1h 27m 46s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/167d4719fe4d47a988cba29048174efa : SUCCESS in 1h 39m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9f4971eabd794e6ab58fc8a361c3841b : SUCCESS in 58m 57s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d2e7c9a04de452db9503cd149dafb78 : SUCCESS in 1h 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a47e49344166d99f26ec267345e7ea988b835ab5","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-09-24 20:52:37.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d215d3a89101df45f139db3a943e5c11a8c3397a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 10:51:39.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"d3b555acec5563db1527b879ab03be0be2ea6944","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-27 11:02:59.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"f3dfe3437db48ff3fd77505a4a976e0f9917199a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:39:20.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"97d175e4e97c5339f65f0a2893ea88d7bcdb927c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 08:50:49.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"ea74a4b7cf27444a1fa591a5ec3e5d3b67f555c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2025-12-26 09:19:44.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"751ee97a28ca240560b475abede2c55bd62cf98c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-26 10:57:22.000000000","message":"Patch Set 19: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f23a43dc7000442890cde069f4338381\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/08b21ce240ae4c59b91808adc650f97f : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5cf52914b476422dabaa6e1865717474 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/941287c53b22465589e6b2d62605aaa6 : FAILURE in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7efd490333004c9cb6f073205ea7339b : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e6e2690f8e4d218ee8e9d4c4839832 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c73e245489124530ba50b0cc3ee2ee46 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8488706be7c7404b8839cd49d65f052d : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bd9769f1644914ad916236bceb23b9 : FAILURE in 3m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/325fedd00b6f4776a0ba4afbd975a414 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/6bcdbbb93b8440c9af80a4a3f7acaa27 : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/347f3452a2f540f68cddbbfa87beb393 : SUCCESS in 1h 21m 22s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5e2cfedc560647fdbb612ea65c69907d : SUCCESS in 1h 34m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/22c87f37607541a48522f86e95696e80 : SUCCESS in 59m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/571459b7e71a4abd91ec48fc7683ebfd : SUCCESS in 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ac58fa54a87b044cfdb8bf8113acb1315d91749b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-05 09:09:45.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"0d2062b92135cf095789329f9c6d1c30b69a0be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:12:14.000000000","message":"Patch Set 20: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d22687a2439490381cf6b912d25e308\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac908e9dfc7e42f3aede4c0883b83341 : SUCCESS in 1m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3976cd1a7f3b48eaa5cd6c33cc13b7f5 : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbb391d3832c42a998379270c25e1604 : SUCCESS in 55m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0767c7286c5e4176be61858ed17cf96d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1521342671b44f82970937894a5f0c48 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a151748759374dffb289e1ea03fd9ed0 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/636cdebfd0e04028ad311d8956e94c4c : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a48c53185648d0b271827620e4d7d3 : SUCCESS in 4m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e5259c8300864deda8421030901dee12 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/7914daa8d8ab437e9e1c7d8c040be2b4 : FAILURE in 7m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8c3e0126ff994f83aa9a7b5211f70f85 : SUCCESS in 1h 49m 39s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f429ddc8238c458e9604d19b4ab8632a : SUCCESS in 1h 41m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/971eb29ea71047b598720111b5e589b3 : SUCCESS in 29m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef6b4b1fe9924c03a2bc0e187536df6e : SUCCESS in 1h 12m 29s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"d26db756aab95d7cd449e861cbfd5215f3a08857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 08:30:42.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"0a3493c72f3da43cc701f1d42cb80898d898838c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 10:36:53.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfbe975b14094cd0a9064ff4ad4ff9fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b60bf05826e43be93400b36cc900278 : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/41a20e85f28142448b62ec22a71fe674 : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/458bccbfa00e44fbb8201d9ba3c77fd4 : SUCCESS in 1h 46m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db2819e2a4cd44769056863f49251686 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf930093ec4141518a0dcdf52c924441 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6abb8f23fc0412d8df6b01eac80a7e6 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46124002e2704991bfad449679cff341 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000f0a663062405f8138dad54b9e3a51 : SUCCESS in 3m 33s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/beb373fdc5ae4c28a40bee5a6f58c454 : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/bbea494cfd0f411a9b6804c42cf4339a : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6a70db39886141ce90a5295de0829f34 : SUCCESS in 1h 29m 45s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1677ecf817b246c9844af6aa12986ca7 : SUCCESS in 1h 44m 17s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/24d37a5f565c48c6b3ef2fee0f6c1c9b : SUCCESS in 1h 06m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f5f7d9ddbdf436dbb84eadc3ed7d25a : SUCCESS in 1h 09m 36s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c651e871c7bdd17003be911ac36bc462a7c29e07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-06 15:51:27.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"0f00095153f7f27bb5c9d7481595f4002f59dd6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:47:20.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cc51972e5044fe9db71ce784a59934\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7604177200f343f2852a17b19518fa17 : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b417113b5234292a0bd885b217e143a : SUCCESS in 2m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68b0e7151b3d45e696d5c4492edde569 : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e516765497dc42988613a96d22e9a9e8 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d27e803895246f29b5c74066555e812 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f42295664ee47ef8e985cb91d07012a : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4a4bb4eafeb403182c85181abd0f5e5 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7348aad19f1640219ff6c934a3b2c20d : SUCCESS in 3m 04s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b8b54d5c115c47b1b09cccb29bb4e32c : SUCCESS in 6m 38s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/5a4b710cc1894e4587a8984d93c4671b : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9d9193131dec4b429600aeb87b91ad6c : SUCCESS in 1h 22m 41s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e6d9cf3698ad464d8881f3001b7f01d5 : SUCCESS in 1h 48m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cdc1a76d25d043bb8123a66e1bd5194d : SUCCESS in 58m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/40ee4a8a2e6c43f6898acc230cceaaa0 : SUCCESS in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"78840c66a82f48f11d827da935d83644117836c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-07 16:06:38.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"e11a98e98ab224035458d5226b266c4e66837e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-07 16:14:17.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"9c6e9f4b457926b939d5b352803368705c971cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 17:51:18.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69513751a47a403984625fe212dd7460\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e79d98f45b064752b0f11413643540ab : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/08d39e866890490f89acbbae61a09d39 : SUCCESS in 3m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9758d28ed8ea4899925cd9f846188b82 : SUCCESS in 57m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05838bcb398c47369b4add0637b9ac9d : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27affe54c04e4eb68f44b7d64d5739f4 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa775c6200e24918a342f96c2f67fbbd : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4c67d52985d48b7ab26169ba0811f9d : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef331b95c974b109a5b6b0552defbe3 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65c2106009bd43eebf53403aa35a1f33 : SUCCESS in 3m 35s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/26da4c24f8d9403dadfd1a1f572c17a9 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/233dfda6e28a415bbdbc85a4026e4a12 : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0a8e909402a84029872a64661cd19b8c : SUCCESS in 1h 27m 44s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ec31aa245cf8424aacb8449670996441 : SUCCESS in 1h 29m 28s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/99ef2506e9dd48b7afda33c7454a6a16 : SUCCESS in 1h 02m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/021b309f1fcf49d4823c81cac23aa934 : SUCCESS in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"eac1bcd1d161dffa36b649ab1afe85b6ec06012d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-08 01:12:40.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"8a74bf85df299874676b1d0e961967cc8f4f56cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-08 01:53:16.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"c32cba9c232b02549fc0c31141edeb57d967ec8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 03:50:29.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e18ff73db3845e2b6062e61c67412e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f5cbd15b89884eafbd5d179a422ae9d2 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aba050ee60be4791967d2b8395d99c76 : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/584e8386101f4d8b86f09b2cc701edd8 : SUCCESS in 54m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7775c3cdefc4f3d8a3405515f168b3b : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92d95c9ba284cd495e74ee889a81929 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdce22734a3b4c2fb4a74fcf37bf2032 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a937e7ddb0174cdb9542894e0a3d81a1 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13bce4ca35e409686e494f3bae8d75c : SUCCESS in 2m 02s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c5c3f4bff82f4e0bbbdbd053e87add61 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/8acf9499d9fc4e2aa82ddd84f1889367 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/3d97468bfdb447948dfed329aec62d07 : SUCCESS in 4m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bf2c4e3311aa47aabef501916c127d8e : SUCCESS in 1h 49m 31s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b022eef0bbd04b5f99d5a58effcbdc37 : SUCCESS in 1h 46m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a7516df9a03d4765a02d146310097a5c : SUCCESS in 31m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/13dea01228184be7805ffad49d9bb42b : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9861944e30f264b1ead80417a526643c83fb14b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-08 16:38:26.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"b8a525797e2a59531efc69bcb49e995b2215d3f5","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-08 16:40:55.000000000","message":"Patch Set 27:\n\n(7 comments)","accounts_in_message":[],"_revision_number":27},{"id":"863e3ee56e180d427068c12289adf5aabc40faaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 19:15:15.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40dc8b0f2b5a42198f9a303212e59bc4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0a9e44fa0d624fbd9afb920575c732c3 : SUCCESS in 5m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/329956a215604e75a48bfe48f8b20381 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0b3afb901e540aba124874c011c9971 : SUCCESS in 1h 41m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d3d301fd144b8894bc71e56ef7bbc6 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbfcb4f52ff43c6b9068c0a936bf44f : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71e278c8d21d4d8cb3497814986df46f : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55f511ae3fae4faca8e8adf3477802bf : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7380cd5bed4f4d709448a904c0c685a7 : SUCCESS in 3m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a4345fa5a9f8493f941655b368177077 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/13a80ff7d85b48c58bed61359395f277 : FAILURE in 6m 57s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/7c6f1f40297d4ff890e612fb888f098d : FAILURE in 4m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f306283ae2224a6b901be43fde83ca3a : SUCCESS in 1h 49m 42s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f0df3ff5054e4312993c4e0c0f66fc5a : SUCCESS in 1h 29m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/620377e06a824d42ae8153eb28172cbd : SUCCESS in 1h 04m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5b56c0668724126ba9ffe63da07a9f2 : SUCCESS in 1h 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"44ee4d905fddecc4e796c8d25a2a27c9ed0ac8a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-15 09:35:57.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"7c4f0cc9f41abd1102de27d19ca32efc4e492871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 11:55:33.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/311ab18a5e5a441ca8f2acfa7ddda191\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e49463dacb3049fb8ed7ea8220ac3817 : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/83228cf8effc4c5393badda532aed230 : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b55232479b2c4a24ae1d0907a72fac3b : SUCCESS in 1h 23m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acedac829e164ba7b94dad9d3c4089e0 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0475378d73a74c4b97263b8f59ea0f61 : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da92f6adca49484b8a6aa221d30b1b78 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f17a9a0d4dae4b1490bd00f1b2a5ab70 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b019f3b98941d281fe9372b2ed8a44 : SUCCESS in 2m 28s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6a518d0eb537424aa302da73d86ce5fd : SUCCESS in 4m 34s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0084ef9dcde245ee90556c48d3a0082e : FAILURE in 6m 46s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/ab0f71726e10467ca7accac8ed8402f6 : FAILURE in 4m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/18e7cf1fe9f9490aaf69e72d07723b0a : SUCCESS in 1h 29m 25s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b148e93c4a82418e9344a337017d3fd2 : SUCCESS in 2h 07m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ef3f6f42f17a4329a5b4e85a4b6c3c79 : SUCCESS in 1h 05m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac514161771c427693503fff1777d100 : SUCCESS in 1h 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"d9de3bf7fa49f6cf2d779ad5f693e5e5253cd1c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-16 04:36:53.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"0e71a6998ab59f776a9b485198eef58031c74f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 06:35:31.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616e8e8f21674ac280f129104b8b8ee3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d07d8fe6830498882f358f540275bcb : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb2029a17fee4725a4f2043448cbabb1 : SUCCESS in 1m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ddd3ee49a7b40ec9e67eeb98981c8c5 : SUCCESS in 1h 39m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394f0a69c553453d8bc9ba2f12cba30c : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e092b45696e94f609b141ab0e3e7ec4f : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8f4b05418c749afa6e016a7a8f8e02d : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f8ad1804cb5457f95fa7ffd575b3ed9 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1838e5b6f07433d96d37b9a772bb9b0 : SUCCESS in 3m 32s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/6aae3b09304349e9ad697b64bf54dd8e : SUCCESS in 5m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4be3139a9a0848f69612fa4ab2709abc : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/06705206075c4e69965c67394db236b9 : SUCCESS in 5m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e4e8c3c8b402456289f4ff61189d560a : SUCCESS in 1h 39m 02s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/31828682b0844cc08e61475eb59b28ef : SUCCESS in 1h 53m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a86731e403c64f04b5f8760a82e6bc1f : SUCCESS in 52m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/67d6a2ea56a84e55be46b21e8169a69c : SUCCESS in 1h 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"6629cb68e3cc16f68dff00f62755978aaf642709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-19 14:48:53.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"0c831fa6aacf932dbf126a500dd8ec43d7ed7d6b","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-19 15:08:04.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"c225ff9d38fcb8ac53e5ce941e9d305156e82020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 16:59:22.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6673f59a3ed44988ad3171bcb851b80\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2e08fe33236148878051259044fe790d : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43189e8fac2d428ab68e1dc3c8bc2899 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8568b1237c043938a0cc9d32661943f : SUCCESS in 1h 51m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/772bd60a6f9041f09d9ec2f860048214 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba98b75f131a43b09cdc22ec6eae9ed1 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18eec8e0c8d940d5a3845c49445dac77 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95dde36cb4414276b7fb996c79adf796 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee794f8c5a04bed897c19308bda9676 : SUCCESS in 3m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/b95ed7abf7814b238c86666ff7a1e3b1 : SUCCESS in 6m 32s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/22b8dc4eb87f466895bcf49bd3b61b1e : FAILURE in 8m 05s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/0b357cceb13e454cab1787c498878058 : SUCCESS in 4m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/719f003d690e41e0945d509796400605 : SUCCESS in 1h 48m 23s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cb6e6397baa9411da2a14677e673e69f : SUCCESS in 1h 50m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ce08f9fdc374461c9bc5918a742f0ac8 : SUCCESS in 49m 52s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/25e8d811fbf54c6eaa7d7ef8d7b7aeb2 : SUCCESS in 1h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2eb1c7db2f60a1d691554860c7fd4fc9beab6af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 07:45:37.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"97d6c4d66589669e55515e09f5b0be6ae7f6c333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:45:43.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97191db276034b5091acd1aa2f7ae91e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa9157fdea3e4c6da61ee6018a581d4a : SUCCESS in 5m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/854916390c96451085bfd7f27f7a815f : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/733b3c048b884bc0afe87e4b39008790 : SUCCESS in 1h 05m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed4af6a141b45c88068afe886675545 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b243a0176c42fbb9cfda611d51d472 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b236ad35e1044c399ca6cfb1b57b93e3 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a17e8aae8f5406e9f8d0d9d4f8127fb : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0a35c1325c47d79534b49cc364cef7 : SUCCESS in 3m 01s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ad9325ffa7e94e3eb9a0bca4dca86ce0 : SUCCESS in 4m 16s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/52109cd866e849668189b46326212181 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/e2312351d4a845a9a356f4713b09a52b : SUCCESS in 6m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3fb471e07a5e4a2cbd6dca3988b6979f : SUCCESS in 1h 26m 13s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/710a450eb7604030b9dc8e54f984f4dc : SUCCESS in 1h 52m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9645e5188e8b4f50a02860446b2f0e58 : SUCCESS in 1h 01m 00s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf2f2b33c24e4371b78e3f5089aee274 : SUCCESS in 1h 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"052340dd5dbc3f374e2f8966501684b0a64e04d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 11:23:17.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"99746d2bdc22c481723a66e6753f0202b6bfe26f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 12:33:33.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"c99ab45b3d26c5f86dda2a31a33e5ca849bab64a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-20 12:54:21.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"c3745fb1eebc89cf309d3a674abe58812cd25fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:02:14.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b39db8522b4eada290e94ddd99b48d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0d079664c6994ad688c6dd8484105ed7 : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc806875e91943e289fb0f9993fdc0fd : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0d36129e8924db78c974e273918c5f0 : SUCCESS in 1h 46m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba8ba242b0634b408249f72baaab57ab : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5310e7ca1b934e8ebd23bfbd3368e3df : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e29fd66c47ea4edd9c01e1b839589f89 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d2b8fe2e6ff445a9f9f924d75febe6c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747a5f64d1494b639de8d0b7774e6640 : SUCCESS in 3m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/984d0f2465e442e99b83f8cc7f80da60 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/444adf5237cd43728e39b22bd2e3e665 : FAILURE in 5m 26s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/53462a9cf74b4a1b86fb3f161225a265 : SUCCESS in 4m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/22a6802fdda842629c0d32d97881734c : SUCCESS in 1h 47m 11s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8ceb4ed6a29349e3ab3ae0c81509f55a : SUCCESS in 1h 28m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/05530e9e75514dbeb228b67b63e9940e : SUCCESS in 58m 02s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7ca46a86be54f2ab016e3778301b47b : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"8c8a0522c6f668cb6b2301f8aa255e74fba48330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"date":"2026-01-26 08:24:40.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"4fdc92f8c749f057c6434054d69ef04bb1d5ada1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 10:16:38.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aeca7be5b1d44e98b4de5302fe929c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20b5fc916b38436b93944099c6d97283 : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf6901f6c39a4bcd8277edd4adc3e74c : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e959934181464ceb9a2d2e362e91a105 : SUCCESS in 1h 43m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/553fc82511d84fbcb7c6422cb3691fd6 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535eb2dfaa0f499c8e7cfd4fa17b6cc1 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0828ac4ceef740248a38198375de62fb : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6f7adb01a2f4f02bbc78ffa66641a02 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887744314cd64b17887294826b4ff44b : SUCCESS in 2m 50s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/778fd92d2e4c4aaa865d77afc94568df : SUCCESS in 3m 55s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c54f0dc222b3433c82ab1a929c589171 : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/97f6ea604e7342a5bf87f413935cbb59 : SUCCESS in 4m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9749ea80102a4bd486269c6494510d27 : SUCCESS in 1h 35m 27s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ecd37130cac34f8594d4907d0d6c0b89 : SUCCESS in 1h 30m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2fa04e1f760544ce90d96020e83c4b4d : SUCCESS in 29m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7063c63096844a1b70e183877d95ba5 : SUCCESS in 58m 06s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"f063e188dc7fd5a32e91180b7e758b911e2bc82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"date":"2026-06-15 01:10:43.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"2f2c4a14892b674a00a8a4f2e7c4401166308ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 03:04:41.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2f0f040d8e42acbc921dffa4e23cf9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a5dd48ea9022449bb127f9126cdc2724 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b1f23aac5ce43ea9820a8553e3ded85 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a2bd21c0bc8496c93770572157c594d : SUCCESS in 48m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6da89b8f8f0496badc6261a00a685c5 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705f1ac140f04b01badabcf62f5b9b9e : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e178352c372049f8b1d91db1fe64c909 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ba934951a4d4b7593b68f60b9cb4b2b : SUCCESS in 5m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7afbc295c13b42e1a6b5c83cbfa8a20a : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de10adbf025448da8831efffa8ea9e5 : SUCCESS in 2m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51c8860e5c4549e1bdc901fd34b26d5f : SUCCESS in 3m 44s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/648cf9b43f424089a2248934d6081fde : SUCCESS in 4m 52s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/01cc5abfcfad4cf086f0262ecadece78 : FAILURE in 6m 57s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/902594d7f0c445dfa3d37e09f2afdefa : FAILURE in 1m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1c5c4faad7ed47f198240440eb23432d : SUCCESS in 1h 46m 51s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7d3cf77b115c4e76b796e4ff7f6099c3 : SUCCESS in 1h 46m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bcdde59d9cbe4e81b40f2de558650907 : SUCCESS in 44m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1769b1c2d6954969828764b1b370f487 : SUCCESS in 1h 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"267503a176e27df204f37db65f34aadfe59fafd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-11 21:36:13.000000000","message":"Patch Set 37: Patch Set 36 was rebased on behalf of \u003cGERRIT_ACCOUNT_33451\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"_revision_number":37},{"id":"5d3a5d912670915399e7e3ab99a9e98a3cdac039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 23:16:19.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5ade0a3bf414e39b5d3f4f782484dbd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dae6575564f1415d8b21d7b1806e0a99 : SUCCESS in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e0ee64678de34a558fb399c8b84a8dfd : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/672be1e673bb4bdcb635b4222809952a : SUCCESS in 39m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db1d52ccc81341c791def2c44289714d : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e3f2c3e05cd4dbbbf079c08f5e2c781 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a49c9d896d34d289e399b7c7382c241 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7e753288b9a427db95733330b457256 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5ec0fa751f204881a07b274855e38b73 : SUCCESS in 8m 07s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/036c5337b04041d3bb07cc3ff8a5eb92 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fc04048fde4fe1b80cf4b088421ee2 : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9e5c5358f0d4381a12b6874e32d2387 : SUCCESS in 2m 43s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/065f12738a834e3fb5d5d982b5571e82 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/82423f6d820b4b73a69eb0e46f0e6930 : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/b3a6f0c618774ee9adff75b48e9dc70f : FAILURE in 3m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/62318cc77ca241e2aa3b31ad189f7587 : SUCCESS in 1h 28m 37s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ca4f2ee52d2740ccaf1727d6a7639ce3 : SUCCESS in 1h 36m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e6dcd82545604ecb89dd64ac7279c493 : SUCCESS in 1h 04m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/81954e2bcdab4cb39c449cd25f6a6cac : SUCCESS in 1h 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"a8f39f423350aba4c314a144117d1e4ca599a955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-07-17 05:37:02.000000000","message":"Patch Set 38: Patch Set 37 was rebased on behalf of \u003cGERRIT_ACCOUNT_33451\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"}],"_revision_number":38},{"id":"7a9c44fc4442c383137d894835a3989a9c4bcc37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 07:24:27.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b983fa5966a04ab48b02443e5f0133f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8713c7a3d65c406a953cca5f9065710d : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ff3e05dfb3c48369f439e2806520092 : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2a45a157ae944cd92827d51713c873c : SUCCESS in 1h 25m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad3f52c1db59452c821e34d48424ac27 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2d5c67dda3241a1b421cde22bb35adf : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0ef93a39cb343f79b6d424a403e72a8 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47e7cc6728e2441b85009e959fcca964 : SUCCESS in 6m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d64ff40ef8384fcc96963586a106e2ed : SUCCESS in 4m 44s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/92b6f7cb5bf8494d8fd511d2912b3565 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c45ebfb2b54537b9e941be0fc7fe31 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35ab9bed64b3477e80e781cb848b0b72 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/e68d5e51e6804c23a852bee636fab9a8 : SUCCESS in 5m 24s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/d3f3cf0819254fe98adaacda4ed6b4d4 : FAILURE in 2m 47s (non-voting)\n- oslo.messaging-tox-py310-func-http https://zuul.opendev.org/t/openstack/build/3d0ed3a097ca4903aaf6ada3c21eed8c : FAILURE in 1m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7160ddc1e2f64545b271fd62997b625b : SUCCESS in 1h 40m 35s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ff410a0160b14cae8edfcba07bce47a4 : SUCCESS in 1h 33m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c84670248bc8469c84a19a9c0e765927 : SUCCESS in 1h 05m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/835a38a7b86c4472a80df79863850f1b : SUCCESS in 1h 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"b4e2165b55466b51f4b37794776ca598c0dd8183","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2026-09-10 23:33:43.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38}],"current_revision_number":38,"current_revision":"0196b876ed2032aeb1b090710f7452f505bcf639","revisions":{"7f3c49209c5c7d9039c304f6cf49f0db1c1c533e":{"kind":"REWORK","_number":1,"created":"2025-01-07 07:47:38.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/1"}}},"commit":{"parents":[{"commit":"2a904bda5bf1308f01bf215adff29f3727a9c1af","subject":"Add OpenAPI generated code and functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a904bda5bf1308f01bf215adff29f3727a9c1af"}]}],"author":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-01-03 11:11:57.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-01-07 07:47:22.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f3c49209c5c7d9039c304f6cf49f0db1c1c533e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7f3c49209c5c7d9039c304f6cf49f0db1c1c533e"}]},"branch":"refs/heads/master"},"afba90e32abf3d8c086ac7d9255ed32e00a84ed4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-05 05:19:19.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/2"}}},"commit":{"parents":[{"commit":"ce5f33834fdf732413fcedd62f62c5646ac57354","subject":"Add OpenAPI generated code and functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce5f33834fdf732413fcedd62f62c5646ac57354"}]}],"author":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-01-03 11:11:57.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-05 05:18:15.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afba90e32abf3d8c086ac7d9255ed32e00a84ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afba90e32abf3d8c086ac7d9255ed32e00a84ed4"}]},"branch":"refs/heads/master"},"b0c4eb4c2e1dd3a9af7e13887ae58e1d0f22a4e1":{"kind":"REWORK","_number":3,"created":"2025-03-19 08:34:25.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/3"}}},"commit":{"parents":[{"commit":"a7a64e0391f34b7349df8bbfc5b903a9702fedd0","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a7a64e0391f34b7349df8bbfc5b903a9702fedd0"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 08:33:36.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b0c4eb4c2e1dd3a9af7e13887ae58e1d0f22a4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b0c4eb4c2e1dd3a9af7e13887ae58e1d0f22a4e1"}]},"branch":"refs/heads/master"},"3917f1f39f0a23de6c55751b0bce5370a4670fbb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-03-19 08:40:18.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/4"}}},"commit":{"parents":[{"commit":"b4a5a13b4c36982c1ec7c811efa902f02f6478ad","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4a5a13b4c36982c1ec7c811efa902f02f6478ad"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 08:39:37.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3917f1f39f0a23de6c55751b0bce5370a4670fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3917f1f39f0a23de6c55751b0bce5370a4670fbb"}]},"branch":"refs/heads/master"},"689fc3a0cc6d26f878ffb2b81963505f6cca12a2":{"kind":"REWORK","_number":5,"created":"2025-03-21 11:03:04.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/5"}}},"commit":{"parents":[{"commit":"305882843e39202cb17704b57d856a546b6243bf","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/305882843e39202cb17704b57d856a546b6243bf"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-21 11:01:19.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/689fc3a0cc6d26f878ffb2b81963505f6cca12a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/689fc3a0cc6d26f878ffb2b81963505f6cca12a2"}]},"branch":"refs/heads/master"},"a4d4454be09e58e30d5aedefb81d002a93764b3f":{"kind":"REWORK","_number":6,"created":"2025-03-28 12:42:07.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/6"}}},"commit":{"parents":[{"commit":"0aeb6ca38bb31ca764779c8ac0760c71b5bc31c2","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0aeb6ca38bb31ca764779c8ac0760c71b5bc31c2"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-28 12:41:47.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a4d4454be09e58e30d5aedefb81d002a93764b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a4d4454be09e58e30d5aedefb81d002a93764b3f"}]},"branch":"refs/heads/master"},"12e8d52ac63a321f78267cb76cae1b06ebf514de":{"kind":"REWORK","_number":7,"created":"2025-03-31 08:19:29.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/7"}}},"commit":{"parents":[{"commit":"2dc7670f45afc30e663a831fb51b6d7892ed2ccf","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2dc7670f45afc30e663a831fb51b6d7892ed2ccf"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-03-31 08:16:53.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/12e8d52ac63a321f78267cb76cae1b06ebf514de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/12e8d52ac63a321f78267cb76cae1b06ebf514de"}]},"branch":"refs/heads/master"},"9c7138a1be778ce9bcacc162bd133a9849652cc0":{"kind":"REWORK","_number":8,"created":"2025-04-03 10:46:00.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/8"}}},"commit":{"parents":[{"commit":"6eaf38e4fb646fda6cce0e643e958a68fc3bd53d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6eaf38e4fb646fda6cce0e643e958a68fc3bd53d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-03 10:44:12.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c7138a1be778ce9bcacc162bd133a9849652cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c7138a1be778ce9bcacc162bd133a9849652cc0"}]},"branch":"refs/heads/master"},"3c72965e006f819563dfe19358820510d5f54306":{"kind":"REWORK","_number":9,"created":"2025-04-09 16:13:52.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/9"}}},"commit":{"parents":[{"commit":"6eaf38e4fb646fda6cce0e643e958a68fc3bd53d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6eaf38e4fb646fda6cce0e643e958a68fc3bd53d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-09 16:12:17.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c72965e006f819563dfe19358820510d5f54306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3c72965e006f819563dfe19358820510d5f54306"}]},"branch":"refs/heads/master"},"e0f15dde3c6332c7c2de567a51a12f704699d790":{"kind":"REWORK","_number":10,"created":"2025-04-30 10:14:34.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/10"}}},"commit":{"parents":[{"commit":"e93bdb884dcd242e0a20a14cbb0290b080e43aed","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e93bdb884dcd242e0a20a14cbb0290b080e43aed"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@linecorp.com","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"xiang-roger-wang","email":"xiang.wang@linecorp.com","date":"2025-04-30 10:11:42.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e0f15dde3c6332c7c2de567a51a12f704699d790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e0f15dde3c6332c7c2de567a51a12f704699d790"}]},"branch":"refs/heads/master"},"8d6e266ca6056188e63cdb9d17e3c333a808b8eb":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2025-07-29 09:10:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/11"}}},"commit":{"parents":[{"commit":"f16ab3a0506356969d4ce895e13232f6df7f9bcc","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f16ab3a0506356969d4ce895e13232f6df7f9bcc"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-07-29 09:10:13.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d6e266ca6056188e63cdb9d17e3c333a808b8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8d6e266ca6056188e63cdb9d17e3c333a808b8eb"}]},"branch":"refs/heads/master"},"e046a7229762e596a409a3424cc1ef6d83dcd5bd":{"kind":"REWORK","_number":12,"created":"2025-09-18 04:49:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/12"}}},"commit":{"parents":[{"commit":"f3f152cfb9aadfaba7f82da4c73ecc9848dfd5f6","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f3f152cfb9aadfaba7f82da4c73ecc9848dfd5f6"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 04:43:36.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e046a7229762e596a409a3424cc1ef6d83dcd5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e046a7229762e596a409a3424cc1ef6d83dcd5bd"}]},"branch":"refs/heads/master"},"d6f737a2c681c4363b9fbbba47402fd3e05a99fc":{"kind":"REWORK","_number":13,"created":"2025-09-18 09:23:14.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/13"}}},"commit":{"parents":[{"commit":"8ee1c2189fdfb849180ed12bb5bb8764ec58cf8a","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ee1c2189fdfb849180ed12bb5bb8764ec58cf8a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 09:15:27.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d6f737a2c681c4363b9fbbba47402fd3e05a99fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d6f737a2c681c4363b9fbbba47402fd3e05a99fc"}]},"branch":"refs/heads/master"},"09dda2e9a845df1f1aa9e6021b3e2601b1b64f96":{"kind":"REWORK","_number":14,"created":"2025-09-18 09:32:12.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/14"}}},"commit":{"parents":[{"commit":"8ee1c2189fdfb849180ed12bb5bb8764ec58cf8a","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ee1c2189fdfb849180ed12bb5bb8764ec58cf8a"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-18 09:28:30.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09dda2e9a845df1f1aa9e6021b3e2601b1b64f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09dda2e9a845df1f1aa9e6021b3e2601b1b64f96"}]},"branch":"refs/heads/master"},"ce704bd8ffadcef7f4d16a26f3056ed1f1af6242":{"kind":"REWORK","_number":15,"created":"2025-09-19 01:37:23.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/15"}}},"commit":{"parents":[{"commit":"2464e4e94afae41eab5419359139a6f1b4c281d5","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2464e4e94afae41eab5419359139a6f1b4c281d5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-19 01:27:23.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce704bd8ffadcef7f4d16a26f3056ed1f1af6242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ce704bd8ffadcef7f4d16a26f3056ed1f1af6242"}]},"branch":"refs/heads/master"},"232d5caf15759102011f3d599dbde81f3b7b5e7d":{"kind":"REWORK","_number":16,"created":"2025-09-24 04:31:01.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/16"}}},"commit":{"parents":[{"commit":"2464e4e94afae41eab5419359139a6f1b4c281d5","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2464e4e94afae41eab5419359139a6f1b4c281d5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-09-24 04:28:16.000000000","tz":-240},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/232d5caf15759102011f3d599dbde81f3b7b5e7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/232d5caf15759102011f3d599dbde81f3b7b5e7d"}]},"branch":"refs/heads/master"},"683aa685ee5506a8782c03f82eceb3033ee984f1":{"kind":"REWORK","_number":17,"created":"2025-12-26 08:39:20.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/17"}}},"commit":{"parents":[{"commit":"46987e447c021cf93c63e4ba0154638ad6f2832c","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46987e447c021cf93c63e4ba0154638ad6f2832c"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:29:03.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/683aa685ee5506a8782c03f82eceb3033ee984f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/683aa685ee5506a8782c03f82eceb3033ee984f1"}]},"branch":"refs/heads/master"},"79615966cc2b69a32674d6d534fa2b1ae0ea6f2d":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2025-12-26 08:50:49.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/18"}}},"commit":{"parents":[{"commit":"bae9c9a73a54183319f637b4b6b89d92c8ede882","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bae9c9a73a54183319f637b4b6b89d92c8ede882"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 08:50:21.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79615966cc2b69a32674d6d534fa2b1ae0ea6f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79615966cc2b69a32674d6d534fa2b1ae0ea6f2d"}]},"branch":"refs/heads/master"},"530f3f59b1a705d90648ae625f3b3f404267c078":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-12-26 09:19:44.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/19"}}},"commit":{"parents":[{"commit":"85149d90d0176de6b5b9b7ca9b82d5cf751c387d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/85149d90d0176de6b5b9b7ca9b82d5cf751c387d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-12-26 09:15:35.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/530f3f59b1a705d90648ae625f3b3f404267c078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/530f3f59b1a705d90648ae625f3b3f404267c078"}]},"branch":"refs/heads/master"},"9c7aa3f95fb24d51813d0fb26bb3957102e7eda3":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-01-05 09:09:45.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/20"}}},"commit":{"parents":[{"commit":"faf57e7c98c80939333e2eaeca274180b617f8fe","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/faf57e7c98c80939333e2eaeca274180b617f8fe"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-05 08:54:43.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c7aa3f95fb24d51813d0fb26bb3957102e7eda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c7aa3f95fb24d51813d0fb26bb3957102e7eda3"}]},"branch":"refs/heads/master"},"72da5d036383695c3147ccf3a27c64452b14a173":{"kind":"REWORK","_number":21,"created":"2026-01-06 08:30:42.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/21"}}},"commit":{"parents":[{"commit":"919573da77865f7bcd061fcee642cbbead3ff19c","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/919573da77865f7bcd061fcee642cbbead3ff19c"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-06 08:23:56.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/72da5d036383695c3147ccf3a27c64452b14a173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/72da5d036383695c3147ccf3a27c64452b14a173"}]},"branch":"refs/heads/master"},"c2f72c618e88f9db735911dc035134afadf969b6":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-01-06 15:51:27.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/22"}}},"commit":{"parents":[{"commit":"e4b064c17eaf469433e41a7eb78218f55e9734cc","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4b064c17eaf469433e41a7eb78218f55e9734cc"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-06 15:39:28.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2f72c618e88f9db735911dc035134afadf969b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c2f72c618e88f9db735911dc035134afadf969b6"}]},"branch":"refs/heads/master"},"c7e67ee5deba1b210057f271c8a942fe1125c3af":{"kind":"REWORK","_number":23,"created":"2026-01-07 16:06:38.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/23"}}},"commit":{"parents":[{"commit":"475d5fe7e3dda34434e51482ea76d499ec7f454d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/475d5fe7e3dda34434e51482ea76d499ec7f454d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-07 16:05:29.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c7e67ee5deba1b210057f271c8a942fe1125c3af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c7e67ee5deba1b210057f271c8a942fe1125c3af"}]},"branch":"refs/heads/master"},"e13e7e67a7d71e38594b0b5e6061e90258ed3e3c":{"kind":"REWORK","_number":24,"created":"2026-01-07 16:14:17.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/24"}}},"commit":{"parents":[{"commit":"475d5fe7e3dda34434e51482ea76d499ec7f454d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/475d5fe7e3dda34434e51482ea76d499ec7f454d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-07 16:14:00.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e13e7e67a7d71e38594b0b5e6061e90258ed3e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e13e7e67a7d71e38594b0b5e6061e90258ed3e3c"}]},"branch":"refs/heads/master"},"fe78e75c9377af4c0985d374f6b0203c8b1ae431":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-01-08 01:12:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/25"}}},"commit":{"parents":[{"commit":"767017e86adcd6a3db325798bfa904f67a1e4312","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/767017e86adcd6a3db325798bfa904f67a1e4312"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-08 01:12:21.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe78e75c9377af4c0985d374f6b0203c8b1ae431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fe78e75c9377af4c0985d374f6b0203c8b1ae431"}]},"branch":"refs/heads/master"},"de128397a71c87c0bb5e379f1eb8bc4dc53ced75":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-01-08 01:53:16.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/26"}}},"commit":{"parents":[{"commit":"61f5a7c6631b84b3cd733d92761049dd59b5ed5d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/61f5a7c6631b84b3cd733d92761049dd59b5ed5d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-08 01:51:21.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/de128397a71c87c0bb5e379f1eb8bc4dc53ced75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/de128397a71c87c0bb5e379f1eb8bc4dc53ced75"}]},"branch":"refs/heads/master"},"136484502637725e318b750216cc35043e050c5b":{"kind":"REWORK","_number":27,"created":"2026-01-08 16:38:26.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/27"}}},"commit":{"parents":[{"commit":"61f5a7c6631b84b3cd733d92761049dd59b5ed5d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/61f5a7c6631b84b3cd733d92761049dd59b5ed5d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-08 16:26:31.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/136484502637725e318b750216cc35043e050c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/136484502637725e318b750216cc35043e050c5b"}]},"branch":"refs/heads/master"},"768d759e0b62bcf4f84801a31e845eb632cc131f":{"kind":"REWORK","_number":28,"created":"2026-01-15 09:35:57.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/28"}}},"commit":{"parents":[{"commit":"76ce169d179207bb7f0af16cd9f514272ad782ba","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76ce169d179207bb7f0af16cd9f514272ad782ba"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-15 09:34:42.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/768d759e0b62bcf4f84801a31e845eb632cc131f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/768d759e0b62bcf4f84801a31e845eb632cc131f"}]},"branch":"refs/heads/master"},"9d0358ca5f996b72f4d7a0472519c417746e6b8c":{"kind":"REWORK","_number":29,"created":"2026-01-16 04:36:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/29"}}},"commit":{"parents":[{"commit":"76ce169d179207bb7f0af16cd9f514272ad782ba","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76ce169d179207bb7f0af16cd9f514272ad782ba"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-16 04:35:22.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/5)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9d0358ca5f996b72f4d7a0472519c417746e6b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9d0358ca5f996b72f4d7a0472519c417746e6b8c"}]},"branch":"refs/heads/master"},"b98881235c611d7fdb2c480c48123427b9f62aec":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":30,"created":"2026-01-19 14:48:53.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/30"}}},"commit":{"parents":[{"commit":"a5086bfcd91117a30c917541d1fe1c6698143b0e","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5086bfcd91117a30c917541d1fe1c6698143b0e"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-19 14:40:14.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b98881235c611d7fdb2c480c48123427b9f62aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b98881235c611d7fdb2c480c48123427b9f62aec"}]},"branch":"refs/heads/master"},"e6d93cf32e2e0021f64411414828c40cba707850":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2026-01-20 07:45:37.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/31"}}},"commit":{"parents":[{"commit":"12215747853a78a800a1e4e5be72ed01b7c0d4d3","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/12215747853a78a800a1e4e5be72ed01b7c0d4d3"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 07:43:26.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e6d93cf32e2e0021f64411414828c40cba707850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e6d93cf32e2e0021f64411414828c40cba707850"}]},"branch":"refs/heads/master"},"7bc163912c3ce4e13ab9d5e6f0b22eabe33b798f":{"kind":"REWORK","_number":32,"created":"2026-01-20 11:23:17.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/32"}}},"commit":{"parents":[{"commit":"d589e52f9b04ddd4d9ddaf8c40c6942584ddaebc","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d589e52f9b04ddd4d9ddaf8c40c6942584ddaebc"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 10:55:21.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7bc163912c3ce4e13ab9d5e6f0b22eabe33b798f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7bc163912c3ce4e13ab9d5e6f0b22eabe33b798f"}]},"branch":"refs/heads/master"},"c9909c4ecc487c4797da934a8514c4e6db799b7b":{"kind":"REWORK","_number":33,"created":"2026-01-20 12:33:33.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/33"}}},"commit":{"parents":[{"commit":"42b26a7e6ba99d11d9ca24b187e9a19a1ddbce35","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42b26a7e6ba99d11d9ca24b187e9a19a1ddbce35"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 12:33:11.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9909c4ecc487c4797da934a8514c4e6db799b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c9909c4ecc487c4797da934a8514c4e6db799b7b"}]},"branch":"refs/heads/master"},"29c847d7141f0bb995bf4b4599f34ab5876e9b6d":{"kind":"REWORK","_number":34,"created":"2026-01-20 12:54:21.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/34"}}},"commit":{"parents":[{"commit":"7de8ab6dfa57589a181273ef39a1682ca0309772","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7de8ab6dfa57589a181273ef39a1682ca0309772"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-20 12:45:48.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/29c847d7141f0bb995bf4b4599f34ab5876e9b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/29c847d7141f0bb995bf4b4599f34ab5876e9b6d"}]},"branch":"refs/heads/master"},"e8d7fb34885bd68ec9733b2518f219cd4aa4f2d7":{"kind":"REWORK","_number":35,"created":"2026-01-26 08:24:40.000000000","uploader":{"_account_id":36129,"name":"Xiang Wang","display_name":"Xiang(Roger) Wang","email":"rogerxiang.wang@gmail.com","username":"xiang-roger-wang"},"ref":"refs/changes/36/938536/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/35"}}},"commit":{"parents":[{"commit":"0789191e0eb13afdc850e9a1d53e3540234bd59c","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0789191e0eb13afdc850e9a1d53e3540234bd59c"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2026-01-26 08:24:22.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e8d7fb34885bd68ec9733b2518f219cd4aa4f2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e8d7fb34885bd68ec9733b2518f219cd4aa4f2d7"}]},"branch":"refs/heads/master"},"9f8a11a183ec7bdcb49a5336e405294cf4f681ed":{"kind":"REWORK","_number":36,"created":"2026-06-15 01:10:43.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"ref":"refs/changes/36/938536/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/36"}}},"commit":{"parents":[{"commit":"65cc5a702638bb4998b63134ef5fb765903d65e5","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/65cc5a702638bb4998b63134ef5fb765903d65e5"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-06-15 01:09:35.000000000","tz":540},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPass the resolved service name through fanout requests so the\nbroadcaster uses the same virtual-host namespace as direct RPC.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nCo-authored-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\nSigned-off-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f8a11a183ec7bdcb49a5336e405294cf4f681ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f8a11a183ec7bdcb49a5336e405294cf4f681ed"}]},"branch":"refs/heads/master"},"0ce80f571343f561301402246abbfd29cebf68a6":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2026-07-11 21:36:13.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/938536/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/37"}}},"commit":{"parents":[{"commit":"ee6e155b336cb5e48683c562e46c308ff9353d6d","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee6e155b336cb5e48683c562e46c308ff9353d6d"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-07-11 21:36:13.000000000","tz":0},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPass the resolved service name through fanout requests so the\nbroadcaster uses the same virtual-host namespace as direct RPC.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nCo-authored-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\nSigned-off-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ce80f571343f561301402246abbfd29cebf68a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ce80f571343f561301402246abbfd29cebf68a6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"65cc5a702638bb4998b63134ef5fb765903d65e5","ours":"9f8a11a183ec7bdcb49a5336e405294cf4f681ed","theirs":"ee6e155b336cb5e48683c562e46c308ff9353d6d","merge_strategy":"recursive","contains_conflicts":false}},"0196b876ed2032aeb1b090710f7452f505bcf639":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2026-07-17 05:37:02.000000000","uploader":{"_account_id":33451,"name":"Yushiro Furukawa","display_name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","username":"yushiro2"},"real_uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/36/938536/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/36/938536/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/36/938536/38"}}},"commit":{"parents":[{"commit":"61eb350594cc034ae4489e56ef27234e704c41cd","subject":"Add flask app for http-driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/61eb350594cc034ae4489e56ef27234e704c41cd"}]}],"author":{"name":"Xiang Wang","email":"xiang.wang@lycorp.co.jp","date":"2025-03-19 04:48:41.000000000","tz":540},"committer":{"name":"Yushiro Furukawa","email":"yushiro.furukawa@lycorp.co.jp","date":"2026-07-17 05:37:02.000000000","tz":0},"subject":"Add Broadcaster for RPC Fanout","message":"Add Broadcaster for RPC Fanout\n\nSince HTTP protocol doesn\u0027t have an broadcasting mechanism like other\nmessaging drivers, broadcaster is implemented as a separate service\nto fanout the RPC messages to multiple servers.\n\nPass the resolved service name through fanout requests so the\nbroadcaster uses the same virtual-host namespace as direct RPC.\n\nPartially-implements: blueprint oslo-http-driver (5/6)\nChange-Id: I9d67deff03993829c3a7aacbde5e01a897870b26\nCo-authored-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\nSigned-off-by: Xiang Wang \u003cxiang.wang@lycorp.co.jp\u003e\nSigned-off-by: Yushiro Furukawa \u003cyushiro.furukawa@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0196b876ed2032aeb1b090710f7452f505bcf639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0196b876ed2032aeb1b090710f7452f505bcf639"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ee6e155b336cb5e48683c562e46c308ff9353d6d","ours":"0ce80f571343f561301402246abbfd29cebf68a6","theirs":"61eb350594cc034ae4489e56ef27234e704c41cd","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
