)]}'
{"id":"openstack%2Foslo.messaging~942004","triplet_id":"openstack%2Foslo.messaging~master~Ibd3bc28641215109fa3563194e68bb0527f169d2","project":"openstack/oslo.messaging","branch":"master","topic":"metrics_native_queue","attention_set":{},"removed_from_attention_set":{"29525":{"account":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"last_update":"2025-02-28 14:36:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29525\u003e replied on the change","reason_account":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-28 14:36:54.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-02-28 14:36:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibd3bc28641215109fa3563194e68bb0527f169d2","subject":"Use stdlib Queue with native thread in metrics","status":"ABANDONED","created":"2025-02-17 19:51:33.000000000","updated":"2025-02-28 14:36:54.000000000","total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"e5e64f4b98e46323d129b2df59025764ca33f3c4","_number":942004,"virtual_id_number":942004,"owner":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-24 12:19:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-02-24 14:08:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-02-24 16:34:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 21:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 09:27:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-02-24 10:39:09.000000000","updated_by":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-02-24 16:34:37.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"11944a68ef9fd673f689c55f437ca952938d43f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-17 19:51:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f821d12e9eee4262d4b0f6eefdcf408c2abe084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 21:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1503913a2b184e47885f7f612e9968e1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05472151de4c44738c963eee7eb7ef9e : SUCCESS in 1h 33m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4e6f49ccdc748dea527c3ebddeb651d : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f9bc04ac47941959ea5fd572cb837e5 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7901bf742cb04c6291db14ca609a7feb : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb7d8de65ff042a7a9957a9a3d5e610f : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2913210c3d6c4ad8ad47cf45efd39735 : SUCCESS in 3m 00s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ccfc8de5ca4248d195f219850170f92e : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/01bcd1688adf416283128d6c17be1c7e : FAILURE in 5m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3e813efa7ae642fbb7237676d54e3d8b : SUCCESS in 1h 56m 49s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a92f19b846c246e6880df05eb7422e77 : SUCCESS in 1h 48m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0e3c9334c2ea4d86b924a278021fd4c6 : SUCCESS in 1h 03m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e79981db394a4a2da877987e231110d4 : SUCCESS in 1h 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fe564f672f0866c5a7111b49a8b42e770e39046","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-24 09:27:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d982dba8ed077d8ad690e59ce1fc1aba2d3e425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-24 10:31:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2e0992205acbbbbe7f73a47b0590f0c8cc4cd68e","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-24 10:39:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eec464dd20f01522e87a8ad08960f71842e8452f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 12:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3acac4653425478a9b734d7ecf298f8d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b53a01e3f304898bfc620d692f3b80c : SUCCESS in 1h 14m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20798873ddf14b518c6f0b549dbe8b8e : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997d1842d2214451b65ce81cfd04d3c2 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aae6b67671c34b3a8d9060ee70eb4c2c : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afca510599954cc084f7c1b2070241cc : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a070a8ebf5984142bd02e075cd951092 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34c6bb51657426dbdb8e648546ab1a5 : SUCCESS in 2m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f2b5783b476d412ebfa470739fd8e115 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/fc41b79eb8a9448cab2574205c8c42a4 : FAILURE in 6m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/978e69780be040fc8bc7237f575afc31 : SUCCESS in 1h 35m 51s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ae48f6931c1d4e1180f87cb0a450607c : SUCCESS in 1h 24m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/64add54779624a5f93cdbc33cfa25b08 : SUCCESS in 58m 58s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8827dd0866b4449aa4aa148f60a56337 : SUCCESS in 1h 01m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"034f01f733e33e58e692d0dc1d2f040439d8ddec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-24 14:08:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aff67ece5bb961e047377d5ece13469693a92ed6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-24 14:11:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbbfa5195172bae61d238067986c62fd6d51d510","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-24 15:02:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50e63e3e48ef1d9bbda1f89b19beace924f1ac90","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-24 16:34:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6d7b50f97b0d0ad235d3a9c1db3cde6497920643","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-02-24 16:35:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7aafc411c4f1d36760164808b4149b572e6c8627","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-24 18:01:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0f912b72cafdee0ec87339ca4b9b8307d2944a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-28 13:53:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac0b6e4cce2817d7efa8232130fc36aa458e46e3","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-28 14:36:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5e64f4b98e46323d129b2df59025764ca33f3c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-28 14:36:54.000000000","message":"Abandoned\n\nfixed in https://review.opendev.org/c/openstack/oslo.messaging/+/942435","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ba183327af05823f5f1eb65dba79b003f3947eb","revisions":{"2c6408381307976128bdf52f7344232edd91fbc3":{"kind":"REWORK","_number":1,"created":"2025-02-17 19:51:33.000000000","uploader":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"ref":"refs/changes/04/942004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/04/942004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/1"}}},"commit":{"parents":[{"commit":"e91bb459c0c6f3ffbfbd1650d172a2b31d82b410","subject":"Merge \"Fixed: use self.start_time in QManager shm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e91bb459c0c6f3ffbfbd1650d172a2b31d82b410"}]}],"author":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-02-17 19:49:29.000000000","tz":180},"committer":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-02-17 19:49:29.000000000","tz":180},"subject":"Use stdlib Queue with native thread in metrics","message":"Use stdlib Queue with native thread in metrics\n\nFixes delayed message processing caused by mixing a native thread\nwith an eventlet-patched queue. The send loop\u0027s blocking read timed\nout incorrectly, leaving metrics unsent until the next message.\nSwitching to queue.Queue ensures immediate processing and prevents\ndelays.\n\nCloses-Bug: #2098714\nChange-Id: Ibd3bc28641215109fa3563194e68bb0527f169d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2c6408381307976128bdf52f7344232edd91fbc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2c6408381307976128bdf52f7344232edd91fbc3"}]},"branch":"refs/heads/master"},"0ba183327af05823f5f1eb65dba79b003f3947eb":{"kind":"REWORK","_number":2,"created":"2025-02-24 10:31:59.000000000","uploader":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"ref":"refs/changes/04/942004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/04/942004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/04/942004/2"}}},"commit":{"parents":[{"commit":"e91bb459c0c6f3ffbfbd1650d172a2b31d82b410","subject":"Merge \"Fixed: use self.start_time in QManager shm\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e91bb459c0c6f3ffbfbd1650d172a2b31d82b410"}]}],"author":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-02-17 19:49:29.000000000","tz":180},"committer":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-02-24 10:31:20.000000000","tz":180},"subject":"Use stdlib Queue with native thread in metrics","message":"Use stdlib Queue with native thread in metrics\n\nFixes delayed metrics processing caused by mixing a native thread\nwith an eventlet-patched queue. The send loop\u0027s blocking read timed\nout incorrectly, leaving metrics unsent until the next message.\nSwitching to queue.Queue ensures immediate processing and prevents\ndelays.\n\nCloses-Bug: #2098714\nChange-Id: Ibd3bc28641215109fa3563194e68bb0527f169d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ba183327af05823f5f1eb65dba79b003f3947eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0ba183327af05823f5f1eb65dba79b003f3947eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
