)]}'
{"id":"openstack%2Foslo.messaging~942435","triplet_id":"openstack%2Foslo.messaging~master~I3c5ce7cc399c76afcee57d92d6a3a0f2b680efce","project":"openstack/oslo.messaging","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-05 13:38:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-02-25 09:33:33.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I3c5ce7cc399c76afcee57d92d6a3a0f2b680efce","subject":"Do not un-patch threading","status":"MERGED","created":"2025-02-21 12:17:24.000000000","updated":"2025-04-05 15:28:06.000000000","submitted":"2025-04-05 15:26:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"942435-eventlet-removal","meta_rev_id":"729e95ab02e8a8486ff4a275babfbb860a0df6cd","_number":942435,"virtual_id_number":942435,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-05 15:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2025-02-28 14:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-03-24 16:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-03-03 10:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-01 15:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2025-03-03 11:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},{"value":1,"date":"2025-04-05 13:38:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 13:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 14:15:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-02-28 14:59:50.000000000","updated_by":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"reviewer":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"state":"REVIEWER"},{"updated":"2025-03-03 10:46:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-03-03 11:17:07.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-03-24 16:17:24.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-04-01 15:56:32.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"e5e381fd35a056004b490872cf450eb42a8c39a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-21 12:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38cf65df87f3dedd0368fda9e81535cea342002f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-21 12:17:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"831891e656e390b883915d41785091146afbdf7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 13:57:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f276d8b32ac42989237b15151339790\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6566fa42c1af4d60b4eef541b1e8cf4a : SUCCESS in 1h 20m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a031b3f02ba64d69bde12465c5baf5c7 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa04805d1d40438395b80f5b2607e9f0 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd23e8820e0b4eec99dbc74d190fed42 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/915ec66c3fa94d2188aecbe65544a66d : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310e126f1ae240959d9f122e2b46e71b : SUCCESS in 2m 47s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/050e21b251be4aa68e0f5b3dfc36b7fe : SUCCESS in 4m 37s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/cf958b4a90f1429ebf7d91923a97444a : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/921b26fe8f714cbb859b74fabbe04180 : SUCCESS in 1h 38m 58s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/954ce72237b04bd1a12ad264bd2c6a3f : SUCCESS in 1h 31m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/53f5bc41e3164dac97358629b28fa6e1 : SUCCESS in 34m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd27300cb49f4cddbe6ac66abae7449a : SUCCESS in 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c567425098fb80c56674d4d467a9f4d1fdfb1fa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-24 14:15:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a38e5615b610888a72bbf441f8252b9d79ed5ede","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-24 14:15:31.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"f14b5cf53ffa4d51c45ca355def1cbad341c13fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-24 14:59:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"021ddeffbaf97b74a57d17b0e7513a78f2030a2d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-02-25 09:33:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d37cdb9e8a630143c12b971e3ba4a29ee0983b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-28 13:51:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3902bc57fe1ee2e20bfdc89ac580682b06eab414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-28 13:52:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93df39a4a5739bfa39963c46da0360235b57c235","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-02-28 14:59:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fa718781bd992262fbb0e1e6ec99b5d7ef36312c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 15:47:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0be229e78741ffafd6a57729de59ad\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4af4b26c454471794923d8562b345d6 : SUCCESS in 1h 41m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66b6a0de1ec946efa50ab04776e981f0 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a8a2f290c9473bb421052bb60a70e2 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e077b72faacb455f99b5c920101bea3e : SUCCESS in 7m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66210b5ab4b44d20a8b56f5466be581b : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74033a0fc7e04fbcab3e068d6aedeabb : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6df676a39fc4a40a0e56c06e63fd653 : SUCCESS in 4m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f5401fc91df341dcb8365c2a7f8e29de : SUCCESS in 5m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9c7f7d3de5d149eeb18f60833e9ee20d : FAILURE in 7m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/efbd271182e74773b290e046cb70a9ea : SUCCESS in 1h 49m 50s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e97c552626ec41aab64cb2d4403bd3ad : SUCCESS in 1h 52m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d60eb037002841008ba89ac310bbbaae : SUCCESS in 56m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6892795b5e8449b968e9760b3e6dd66 : SUCCESS in 1h 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f47abca0754093b96f8eaa671e29cef05f65d479","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-01 03:05:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ea9c549bbc7ec89bf37a9ab0b47e6bd27efd4ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 05:11:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54b13ab8981b48f0910d0c0d0f8fe01c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a2d4059c7764b29aac4bbb1a108bb16 : SUCCESS in 1h 55m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a90940181e4980bde65279e42eb918 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11b8270d78f445fb17f6f9e64adef6e : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b94fddf39d0341ed8ea5d1b1316fcbc1 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2591c78f475848f28d2c8ac155d9c092 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3957e3642f447391e9c00cf65601b2 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10005c71589745169af96118b4f7611e : SUCCESS in 3m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1a9f12b4d62d439b87eb5041d0de04e1 : SUCCESS in 5m 27s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/c94a9993534c4af0b0b5d6d1161457a8 : FAILURE in 5m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ffee9e5f4e554ace8735c14b91029246 : SUCCESS in 1h 36m 46s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/11c35402a31c4f55aab867eb0abb55be : SUCCESS in 1h 58m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0eb1b382383f4c46903d2818d418dd71 : SUCCESS in 55m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/52a597734e394066865fe196941ca296 : SUCCESS in 1h 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bed51a67cee4d8a1fabffc2e8a5ebdeaa84e9199","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-03-03 10:46:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9c41e74ffeea082cbf10f4172ede73a32b278af9","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-03-03 11:17:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ec9a419e90b7217cf1f259dea1578a32b6dddde0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-24 16:17:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7dcb518dfc69ee5aaeea0dfe617c62d0319fb5e4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-04-01 15:56:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"332cd1efabece337ef756123479e86f88d07739c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-05 13:38:16.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4abae87bee2b540fdd4454d3826dbc08fa88fede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 13:38:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"624db4b5a903f8ede41d44e5eb3f12c6aa8cdf90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 15:26:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ea1a1409de45fabfe297a5f4dfa2e0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe7b9ca7d39245e6851389146c9b5a82 : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71e9a6a533b470b873f654a78d085d3 : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e7345357d0d425ea1c5a7d12926672d : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67c874cf3f444df1a98c2ac1a887aa60 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2c29c2bb1d422eadad2d264466b025 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17e17dcb8f60434caa518586f71e9652 : SUCCESS in 2m 10s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bab79ad0259c416bbcbfa948c79e7202 : SUCCESS in 5m 53s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c701ec50851d48bfa2f29fc8dcd24505 : SUCCESS in 1h 48m 04s","accounts_in_message":[],"_revision_number":3},{"id":"6bd2dd22c1f61aa62212b4ede05b42ca1b71e36b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 15:26:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"729e95ab02e8a8486ff4a275babfbb860a0df6cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-05 15:28:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed726bd1511740a9999c56c9912e1149\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5f56ad076a480aab56993cdbf12fe5 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc7bac444596485380ba5fd3bd9e24b1 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3ec37a36082ba4ac97993961d4073bf43177d38","revisions":{"05ea067e4aa274c7d7ed6bcdc22f2d27940fdb9f":{"kind":"REWORK","_number":1,"created":"2025-02-21 12:17:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/942435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/35/942435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/1"}}},"commit":{"parents":[{"commit":"89f4ce95f28eef95ad3950d3520870c0a8e37c91","subject":"Skip installation to speed up pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89f4ce95f28eef95ad3950d3520870c0a8e37c91"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-21 12:14:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-21 12:17:16.000000000","tz":540},"subject":"Do not un-patch threading","message":"Do not un-patch threading\n\nCurrent un-patching causes conflict with a few other moduels such as\nqueue patched. Because there is no clear reason why we need this\nunpatch, let\u0027s remove it to use unpatched modules or patched modules\nconsistently.\n\nChange-Id: I3c5ce7cc399c76afcee57d92d6a3a0f2b680efce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/05ea067e4aa274c7d7ed6bcdc22f2d27940fdb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/05ea067e4aa274c7d7ed6bcdc22f2d27940fdb9f"}]},"branch":"refs/heads/master"},"9c02d20cbbe611f81e35e4c4ce92ecc0a8a0d517":{"kind":"REWORK","_number":2,"created":"2025-02-28 13:51:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/942435/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/35/942435/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/2"}}},"commit":{"parents":[{"commit":"89f4ce95f28eef95ad3950d3520870c0a8e37c91","subject":"Skip installation to speed up pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89f4ce95f28eef95ad3950d3520870c0a8e37c91"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-21 12:14:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-28 13:50:42.000000000","tz":540},"subject":"Do not un-patch threading","message":"Do not un-patch threading\n\nCurrent un-patching causes conflict with a few other moduels such as\nqueue patched, which is causing undesired delays. Let\u0027s remove it to\nuse unpatched modules or patched modules consistently.\n\nCo-Authored-By: Artem Vasilyev \u003cartem.v.vasilyev@gmail.com\u003e\nCloses-Bug: #2098714\nChange-Id: I3c5ce7cc399c76afcee57d92d6a3a0f2b680efce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c02d20cbbe611f81e35e4c4ce92ecc0a8a0d517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9c02d20cbbe611f81e35e4c4ce92ecc0a8a0d517"}]},"branch":"refs/heads/master"},"e3ec37a36082ba4ac97993961d4073bf43177d38":{"kind":"REWORK","_number":3,"created":"2025-02-28 13:52:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/942435/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/35/942435/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/35/942435/3"}}},"commit":{"parents":[{"commit":"89f4ce95f28eef95ad3950d3520870c0a8e37c91","subject":"Skip installation to speed up pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/89f4ce95f28eef95ad3950d3520870c0a8e37c91"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-21 12:14:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-28 13:52:00.000000000","tz":540},"subject":"Do not un-patch threading","message":"Do not un-patch threading\n\nCurrent un-patching causes conflict with a few other moduels such as\nqueue patched, which is causing undesired delays. Let\u0027s remove it to\nuse unpatched modules or patched modules consistently.\n\nCo-Authored-By: Artem Vasilyev \u003cartem.v.vasilyev@gmail.com\u003e\nCloses-Bug: #2098714\nChange-Id: I3c5ce7cc399c76afcee57d92d6a3a0f2b680efce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3ec37a36082ba4ac97993961d4073bf43177d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e3ec37a36082ba4ac97993961d4073bf43177d38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
