)]}'
{"id":"openstack%2Foslo.messaging~946720","triplet_id":"openstack%2Foslo.messaging~master~I42f02490372322da0017c5e205f28521ff69a337","project":"openstack/oslo.messaging","branch":"master","topic":"bug/2106547","attention_set":{},"removed_from_attention_set":{"29525":{"account":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"last_update":"2025-05-05 14:31:45.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-17 09:32:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I42f02490372322da0017c5e205f28521ff69a337","subject":"Fix RPC metric labels being reused incorrectly","status":"MERGED","created":"2025-04-08 21:59:56.000000000","updated":"2025-05-05 14:32:37.000000000","submitted":"2025-05-05 14:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946720-bug/2106547","meta_rev_id":"fff21d7df3554cf67304c3c34b0ab92db5977223","_number":946720,"virtual_id_number":946720,"owner":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-05 14:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-04-17 09:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-05 13:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-05 13:06:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 23:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 12:23:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-04-17 09:32:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-05-05 13:06:28.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"4689b12627001d1753228e418deac3e51405f1b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-04-08 21:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f04658ba60b497090c7c91581af6c186c3608507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 23:50:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31691c124a9b454a9a335787ed040b70\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15f7a761df004ad9b1e5d1734600951b : SUCCESS in 1h 27m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8793520550a04a3c9da47f9ee7adaaca : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ad15c991a24d3d8f4c67534b3c7f6e : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5af1b3e3603841808d1b008b067ed792 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfc5edb547374548acbfe5f04297b5fd : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c82216c5107d4fd58525c07e3771beec : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660f7b705b654591b43ca01a25dadb94 : SUCCESS in 3m 56s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/474f941187584f1d9eb2f32291cbda51 : SUCCESS in 4m 22s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/baf66eb037074b08b7b75e3eb2d72cff : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aadfb6aed8984a2190d089fb345f5cb7 : SUCCESS in 1h 45m 34s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/bc1cae97004b4b9b827633875a4deb73 : SUCCESS in 1h 22m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/85ebc15cf9524be9a813254efb6a3744 : SUCCESS in 55m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4eee6421eda34a49914f88a2bd7219af : SUCCESS in 53m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51cfb14128365552b5c7228b4b1d31bf11009112","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 12:23:37.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d09d832e27d046079fb04387f320076a47b2902d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-04-16 14:59:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9060a64c2df459b23306e937f00ccfed3b50daaf","author":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"date":"2025-04-16 15:01:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71ce35b1042142b6b94a23da8ce3faf47a00c72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 16:45:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29109d980aa4d34a6edf8daa7fa7412\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bd75d771f4844e984c910ee361b641f : SUCCESS in 1h 42m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2d519cb48e4e1782ae5d581bff3c92 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882d72a5d53642f696ef1c2a33e30306 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/304f9fb63a134cd1a131cea5f6ab13b8 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b58f7241eac44eaeacc08c813e73aa9d : SUCCESS in 7m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b16d1a92196e486c8841a6f71b2d3619 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9923c18197b64cde8f6b0b1cdaeb9872 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1f8a33b90d8247e19235d8d66e697a13 : SUCCESS in 4m 22s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0be81890baaf42beac9bf3a9ee31b42b : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/531db72290864c4d815cdebfcd7fef47 : SUCCESS in 1h 33m 22s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/55aa8a26cb9b457f957a5371b33e27cd : SUCCESS in 1h 37m 20s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7bacfb337d9f4a71b15c67b20e9c7f96 : SUCCESS in 55m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7c9e91f50a2495c98052bce7fbc9b00 : SUCCESS in 1h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b8e21e76fd609aba420b6b0b0b8c27a1a79d755","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 09:32:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c65ab2e620d3970b68a9fd8655ccdb5cf7ebd89f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-05-05 13:06:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf5b9ead6c87ad267bdb1d2983cc2b03c2562ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:06:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"577c20a83907c7c955dc9e31b23dc5fc95466a8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:31:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6f77147550c4e2eaad5eef3729de6bd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79ed6d8eae3149639c5af8a47db6fe50 : SUCCESS in 1h 24m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc1283f27b14db287416a09890b8001 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/660680e25d07424c96b9631469ee44d7 : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f13ac94c2a8c45f78a21e076637fc518 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/586c6492b443459a8e06114e8821d6b1 : SUCCESS in 3m 42s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1243d4b88e564a5f935fa09356267995 : SUCCESS in 4m 18s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1280f6a51a50468f9a629f8db476ef87 : SUCCESS in 1h 23m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e3fb97cfbce13c3f04fba23085ddf5cd7a91bf4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:31:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fff21d7df3554cf67304c3c34b0ab92db5977223","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:32:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0e02661b00437cabc26ab9f54b5f7b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddc94e2fbbb419d80231f7fd5f03f0b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fedf06d75ace0e9da4c328d18aa7da8909352578","revisions":{"3b1e3237f58be904f5a8629f442cd3b418dc2887":{"kind":"REWORK","_number":1,"created":"2025-04-08 21:59:56.000000000","uploader":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"ref":"refs/changes/20/946720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/20/946720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/1"}}},"commit":{"parents":[{"commit":"c058e79bafb1976a2e40a1a6dbd05036acf1a33c","subject":"Merge \"Do not un-patch threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c058e79bafb1976a2e40a1a6dbd05036acf1a33c"}]}],"author":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-04-08 21:57:11.000000000","tz":180},"committer":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-04-08 21:57:11.000000000","tz":180},"subject":"Fix RPC metric labels being reused incorrectly","message":"Fix RPC metric labels being reused incorrectly\n\nMetricsCollectorClient cached the first metric\u0027s labels and reused\nthem for all later RPC calls.\nRemoved label caching and moved context manager logic to a new\ncontext manager that takes explicit labels, ensuring correct\nper-metric labeling.\n\nCloses-Bug: #2106547\nChange-Id: I42f02490372322da0017c5e205f28521ff69a337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3b1e3237f58be904f5a8629f442cd3b418dc2887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3b1e3237f58be904f5a8629f442cd3b418dc2887"}]},"branch":"refs/heads/master"},"fedf06d75ace0e9da4c328d18aa7da8909352578":{"kind":"REWORK","_number":2,"created":"2025-04-16 14:59:50.000000000","uploader":{"_account_id":29525,"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","username":"artem.v.vasilyev"},"ref":"refs/changes/20/946720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/20/946720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/20/946720/2"}}},"commit":{"parents":[{"commit":"c058e79bafb1976a2e40a1a6dbd05036acf1a33c","subject":"Merge \"Do not un-patch threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c058e79bafb1976a2e40a1a6dbd05036acf1a33c"}]}],"author":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-04-08 21:57:11.000000000","tz":180},"committer":{"name":"Artem Vasilyev","email":"artem.v.vasilyev@gmail.com","date":"2025-04-16 14:59:33.000000000","tz":180},"subject":"Fix RPC metric labels being reused incorrectly","message":"Fix RPC metric labels being reused incorrectly\n\nMetricsCollectorClient cached the first metric\u0027s labels and reused\nthem for all later RPC calls.\nRemoved label caching and moved context manager logic to a new\ncontext manager that takes explicit labels, ensuring correct\nper-metric labeling.\n\nCloses-Bug: #2106547\nChange-Id: I42f02490372322da0017c5e205f28521ff69a337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fedf06d75ace0e9da4c328d18aa7da8909352578"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fedf06d75ace0e9da4c328d18aa7da8909352578"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
