)]}'
{"id":"openstack%2Foslo.messaging~947094","triplet_id":"openstack%2Foslo.messaging~master~I89eb7b34b0acf64dcff52af0cd45845e4dad4639","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-05 14:57:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89eb7b34b0acf64dcff52af0cd45845e4dad4639","subject":"Remove quorum options with typo","status":"MERGED","created":"2025-04-14 12:13:13.000000000","updated":"2025-05-05 14:59:17.000000000","submitted":"2025-05-05 14:57:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947094","meta_rev_id":"dea4b2493b9d9f23febcc351a92c1e2f9dd569e4","_number":947094,"virtual_id_number":947094,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-05 14:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-04-14 14:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-04-15 06:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-05 13:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2025-04-15 08:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-05 13:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-2,"date":"2025-04-15 10:38:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2025-04-15 08:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 13:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-14 14:53:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-04-15 06:09:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-04-15 08:58:24.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-05-05 13:06:57.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"4dd800960332f0cb199383030cd96c0428dab388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-14 12:13:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"121828da5c52a1b43490da8101f3373dfa49da6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 13:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4bae81f3aff45e7bb9f0d08ea01c3de\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51c155b5be874f2ab3f7551843819ecd : SUCCESS in 1h 38m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de15291b9a2142f4af8b3cd3fceb7829 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/492980659d654468929092f7ade6aa53 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0726fcbb600b4336a9c5e26eae86a9de : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/128f9376aa364ad7876ee099a0c12152 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/870a3a998fb242b4a6f5a5bafd2c6c4f : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb26273013f5407d931e428c2694660d : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/696701289c414345b0d2554917c9ab7f : SUCCESS in 2m 17s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/705eb648826347578ce16b89e16a411c : SUCCESS in 4m 15s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/02e7a44e9e9e454a9302b6402f658d94 : FAILURE in 8m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/504fc6ce538f4d35927edd59594a4272 : SUCCESS in 1h 22m 33s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/44642c5b758d4e309fe4e7d216d1f771 : SUCCESS in 1h 36m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6c18c871fdb34a329c96300284210ccc : SUCCESS in 28m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e0fd6286d0140dc8cabe5feeed5ae1a : SUCCESS in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5d21fabfd3cb3e31b1da0b3cc15a11018bd81dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-14 14:53:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04aaa12bd4212f437e12c61ac54a0cc43852ee16","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-15 06:09:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8443e226a27342031300c544ef63d5856890bdce","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-15 08:58:24.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"76728fb7b34059b582c73b6fc5ee8a1c8a0822e2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-15 10:38:41.000000000","message":"Patch Set 1: Backport-Candidate-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcb87787da3281615747debbefbae74455499e54","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-05-05 13:06:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d0327f5b6f59ce87868d6d15ff268e255ae07b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:07:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01f185ac08e43f4923e97abcd54b087653794a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:57:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa9aa1771cd439180f027f2876cf54f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5661b7134f5f4cbfad262bc3ff60ace5 : SUCCESS in 1h 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e166005ba9a4aa5a55d026e0f537cf1 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caf5555b899640cb94396c0523cd5da1 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fba468b5d664703b452081a44c6e954 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f789dd3282c848aeb933796fe786a1cb : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5944056c7d946d1b124752677f3398c : SUCCESS in 2m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/177f8ce65a7b43cbb63590fe924a424c : SUCCESS in 4m 05s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/758d294796a84a70a18aab54d4869104 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5b2b9123116cbbdbf0e26733df84e550d433f20b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:57:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dea4b2493b9d9f23febcc351a92c1e2f9dd569e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:59:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c8e104edaf74f859003d98b0c2d708e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80779042134e473294d5224d278f36f1 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb366f0c7d364dbea32c543a58a19d76 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf39b36b31975f5d080d4474519583411c35849","revisions":{"4bf39b36b31975f5d080d4474519583411c35849":{"kind":"REWORK","_number":1,"created":"2025-04-14 12:13:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/947094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/94/947094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/94/947094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/94/947094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/94/947094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/94/947094/1"}}},"commit":{"parents":[{"commit":"3a0b6cb9813614077c6712d10df57190a98583b6","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3a0b6cb9813614077c6712d10df57190a98583b6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 12:11:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-14 12:12:28.000000000","tz":540},"subject":"Remove quorum options with typo","message":"Remove quorum options with typo\n\nThese were deprecated during 2023.2 cycle and we even backported\nthe renaming to 2023.1 so that users can use the correct name. So\nthese are now ready for removal.\n\n[1] 115cfb5b7cff2a9803d64d02b98250b7b0fc09f0\n\nChange-Id: I89eb7b34b0acf64dcff52af0cd45845e4dad4639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4bf39b36b31975f5d080d4474519583411c35849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4bf39b36b31975f5d080d4474519583411c35849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
