)]}'
{"id":"openstack%2Foslo.messaging~959784","triplet_id":"openstack%2Foslo.messaging~stable%2F2025.2~Ia3689c9b8ba31e231542a87b33a49bcf20e43095","project":"openstack/oslo.messaging","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"Ia3689c9b8ba31e231542a87b33a49bcf20e43095","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","status":"MERGED","created":"2025-09-05 12:23:47.000000000","updated":"2025-09-05 13:51:19.000000000","submitted":"2025-09-05 13:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959784-create-2025.2","meta_rev_id":"adaeef38ab3de94a6981ae206febc1b86d52cfb6","_number":959784,"virtual_id_number":959784,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 13:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-05 13:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-05 13:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 12:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-05 13:36:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a6e4613c96056b61df7519b71f1d438f1b1bc243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-05 12:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44c612a58ded3f83c693d64e9016d13557a94b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:51:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164c4175ffaf4c678e474b7b5d7f338a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f64dbc79fc04e45ab68fa5b6004664e : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b646425e86fe4a838f332973b9d60e0d : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baf61a0ea9a64e93a67a5956f335a24d : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/127843cbc0b348af90a2d13f559934c6 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9280b217756348e1aa707c9514e822d9 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d19b01c87434d34be488155b62f99d3 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51343a74e9bb42a4b7245b03f482f6c5 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/5c851c137da94e23a36841efa9c96cf1 : SUCCESS in 5m 56s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3d06386abc5e4d489ba7ba39755beae1 : FAILURE in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e396434eaa04a02e453bb288eb426463167f9b0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-05 13:36:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17edd4263822b94660504260e87879298321b1ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:37:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05ff2189d7ef179e57f046e4e277f4d6473b9cdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532a1c79010c4109ace8b812de37f20a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c52f36bb264ea78eaf6bda6e1f7eaa : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b000fd29efd461cbd910152a4bb2efc : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4802c148954a4b51b5a2eb7d420d0881 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3be61a86ed14373aacc08c7f738253a : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82380ef03ced4a3d98d9f83be6591d2c : SUCCESS in 2m 37s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/d8f95b3a263247d5b8e154a9b9a5f8bc : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c2d05b50d4ed1952b203a9c7dcb0d238e14280cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adaeef38ab3de94a6981ae206febc1b86d52cfb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:51:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e11bbc73b7a41948f27602d5474e91e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592263eb85af4e8ca86f70b2e11022b0 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a45107f1f3594b9a9f138d06cd3182c4 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d6a623a480ccec7989bb68128f3a897d3f684b0","revisions":{"6d6a623a480ccec7989bb68128f3a897d3f684b0":{"kind":"REWORK","_number":1,"created":"2025-09-05 12:23:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/84/959784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/959784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/959784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/959784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/959784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/959784/1"}}},"commit":{"parents":[{"commit":"adccb14c64c31f8f148492fe8ea927f972d6c40c","subject":"Update .gitreview for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/adccb14c64c31f8f148492fe8ea927f972d6c40c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-05 12:23:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-05 12:23:45.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ia3689c9b8ba31e231542a87b33a49bcf20e43095\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6d6a623a480ccec7989bb68128f3a897d3f684b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6d6a623a480ccec7989bb68128f3a897d3f684b0"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
