)]}'
{"id":"openstack%2Foslo.messaging~960603","triplet_id":"openstack%2Foslo.messaging~master~Id005b8d0252734fbfd6c90d7ab3fd9531533d5e2","project":"openstack/oslo.messaging","branch":"master","topic":"pyproject","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-13 21:20:40.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-09-13 21:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id005b8d0252734fbfd6c90d7ab3fd9531533d5e2","subject":"Migrate setup configuration to pyproject.toml","status":"MERGED","created":"2025-09-11 14:52:24.000000000","updated":"2025-09-13 21:21:23.000000000","submitted":"2025-09-13 21:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960603-pyproject","meta_rev_id":"e9a0aa1b804972312c0a6b233d9ad891c3d73a3d","_number":960603,"virtual_id_number":960603,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-13 21:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-09-13 19:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-09-13 19:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 17:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 14:15:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-09-13 19:29:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"170a80bb91a39e313c8c71997c4d6440dee0a802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-11 14:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b768d49a43f57f4d6c59b2197093b2f74dcf19e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-11 14:52:46.000000000","message":"Topic typing removed","accounts_in_message":[],"_revision_number":1},{"id":"159573bce4db6d5ecbea18513827aaae5f83df22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-11 14:52:48.000000000","message":"Topic set to pyproject","accounts_in_message":[],"_revision_number":1},{"id":"e863b0b4466da7f4a474ac80ab62f1838b2d6b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 17:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d9e3940d1c34e1f873bef0e353b337e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/46790577f41f4c73b2e4a992796f347c : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/710a4a03aff9488c9cc1f83a782b5d71 : SUCCESS in 2h 01m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f7e31ee0f8f4a5d92b192747e54453e : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d9ab72b8914b71a9aec92652463c5f : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c5ffba4a60c469483514d81b1e896d6 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7af35d95d8184ffc890cb4722f6f50e9 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fde0ef0005e94372a74949799891564c : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f260e6b54e7a4edfa541a56a0cbc6296 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ef6769a2096415f96edf9fc6afecf96 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/e24c39c7c10d4b179d096b415ee810de : SUCCESS in 6m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0f67a3a1ae1c4c26967ea12f878dbf12 : FAILURE in 20m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e1fac13d9d664283a47cf283de8305f9 : SUCCESS in 1h 42m 24s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e248af5fe43b4fb591a5779778916695 : SUCCESS in 2h 01m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fdd81004b8814b99803d26f8e5a6f495 : SUCCESS in 29m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/45b090abf0ad4de39fe61d4c6e38945e : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20cef2b38c91a763d69feb5ec1b1befbfda96d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-11 17:20:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"af931c8009dfaeedce76d44818f72e25506c41d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:20:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ebb9cc04ae646238c9b04defffe30d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6a809234f564348b09724596eaadc93 : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcf86dd51b664a0e96e24d91e52e1e7c : SUCCESS in 1h 51m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41f54f57b8a48298bb3daa3b2374549 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d820b07da148c0881247994fccbe43 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a1b1b65e6a74f33af6765827e95b603 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56746a6174804f2caf13d3bf01dd5d02 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bdc4919fba1423c97ebd95db1425a29 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3760dc01f7c6425e8ea1e051bc127b82 : SUCCESS in 3m 30s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ff66692fd3a9462db4133c61cab0bf07 : SUCCESS in 6m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/32376cae9dba4118888c305d97f04679 : FAILURE in 14m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/22abfdc7a3a54d5a8895e79149ec793c : SUCCESS in 1h 55m 00s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a7f9771dc763425fae24dfcd58f5cd5d : SUCCESS in 1h 58m 55s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d08c1428817d4b59960b1eff7a8a7b5f : SUCCESS in 52m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cab4615d77444dda518c0767c9781fd : SUCCESS in 46m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd8b0d5c452da4f883c92a278af108cc12e704de","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-09-12 14:15:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"58b9b984a4ce3b03ac075e8f3a9f41cf1ec280ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 14:16:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2b26324527e03f548e22bc352c7d9686a8cbf27","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 15:36:19.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10e750292c7d71a73710a040aeba7878b4b36c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 15:36:21.000000000","message":"Patch Set 2: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":2},{"id":"85d53f61574ea44adb83b9cb9cc53901fdc1fbef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 15:51:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b804c3b49c4e41dc0839ead1989dadb4907f371","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-12 15:52:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fb45eb5f7fd22d85b8c3ecf795f9ded683fdd5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 17:58:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5fa63cd9971485ba1c41dc5579dfc17\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6a2a2f3cf3b421d9ee2207d49ef1c56 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a459034e6c9c4d2b910dc0d1e0c9d156 : SUCCESS in 1h 41m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5005c4aa7e540cdb7745de0f38e6421 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a9f75cdd21d4402a4167645f28881b8 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94819918799b4095991c4c198866a6bb : SUCCESS in 6m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97c6a945142645fd954570bdc6e0b710 : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77a7b6f44f834ad38539c3b780c27e2e : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c75cbf38bf4a638754c4773e831a3d : SUCCESS in 3m 58s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/31036dd969dc4fcd822ee96e6d09869e : SUCCESS in 5m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/3c144718ceb843f6918d7d5a69dfa481 : FAILURE in 11m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/61e47c041af7411c8a49bf3d6a2970a3 : SUCCESS in 1h 59m 47s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/87dc5f19dd8449648daa5180cb918130 : SUCCESS in 1h 59m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ba7faa9aaf954349b5247db52d70be5a : SUCCESS in 1h 04m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0167b832ee54c369b655f34d20869c0 : SUCCESS in 1h 09m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7904767405d58c793953ef7af6f1d771e3d0dbe7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 19:29:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"016daf641f46f6b2f1ae6db48a797417cdaadab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 19:29:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0d127f714171a5b753db2d8dbfd449af6951692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 21:20:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c501c541804737b8dcb6d92d421f60\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1b7020748ac4dca9d82b69dc6618cf6 : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4c102ff387849c48c156ec6de368de1 : SUCCESS in 1h 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123b785e7b384235afb14059c77009b7 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/527fa66b68dc449a91dbc045f3bc926e : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f8af515ae8b43bb84bd90ac06325914 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f8dcd500a74a0ba1f889a31447fe1a : SUCCESS in 3m 14s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9960ab10377f44e191201d9a4423abed : SUCCESS in 4m 19s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4b0e86591a6c4b48b77f1ada697e0ffd : SUCCESS in 1h 46m 33s","accounts_in_message":[],"_revision_number":3},{"id":"648cb893c90013466d5aa378bb4bab8b4630c584","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 21:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e9a0aa1b804972312c0a6b233d9ad891c3d73a3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 21:21:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/356e91e3111f43a6a528afeb97614442\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae74a33efb5490599b8fcef11b1c747 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4324bb89b958044209f6afa145dc35b5fd2588ab","revisions":{"31421e4586a8eaea26f7450b7f66f81603c8ac46":{"kind":"REWORK","_number":1,"created":"2025-09-11 14:52:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/960603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/03/960603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/1"}}},"commit":{"parents":[{"commit":"56921bcf7f08ad29c19ae8767c694099f0d83450","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/56921bcf7f08ad29c19ae8767c694099f0d83450"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 14:51:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 14:51:57.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\n...and drop Python 3.9 support in the process.\n\nChange-Id: Id005b8d0252734fbfd6c90d7ab3fd9531533d5e2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31421e4586a8eaea26f7450b7f66f81603c8ac46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31421e4586a8eaea26f7450b7f66f81603c8ac46"}]},"branch":"refs/heads/master"},"22acd94afe1bcf2766666cf31c5fea0d043a40ad":{"kind":"REWORK","_number":2,"created":"2025-09-11 17:20:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/960603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/03/960603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/2"}}},"commit":{"parents":[{"commit":"c4589c6665fc44ba26964f7b496cde38c112340b","subject":"Merge \"Drop Python 3.9 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c4589c6665fc44ba26964f7b496cde38c112340b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 17:19:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 17:20:22.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id005b8d0252734fbfd6c90d7ab3fd9531533d5e2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/22acd94afe1bcf2766666cf31c5fea0d043a40ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/22acd94afe1bcf2766666cf31c5fea0d043a40ad"}]},"branch":"refs/heads/master"},"4324bb89b958044209f6afa145dc35b5fd2588ab":{"kind":"REWORK","_number":3,"created":"2025-09-12 15:51:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/960603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/03/960603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/03/960603/3"}}},"commit":{"parents":[{"commit":"c4589c6665fc44ba26964f7b496cde38c112340b","subject":"Merge \"Drop Python 3.9 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c4589c6665fc44ba26964f7b496cde38c112340b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-11 17:19:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-12 15:51:35.000000000","tz":60},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Id005b8d0252734fbfd6c90d7ab3fd9531533d5e2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4324bb89b958044209f6afa145dc35b5fd2588ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4324bb89b958044209f6afa145dc35b5fd2588ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
