)]}'
{"id":"openstack%2Foslo.messaging~971591","triplet_id":"openstack%2Foslo.messaging~master~I059e8861256ea6a41e29bed89d12b6d5b73cf4e2","project":"openstack/oslo.messaging","branch":"master","topic":"remove-FIPS_mode_set","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-01-08 15:21:18.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-20 16:47:26.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-20 18:21:17.000000000","reason":"Change was submitted"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2026-01-20 18:21:17.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-01-20 15:14:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2026-01-20 18:21:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I059e8861256ea6a41e29bed89d12b6d5b73cf4e2","subject":"Deprecate enforce_fips_mode","status":"MERGED","created":"2025-12-20 18:07:15.000000000","updated":"2026-01-20 18:22:27.000000000","submitted":"2026-01-20 18:21:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971591-remove-FIPS_mode_set","meta_rev_id":"1307749eeb7d6b3b374bbce0149e60a080ed0dd4","_number":971591,"virtual_id_number":971591,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 18:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-20 16:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-01-20 15:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-20 16:47:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-20 20:02:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-21 15:14:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2026-01-05 10:45:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-01-05 10:45:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2026-01-06 17:59:37.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-01-08 09:53:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"082bac493dd2ef5d02003b97223ea40d3da3291c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-20 18:07:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2b51e60c1af64ffa1d8361e7b24a7ce3e4ceb75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-20 18:07:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8096b2ffa0efdf8b70e97b3d4960199b1e281770","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-20 18:13:09.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"aa49477ffea5c609050a2d6b0184a2b66dc88ee0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-20 18:22:52.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"bd675ef3d7aa8c4c539380a75f3a6d46fc37d705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-20 20:02:23.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf4da2d5995d4e7389d781927a81aa8f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bae1336dec7241808b44703ecf64c552 : SUCCESS in 1h 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27d5d2df57f04e57b5e18f563d7701c0 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986f4a061faf4c53b094d13794a6b55c : FAILURE in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/021d157aec28463bb22e43a3294fa27f : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f07cd9bd69ae4975a572b52c91e25946 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8ea75159cf4d62b3766c89f6c83109 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8992dc3b2314e2bbeb8e5a77a49c9e0 : SUCCESS in 3m 14s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c003fdef7fc3421a8811ca844ace4096 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/4ca2730b09a245b08dbe3eecd889f24d : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9e9042f1a98d4d4fa5f27a9744dcfcfe : SUCCESS in 1h 39m 39s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/45648b7c16cc4ce3a14ad7be6335222b : SUCCESS in 1h 26m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/190838e639754d15aa8ac8374e3efb8c : SUCCESS in 29m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f0f12e37c094319be7dfcbd593c4ef7 : SUCCESS in 1h 07m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65e544cfc9a255c1675ca37371cc933b422da89d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-21 04:09:05.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4a9510ea3b6b38411e2ac5814c36dd8230eda6b3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-21 04:09:17.000000000","message":"Topic set to remove-FIPS_mode_set","accounts_in_message":[],"_revision_number":3},{"id":"8b5f509f44855721c4277d5005c593d22970d95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 06:08:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346e106c89f441d3bb4f162e5f5a6fe4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a67116053d742bca2219479d2200973 : SUCCESS in 1h 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cebba7824979474dbb1149915f4d0b93 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a416c3ca694d9fbb7d56ff373159f6 : SUCCESS in 1m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88f103f15b984f5d93792de1c488e869 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51e3dccaaff9486fb99c4ac57c1024df : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f363134cc44f9fbce6864166d897be : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4e4caa98de64c8db02a3c574489e538 : SUCCESS in 1m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/bae4c547657646799f7e050c113b407c : SUCCESS in 5m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/9f4a2007b7c44ffb803d2d296029833a : FAILURE in 5m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/519599e278764111915d4b39e2dcd30a : SUCCESS in 1h 54m 10s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/eb279f330ae6472ab6470e568f73fd73 : SUCCESS in 1h 52m 13s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0d6fa375c4604a0e8a36472a3b128bf5 : SUCCESS in 50m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/eaae1ca08f72490ea2d93ff700058208 : SUCCESS in 1h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b3e397f2aa3af07efb8b1c9eede2cf6b425f03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-21 15:13:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f8717c2c372051d7817fec4a79a572938f0eae56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-21 17:06:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a30e45c5564d98b9031ef17cdefa67\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/233bd629b9294361b3823517b624a4e9 : SUCCESS in 1h 42m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a59401d5d9684bd58cffcafc4781b3c4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c6e984a2b34f09b3465a6d4d526685 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25264cf9e94a4b908b4d14cc90efa6e3 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0866bcddab2f4aaa9db170198d283c28 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a263dccede4d539cf93f9e9f095de9 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de43c7f8a957480d8c78b9cef8c435dc : SUCCESS in 3m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/0322eb2afd164a12ba0dd2dc9ae6a397 : SUCCESS in 4m 31s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/fa45f3e5b01e40c2bd810e6f671fc37a : FAILURE in 4m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/001f7a0e5f654d7aa6aa00579cb0867e : SUCCESS in 1h 41m 21s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/25125be0a22f4b78a67728e8304fec8a : SUCCESS in 1h 48m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4929390d5c5543768eeb470438fa0614 : SUCCESS in 1h 00m 37s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b9a7c277c4a4f06b8d42b12a1a5fcbf : SUCCESS in 52m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ad53082758c99cefb02f3c7347b3487eec69cdef","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 10:45:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"613ae9c3a118857e573b162249b74e7aa1c844d3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-01-06 17:59:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a84ab3a5bfebdde20ac6adb0e0c972c12bb4b004","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-08 09:53:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"40377738fcfd1f9af6e975099f607fa4c97a2b7e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-08 12:15:26.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b46beafca8401256457f4403fbd8a3e84d84b616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-08 12:17:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"24d03a3ed4d50a350b3568e1763b8b64587d5b47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-08 12:18:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"265ea5bb93048e5882894f6a4e08f9bbec465042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 13:20:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca0340aede0148b09f311ce6b79f2df3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b59093e1937241ec857f76ba0f75c0ca : FAILURE in 20m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f4c688b1ba4deb81c3c5ddd468b62f : FAILURE in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b963b62bd5674855bff15f7242a48f1f : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712c71d8dcfa46e1930d13d6b37fb9e4 : FAILURE in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff8d90c7488c44e58637804753c4161c : FAILURE in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26ed070ee66435496da4b7d4e4e83bb : FAILURE in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4125dd3c9f564575b209e1aa33942d26 : SUCCESS in 3m 23s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/abc6a5b747584ca880364b73945f61f3 : FAILURE in 3m 35s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f8d30351a6c44a6ca97666b108ce8b6f : FAILURE in 3m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ac5564f9ac7e4160b904db30e7ee52c8 : FAILURE in 21m 07s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2eb2886bedc049a99d954432c7707d1c : FAILURE in 19m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5c01d8a1ca3346a89765c7466738227a : FAILURE in 47m 57s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/82cf185d8c7f4c0a9a5ba1ae3602b916 : FAILURE in 56m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3946b99bacfe0d797a68cb0992bccba8938bbac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-08 15:35:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"089bd99a571e107eeb2ed538b44833c147b62c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 17:23:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5729800213db41e9abd8f4576f1be915\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9740c7c7fe7a403b805bf99445ed5d70 : SUCCESS in 1h 42m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9097f83d3e8f433db03944d6a7bd0c23 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0450055ccb4b2d953fe022127ba376 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c4e061722ea4c558959bd9b59c65d56 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fa7a246d1464dbd92053c00a2257d2b : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16fae36705f46879ae2c9f6680c5167 : POST_FAILURE in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a3457b3334840cda0b541738720685e : SUCCESS in 4m 20s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/66d023435cbe4717b61ebe7e567f70b6 : SUCCESS in 4m 44s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/ee8cfc0be4504bd1b5c222723c7ebaeb : FAILURE in 6m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/73e5c2c227e84aa9aa331df2a72e77c8 : SUCCESS in 1h 35m 23s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5958d99ed35b4a929ff5192d1c512b45 : SUCCESS in 1h 21m 41s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/19d87fc3acd64fe88fac58c8976b488b : SUCCESS in 53m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b981878d93d406ea113704aadc7dab0 : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"529284b271d0c47e99889b0909724ed34b00dec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-09 02:11:53.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"719ab8b76cbc684d8e60d23f7fc22cd5135a3b84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-09 10:57:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2e9fb96611368681b8846607eb6c540673d0bd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:03:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec414cb8f994b6197276a5184ecf044\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22cfdbe3fa724a54b5c9d7c54c3b6581 : SUCCESS in 1h 29m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aed2a42d633a4179a620a932ae7fa193 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/941bc54a17e14283ba1e64f77f1f9f43 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e15b0e334e554333a4e0e83e6ec87241 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/595f92398afa41e0924b7174731dbe28 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0010355169274e8a8a7e1a38e606319d : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77f88ab458b423986aec14d2e5e651b : SUCCESS in 3m 15s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/eae8e1a7cd084e309180f7c9438f78e2 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/42eeb7a399054677bf6dc6187f45c739 : FAILURE in 5m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5106ee4e72e84c5183d7a647e62e9fcd : SUCCESS in 1h 56m 41s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/59ee4de670274f40b10f19a21e61afbf : SUCCESS in 1h 23m 36s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c7751576a7de4344bd8c482fa5f0b047 : SUCCESS in 1h 07m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/37a58f2f57d14b07913c8755450c0fe4 : SUCCESS in 1h 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c499ac7de5256ce7f7879ddb95731788519dbe71","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-01-20 15:14:19.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"846f1fc41ae94757ed634c5652e7fb8760f01bd9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 16:47:26.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b37adc14f6a524ca292c6dc5fc05363e3f01b030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:47:48.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e2585546954d04fadf286503b280200322cf4a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:21:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba977373be24dffbb9fd5477bf19bd0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf4209b77e143328a40e69bac7c6cd8 : SUCCESS in 1h 28m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4059cedbefb49d4a43a3b88993fec3b : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc03bb179bf24ac18a649abde09371ca : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3e2fb0a5c9a471888a520f4712658d8 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8a709ef82b4b098bdd604836f1648d : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/289203215024436cb50adfb7596d6421 : SUCCESS in 3m 41s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/37e0439902484c98a27ce127a9250c7c : SUCCESS in 5m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b6830585a16d4ad0ab738d487f0ed2e0 : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":7},{"id":"3c681d97793d2831fa6576b622cf5563533db838","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:21:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1307749eeb7d6b3b374bbce0149e60a080ed0dd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:22:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bef562c392dc42a58fe2a1359b83ccd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31080cebb4a04f4bb3af1f6d4e26a530 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b48a8816ad2459eb3135b6419514422 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"deb484f477e669ad68240db5abe0d45a0a2a145a","revisions":{"0a316257e6a48edc281a97b10bcb9545cb66d50f":{"kind":"REWORK","_number":1,"created":"2025-12-20 18:07:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/1"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:07:02.000000000","tz":540},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1].\n\nThe package in both Ubuntu 24.04 (3.0.13) and Debian 13 (3.5.13) are\nnewer than 3.0.0 so this interface is no longer functional.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a316257e6a48edc281a97b10bcb9545cb66d50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a316257e6a48edc281a97b10bcb9545cb66d50f"}]},"branch":"refs/heads/master"},"482c6f0e8cd9a45b99ee75d3eccc44221fd5d52a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-20 18:07:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/2"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:07:27.000000000","tz":540},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1].\n\nThe packages in both Ubuntu 24.04 (3.0.13) and Debian 13 (3.5.13) are\nnewer than 3.0.0 so this interface is no longer functional.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/482c6f0e8cd9a45b99ee75d3eccc44221fd5d52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/482c6f0e8cd9a45b99ee75d3eccc44221fd5d52a"}]},"branch":"refs/heads/master"},"f9812766de4dd918dd5cb673280b96d19b43d2ba":{"kind":"REWORK","_number":3,"created":"2025-12-21 04:09:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/3"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-21 04:09:04.000000000","tz":0},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1].\n\nThe packages in both Ubuntu 24.04 (3.0.13) and Debian 13 (3.5.13) are\nnewer than 3.0.0 so this interface is no longer functional.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f9812766de4dd918dd5cb673280b96d19b43d2ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f9812766de4dd918dd5cb673280b96d19b43d2ba"}]},"branch":"refs/heads/master"},"461b072bba5e3a655dfb02c71c7fa7c8e2a77ab1":{"kind":"REWORK","_number":4,"created":"2025-12-21 15:13:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/4"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-21 15:13:21.000000000","tz":540},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1]. The packages in both Ubuntu 24.04 (3.0.13) and\nDebian 13 (3.5.13) are newer than 3.0.0 so this interface is no longer\nfunctional.\n\nSince RHEL 9, RHEL requires that FIPS mode should be enabled at host\nlevel, instead of each application level.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/461b072bba5e3a655dfb02c71c7fa7c8e2a77ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/461b072bba5e3a655dfb02c71c7fa7c8e2a77ab1"}]},"branch":"refs/heads/master"},"a04be8710f1f69837d64a0d1e6a39afa02ab83e7":{"kind":"REWORK","_number":5,"created":"2026-01-08 12:17:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/5"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-08 12:17:02.000000000","tz":540},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1]. The packages in both Ubuntu 24.04 (3.0.13) and\nDebian 13 (3.5.13) are newer than 3.0.0 so this interface is no longer\nfunctional.\n\nSince RHEL 9, RHEL requires that FIPS mode should be enabled at host\nlevel, instead of each application level.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a04be8710f1f69837d64a0d1e6a39afa02ab83e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a04be8710f1f69837d64a0d1e6a39afa02ab83e7"}]},"branch":"refs/heads/master"},"df1bbf45dd96ad78076c6e254f4aa66124549418":{"kind":"REWORK","_number":6,"created":"2026-01-08 15:35:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/6"}}},"commit":{"parents":[{"commit":"bd16d231be1706711e39e97be81bdfe8c9ecae98","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd16d231be1706711e39e97be81bdfe8c9ecae98"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-08 15:35:33.000000000","tz":540},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1]. The packages in both Ubuntu 24.04 (3.0.13) and\nDebian 13 (3.5.13) are newer than 3.0.0 so this interface is no longer\nfunctional.\n\nSince RHEL 9, RHEL requires that FIPS mode should be enabled at host\nlevel, instead of each application level.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/df1bbf45dd96ad78076c6e254f4aa66124549418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/df1bbf45dd96ad78076c6e254f4aa66124549418"}]},"branch":"refs/heads/master"},"deb484f477e669ad68240db5abe0d45a0a2a145a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-09 02:11:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/91/971591/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/91/971591/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/91/971591/7"}}},"commit":{"parents":[{"commit":"cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f","subject":"Remove old hacking check for oslo namespace package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cc15b88d3ce1ed0753752d5f7630fc3bf66bdf4f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-20 18:06:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-09 02:11:53.000000000","tz":0},"subject":"Deprecate enforce_fips_mode","message":"Deprecate enforce_fips_mode\n\nThis option depends on FIPS_mode_set API in openssl, which was removed\nin openssl 3.0.0 [1]. The packages in both Ubuntu 24.04 (3.0.13) and\nDebian 13 (3.5.13) are newer than 3.0.0 so this interface is no longer\nfunctional.\n\nSince RHEL 9, RHEL requires that FIPS mode should be enabled at host\nlevel, instead of each application level.\n\n[1] https://github.com/openssl/openssl/commit/31b069ecea2c567de22b3874c8e71cc37c921ec9\n\nChange-Id: I059e8861256ea6a41e29bed89d12b6d5b73cf4e2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/deb484f477e669ad68240db5abe0d45a0a2a145a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/deb484f477e669ad68240db5abe0d45a0a2a145a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
