)]}'
{"id":"openstack%2Foslo.messaging~979558","triplet_id":"openstack%2Foslo.messaging~master~I0983309357fe915dd8746b8d1cb2408e4fce58db","project":"openstack/oslo.messaging","branch":"master","topic":"remove-heartbeat-in-pthread-option","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-26 15:01:19.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-10 15:30:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-03-26 13:11:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2026-03-26 15:01:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0983309357fe915dd8746b8d1cb2408e4fce58db","subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","status":"MERGED","created":"2026-03-09 12:17:33.000000000","updated":"2026-03-26 15:02:40.000000000","submitted":"2026-03-26 15:01:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"979558-remove-heartbeat-in-pthread-option","meta_rev_id":"9d63cfdf97f186b76ccef2a2e18dd0dfc3a4c9f4","_number":979558,"virtual_id_number":979558,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 15:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-03-10 15:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-03-26 13:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2026-03-11 01:39:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-11 19:35:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-03-26 13:11:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2026-03-10 15:30:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 14:14:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-03-09 14:36:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-03-09 16:44:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 09:15:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2026-03-11 01:39:00.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-03-11 19:35:44.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2026-03-26 13:11:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"af14ad0ffd4cfa13c14a20bc4269479e2407cca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 12:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18877c388d88e11d3f648f42baaeb65cc08912f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 12:22:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bfda8ef41e34a22edea1df36bbf4feea4a28c887","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:14:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"030b48ece1ec6ceaaa29015285cf9a3115d040a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 14:24:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aba45bbaac292f3dc4363dded68756e47aff39e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 14:27:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e784f26fa13c6f85a65624d434b288ae10363968","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 14:27:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8eb17b0d5d256c75b8b8daf4c4c4462f8614911f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-09 14:28:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a4060908250706b8144666bf8ba986c7cec72672","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-09 14:36:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e9cfe38069df38ca01100a50eb2072a755502987","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-09 14:42:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1cbc8d55a2e4b2d97f268516f5a17e28823e5d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:44:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e907228fc729440d823af2554e04d724\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1806e6eb48ac409dba478501525bb45f : SUCCESS in 1h 52m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8c45da7ef2d4a0eb28dc7c022bbac75 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc564c2f57a242748f1491e88290640e : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/868fe427401b4fc09fa53b8bbf414b31 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/162fffcd0ea0483ebd8f3b3c51f0fad7 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9d7293217c44bf3a561781b0979593c : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b7acb1e43644888145c7530c7ff5f6 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3eb23827dad49059390af40be2f924b : SUCCESS in 2m 30s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/34fc11db440e44e8afc9b036ddcc1585 : SUCCESS in 4m 45s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/fd2f067a357a433cab3d9b71dd79f24b : FAILURE in 8m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/77b2c479182949e99818e4d8abb83ef4 : SUCCESS in 2h 13m 22s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ea923514e1684d59b98a974fdce50b4b : SUCCESS in 1h 55m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/02d1fdaf92c14d8b84ab21f94d06e964 : SUCCESS in 1h 00m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/db38a396616e4171a5d8f3211cece9b3 : SUCCESS in 1h 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e5082501282336d8ca1e42737c592a1b95b857c8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-03-10 09:15:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fa9683dc7d43cbb6d177dbfe5fd42bacb9473836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-10 13:14:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"70761f3971cab99e1c0c3af2c4d0415f0713ce5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-10 13:16:19.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9dcf7aad6905ae04e5d5e1063f71433a37b7ce67","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-10 13:18:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fdc6aa7558e18fb33cf52a5c84a29aeb7b87ab5b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-03-10 13:20:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e1daa83321124b9942d02f96eebe7efa303ad96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:27:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dadf95a6bf9740aab4e2236b6cef2f58\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b08d8109a5c46949105cab138859a3f : SUCCESS in 1h 39m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/493dc9ded64f41fc8d579dc68cd6e64e : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18fb7f2ea5a24e3d97f621b3c0048a81 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ccae170d9c45f1bcf67cafc4f08bd5 : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d17ed5272bd47b2af3aa41306113864 : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd44261e5d5140bc8b827a97554a35af : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9146df66c9c147ac866358ffa59b70a2 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cd678d05b264887b366646d0daed804 : SUCCESS in 4m 10s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/524e395002ef45c1bc72225a51608abc : SUCCESS in 6m 21s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f9572c3e2f99451f954495a59f114e06 : FAILURE in 9m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/899bb3ec749d4fa88f5e681bd93537e5 : SUCCESS in 1h 25m 26s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/79c473bfc56f4122b3e3a41c25829c45 : SUCCESS in 1h 47m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8f670e740c8f43069b8e34c58516c373 : SUCCESS in 59m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/44f5156e15154a1c8563f0c92ae1252b : SUCCESS in 1h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f3873e1268f1fd78a5534e0f8f8e5f9a5004ce6e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-10 15:30:05.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"170b5485a9a2929a1df1b5b30e7c93477da1db04","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-11 01:39:00.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eb37a5b501c6431d266ced8356860e4e7de55a72","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-11 19:35:44.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"ffd80da4f2fd89c56b77a8afc318753e59b2d285","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-03-26 13:11:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6f4c978153b68cde9f64d1a40ed3f22d1ca2b020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 13:11:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c8ae7a553d481a76776ae876ec837401e0a121c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:01:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24952589668a42468fba59aa3077d356\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb5b7188d3e342fc9ad916307bcd62d4 : SUCCESS in 59m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2171b53e8c41b4b3feda318b70ed5e : SUCCESS in 1m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a369a0d40dbc4f51ad3a7ddd0c23cc48 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57ebac1a45bc460f8702fa5f48e1e9b1 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6473066ef8264c9daaa4c7b7661a6514 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2b8ed19a7b440968ec6ee683ef39441 : SUCCESS in 2m 39s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/d476822e7181417fa83b874bc40146a0 : SUCCESS in 4m 14s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84b1eeb973ee4d5a8ac75f4d1ce8c0df : SUCCESS in 1h 48m 07s","accounts_in_message":[],"_revision_number":6},{"id":"1360333e837ab7c56a51f6b0c12c8302a3d2cef4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:01:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9d63cfdf97f186b76ccef2a2e18dd0dfc3a4c9f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:02:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed8dfe0f8184ec6928a0493847b578e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b5acdd663b4b11a9afb433e707e276 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/378a703405c0405488fc396569b259fd : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"624f8572485676e2d17b63580948b71fb5ab2b02","revisions":{"5f852e63155dc5bb4107ad8f88522ad1ec9f057c":{"kind":"REWORK","_number":1,"created":"2026-03-09 12:17:33.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/1"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:17:21.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe driver now always runs the health check heartbeat in a native Python\nthread. Remove this option from your configuration if present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f852e63155dc5bb4107ad8f88522ad1ec9f057c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f852e63155dc5bb4107ad8f88522ad1ec9f057c"}]},"branch":"refs/heads/master"},"9ff4f33b4a7b1ea8a1cad610c0a2ff097e5a17e3":{"kind":"REWORK","_number":2,"created":"2026-03-09 12:22:36.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/2"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:22:27.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe driver now always runs the health check heartbeat in a native Python\nthread. Remove this option from your configuration if present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9ff4f33b4a7b1ea8a1cad610c0a2ff097e5a17e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9ff4f33b4a7b1ea8a1cad610c0a2ff097e5a17e3"}]},"branch":"refs/heads/master"},"68f0fffea5d1ea0e6cd1bffb8326cd99cbcc7354":{"kind":"REWORK","_number":3,"created":"2026-03-09 14:24:49.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/3"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 14:24:40.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe driver now always runs the health check heartbeat in a native Python\nthread. Remove this option from your configuration if present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/68f0fffea5d1ea0e6cd1bffb8326cd99cbcc7354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/68f0fffea5d1ea0e6cd1bffb8326cd99cbcc7354"}]},"branch":"refs/heads/master"},"1f20bdbc4f1c4e594d80b6f9b16da3e61109830e":{"kind":"REWORK","_number":4,"created":"2026-03-09 14:27:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/4"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 14:27:25.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe driver now always runs the health check heartbeat in a native Python\nthread. Remove this option from your configuration if present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f20bdbc4f1c4e594d80b6f9b16da3e61109830e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f20bdbc4f1c4e594d80b6f9b16da3e61109830e"}]},"branch":"refs/heads/master"},"3ef57432a6b79d4ba00404f968e31301d8515172":{"kind":"REWORK","_number":5,"created":"2026-03-10 13:14:21.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/5"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-10 13:14:13.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe driver now always runs the health check heartbeat in a native Python\nthread. Remove this option from your configuration if present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ef57432a6b79d4ba00404f968e31301d8515172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3ef57432a6b79d4ba00404f968e31301d8515172"}]},"branch":"refs/heads/master"},"624f8572485676e2d17b63580948b71fb5ab2b02":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-10 13:16:19.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/58/979558/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/58/979558/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/58/979558/6"}}},"commit":{"parents":[{"commit":"47639cb6f5503d77aee9f49de30f8923a51a666c","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/47639cb6f5503d77aee9f49de30f8923a51a666c"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-09 12:16:00.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-03-10 13:16:03.000000000","tz":60},"subject":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver","message":"Remove deprecated heartbeat_in_pthread option from RabbitMQ driver\n\nThe heartbeat_in_pthread option was deprecated for removal and had no\nclear functional benefit. Remove the option and the related code path.\n\nOperators should remove this option from their configuration files if\nit is present.\n\nChange-Id: I0983309357fe915dd8746b8d1cb2408e4fce58db\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/624f8572485676e2d17b63580948b71fb5ab2b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
