)]}'
{"id":"openstack%2Foslo.messaging~988980","triplet_id":"openstack%2Foslo.messaging~stable%2F2025.2~I317ebe5c3728041d15152b0c372cf471e527d1fb","project":"openstack/oslo.messaging","branch":"stable/2025.2","topic":"backport-988095-ps6-stable-2025.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-30 08:23:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-30 09:50:58.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-05-29 20:33:47.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2026-05-30 09:50:58.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-44393","bug/2150316"],"change_id":"I317ebe5c3728041d15152b0c372cf471e527d1fb","subject":"Fix RabbitMQ TLS hostname verification","status":"MERGED","created":"2026-05-18 14:06:04.000000000","updated":"2026-05-30 09:51:50.000000000","submitted":"2026-05-30 09:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988980-backport-988095-ps6-stable-2025.2","meta_rev_id":"07954b561bda1873f786e4a68cbe8f0579fa5832","_number":988980,"virtual_id_number":988980,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-30 09:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-05-30 08:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-29 20:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-05-30 08:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 16:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 14:23:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2026-05-19 15:58:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-05-27 08:08:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2026-05-28 15:57:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"6c12f22a582ce35b0da415a79f16ef887931d852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-18 14:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1944faa7266b2b5eaaddc8d7a2be842158464f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 16:03:27.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8114cf6d570d4564b00c942540c974b5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b762d353197c4067b745d7cbaae015ad : SUCCESS in 1h 51m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2048e4505f847d2bd053841d40ad667 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088a87849a464e02ae51239902867e5b : FAILURE in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4a67b7e12a24e87825b14dda90de169 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c06d1acbaa07421b88078e06dc037ac7 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf5f9c9fd5a7408094badf42490fa40e : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e39ea21f1834cb5b246a149812e731f : SUCCESS in 2m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5cb12554a9945009e973cbb49a3f90a : SUCCESS in 2m 51s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3e2748209310454f93a43529125888e0 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/22ea0f6a5fc54b3084a49c9e7910f697 : FAILURE in 16m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2b0d0999507b450880118e4d85cbc19b : SUCCESS in 1h 39m 48s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/848638cdc6a245dbb47a29381ca06272 : SUCCESS in 1h 36m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/dd3046b3fa9c4bfaaa74e9ab2f707ab9 : SUCCESS in 36m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/96e363db605446cea7be501c5a26f098 : SUCCESS in 1h 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6df4758242cd9856cc59297a8d78eefb580c2605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-19 12:30:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8dfa1ac06372e3d453eecbc117bc002902dac7a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 14:06:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad43bf45f1a47018b275b8729f9ef5f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58b6b16d9df146b3bf73196ca3ece20f : SUCCESS in 1h 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d78efd1435454e960eafd7f24029a6 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643079ecab1547e7b15a8f4dc6f0e7a0 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e27e71644c643d189dde19bffc66b94 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a686d108cf3444f8b450cd88df2808a3 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2ffad2400c94ad1ae9209c03d248289 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5f93c6fe994128ade74d720a8e7b94 : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e138e3171d48e39a9cf30aadb751fd : SUCCESS in 4m 21s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/82a8a234c9ab4392a7ccd94415a0d97f : SUCCESS in 5m 09s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/e03d21af206142aab0589e61d529b88f : FAILURE in 10m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f587af2d29954b89b2b9a017aa1ea9cf : SUCCESS in 1h 17m 57s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/eb8efeeecea64bcd8f70a31cc6134281 : SUCCESS in 1h 26m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4092b519fa104db6aa76b74fe4a48c9c : SUCCESS in 1h 06m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5491edb53be449895f4859b7e5f0901 : SUCCESS in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ab08c60f831ef49ce1b8cf0efb0c6ce6fe621a8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-19 14:23:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0ce2654a1b4c804966c203297eb1f88cbea2aee1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-19 15:58:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"303569b0a18d0908674dd7f0261297eb1ca0345d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-20 07:48:50.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"6979cd911d974a2f73d52a26b16a2b0efc35d3ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-20 13:18:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"3561a50c133610bc0d2f7f387a81f5c3afac4662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:24:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225c3de08f234e9498222d488b9aea27\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64b1147c85f448878a2df1ab69f2c73a : SUCCESS in 1h 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc993bce5f174be9b187a9144b50ed98 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3439901f94cc4ef2977bdccf5582d76d : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de5b6c52a6a5489ca74dae9d6a3ac9df : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/327cddca0d8b4d8eab5d7bcdc196718e : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7aae4642d8b412a93b90b364f57e7d1 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0978a156fa48454f99553aeae6c8586d : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e2c542fffc45888b8bdbd762843681 : SUCCESS in 5m 14s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f262d9d1899d483b83c091c99316f94c : SUCCESS in 4m 40s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/16e8417c7a7f4f4db54dc2a794281b71 : FAILURE in 12m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d41287749b1e4c6abc3496a9f648ea0a : SUCCESS in 1h 59m 41s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/662ca4bd44ca47c587bb8eeb6e4cba50 : SUCCESS in 1h 33m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/502eea9421f54420aaa75af25b8740f5 : SUCCESS in 1h 02m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c9917d7f3d54048ba52daaf3c1609ff : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6efa75bc54dc54ab9deceab963f7d83bf3a9255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-21 14:54:30.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"47961c75452cc57642e8a6e9c3ca4126e250c9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 16:45:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecabd45595624305bb30947482ee647e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e809ebd246d14dfe8e7d8620f94c49ec : SUCCESS in 1h 43m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcf3ce81506044a19d471e3c6ba4fef2 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a0a8f8b87048498ff62046be850820 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b756027598514ed6b965c464ff1b1db0 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cccf3cdacb644d87886fcecb04bc646d : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dba00319e470468dafa5e194f440094e : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c3efb2dc3f467e8248ff6066609a3b : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506330d8cffc4f1aa5ac90ebdd29d4d1 : SUCCESS in 5m 28s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/91a387d655c845f98da4c5c1b9152929 : SUCCESS in 5m 53s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/1b126bb9ab674d84b667d4d0d2980e68 : FAILURE in 10m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/37b86e81b26b49b29f614862c0b41587 : SUCCESS in 1h 47m 45s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cd38ab0faa104cf9b036200e4f21257e : SUCCESS in 1h 43m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7147d842f7e64531b176d258db0d40fc : SUCCESS in 57m 05s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/08db0cd5591c4ff2be71050feb72feaa : SUCCESS in 1h 32m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc404615a80d00b9579aca644167ac841a908402","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-21 17:00:53.000000000","message":"Hashtag added: CVE-2026-44393","accounts_in_message":[],"_revision_number":4},{"id":"23bf8381957e77f37d767cf3548e2ad9260e1f1f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-21 17:00:57.000000000","message":"Hashtag added: bug/2150316","accounts_in_message":[],"_revision_number":4},{"id":"1fbb25ab047102453114c110b409cdf85c1b745b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-25 15:00:26.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b5eabbd091281d831dbbf7c9088737a000588357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 16:37:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c46cfd3dfe4d1ebd1ea274c1eb97ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eee4f892fd564c38adc497b95eeae84d : SUCCESS in 50m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e69956ea9f5a443688a5855a52efb1df : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00590ffe3e0444e843bdb8297d1c582 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49115982d82a496488cb1a04a0b8e4da : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddc19d41c1d7453287b69bdc2c6bac90 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9fc0e5fb791341e19b74a10b620b717b : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248387a1bc8b48a68a75b88133ef86a1 : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2edadae09264902bb985a4eace3c590 : SUCCESS in 4m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/425f70d1a89b4e37aa170de5c7c7eee1 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/83defdd4d0444c47971c43e253fd6638 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c1aec195954544b385faab47c3394ddc : SUCCESS in 1h 35m 16s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/bfceaf9a2d904aa3ae76835a58f594c1 : FAILURE in 1h 19m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d007c2bcc30f42c3958c5232c8e748a3 : SUCCESS in 45m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a02382522ece4623b45bc04a59f4ec13 : SUCCESS in 38m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"015b09c5514e404eb149df81e3553c063516fff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-26 13:52:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"efdf2509e4ea9bd9cabae7e941043cc4659a3f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 15:50:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc3a435f4da4758a18e863e6d29e764\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b526740ebc6f487892446f4113cde0a3 : SUCCESS in 1h 24m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7499587ef844e5cb1933f7d74ffe5bf : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1163747900a6448cb76b94611ee0a45c : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48acceb294564df9a3e931475cfa3ebc : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c56c26dad1ee4d3a88ac7345a69c0b16 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/087a0540506d4581b52803def32161bc : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac77002d4c5d43c2979f8cdbd850ca7b : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/980f15f9683040c38eb4bdbd195ced1f : SUCCESS in 4m 40s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9ab2fea7c71b4ba3b196effcae9755bc : SUCCESS in 4m 39s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/bd0b7f21ed464268b1c61c56157a1483 : FAILURE in 8m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/addcf735462a4d1d9f946e4be46feeca : SUCCESS in 1h 54m 37s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/38ce42a7dc6d499fb2cc0d2ba56b355b : SUCCESS in 1h 47m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0dfe0cf6ef92450e8e8f56bfa73cce59 : SUCCESS in 1h 09m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/4294cd33fe5a49d4a2502c34355cc30d : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3368c35f41be48268662c2f11024529e691567ab","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-27 08:08:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"19074439fbe9cf35506fd3816a0258aa77719ebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-27 12:49:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"0ee573f81cdce93760b1ac0c5209e2448310763b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 12:52:49.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"78062e87eca9023fbdb41051890274d8c9c838f1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-28 08:24:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bfad61b3b649818ae1b184e4ea1f2187332fda8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-28 08:32:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"3b6ada9b03606b8b51e3632e3afc53c52b7ea255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:26:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b660fd018d1349bfb2a093da3d456daf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56898034530744498f63b0710ee32a3b : SUCCESS in 1h 37m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00ec05421be44147874df74e176e4f83 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a038459e14745c39f634b559cdcd65d : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fdc21b75887447ea765032ca192181b : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f20bf637eaf8407cb77ab0d4aee218a1 : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1da914f7d3074fe59f373d6a17e40f7c : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f91bab725e4389820f7739f1e9d567 : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/577df40028f94458b9280876f7bf3c06 : SUCCESS in 5m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/c75d49033d7b46e48544077caef932b2 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/47b94339977a4b2aafc72d024a737a99 : FAILURE in 8m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6f7429507ac54d29aef679cb3cc23459 : SUCCESS in 1h 44m 57s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/98d41d0bd7b74b73bd69a37ecafa6da3 : SUCCESS in 1h 40m 46s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9e3fb969a5bd4863bc4632e4547aed8a : SUCCESS in 57m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3f3387a4bda459e9a2785e47c782825 : SUCCESS in 1h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a4ae805bf5f687bb4287d71528fc818a89fdf882","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-28 15:57:58.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"318d8f51f2e1952bb2bf9119d0105c93f170be39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2026-05-28 16:33:07.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"9dbb6acc142155fdc0f034a2590b055827a77353","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-28 18:27:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"00add5dd857d65c66469c653ee098e45f094164d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:36:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b737206736f420280b498c5c7e6df83\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/094ebe5cad07449eb759619e11d54f15 : SUCCESS in 1h 58m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4430345dc23c4ac7ac982fc23bf11239 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009b2a68dc18480992f448fe8b537e0a : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ab10b9d98844077a568602871a55756 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba4acbdcdb6f41768df9c67829d13a9e : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb7f6041c1824418b4906365856e52fc : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e27309f2e845e188ca498885adb9a8 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f2f1bd29d1e41beb48914755940ce09 : SUCCESS in 4m 28s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/74f2885f49f84779b909f993c7ef324d : SUCCESS in 4m 26s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/94206d4ca34c479f8722f9fb62462d81 : FAILURE in 8m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/96ed3a3986b04a91b29e130a67388097 : SUCCESS in 1h 40m 12s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b5e362a67dc6492b8543622b76cc1251 : SUCCESS in 1h 30m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cbd91329b92c413eb2393bd1ee9cc724 : SUCCESS in 35m 00s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d81e93eccbe743b69733fc02393fe95c : SUCCESS in 1h 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e521ecebc97de8c62ad8b3f691e79dd19955667c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-29 11:43:01.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9286a23d5ecf88d7bffab118b34aa598632eaf93","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-29 11:44:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f021f55d37b04c8b26bdbf8a85029320443348d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-29 16:17:44.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"c745e8efdaf28df69576daad9cbd1c436e92171f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-29 16:21:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d5b22d7a902a3b3564af443eb5b7c1ce19bdeeee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-29 16:23:06.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fca632fb63cd635fdfaa630bd651dbad95ea1544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 18:45:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7708fb746d4f82ab9fb00669e6fc50\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1bdcc227b6b479cb1d04a13e10e2af4 : SUCCESS in 1h 59m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/660dc1904f7f49b593054a405468c678 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9cb7ccf99343b792e0e761a4cabec7 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9745c97a957240b89964997ac496339b : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dee807d780c0477ba06411180b3a42d6 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45d5eb6eed3040438a237412bcb56021 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa3fb4831084675b1a7734857c2a9e8 : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aafc8e924cb749138bb99769d4be5cc9 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/a4d58b3ba8814e25994ba78f9753ccc2 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/bb8fde59960947e595426eecfb8e8ef1 : FAILURE in 8m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6fd96a5d03ec4f31bd5f9d189d25c560 : SUCCESS in 2h 18m 05s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d48a1aec2d9a4af09db9a498eadd43a0 : SUCCESS in 2h 10m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b9f137cee8a04adfb1f7ec2f2bcfd909 : SUCCESS in 1h 03m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/aeddbc61df2d4f11bb0f996e4bc1646f : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fc6a06354b7b5dbf0c9b603c56871209ba312196","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-29 20:33:47.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"74ed33927e266bee7fec3bf91f23e65fc76de033","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-30 08:23:57.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"62048d79d3b7e1074e1fb9cbca4f4b4ffa26cba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 08:24:20.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4a7bb28565a968abc33cd7540fb73034d75e4141","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:50:58.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d46eae43c5410c8594931aa89ffdaf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6164d9dfc7bc4769b06f04faf1e4dc34 : SUCCESS in 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0105f637114a43bbb3d0c1ebfbaf26 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c307e7ade714237bb5c2e541479b8c6 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0243291875f460bbc8b7601e3a8f8da : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3bb17d857db46adb1f607587c6ebccc : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47e2c449943b41cb920288f998d36360 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/06d70d7ac65f46ad803971433fa03c5a : SUCCESS in 5m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/937a51226e50413db36919bf163edb17 : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":11},{"id":"5d970e550ced6d9c448b76c47901b8824588eeb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:50:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"07954b561bda1873f786e4a68cbe8f0579fa5832","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 09:51:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a1f0eaffee543a092bac025574c2c0b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbda4b450b174625ad682a0f50e8a0e5 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73200a8a01684721a15f9e789d49d023 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"3a9ea484debcf208cc65eea557c78579418e196c","revisions":{"559b20e59cbe474c9307880028d7101734479b99":{"kind":"REWORK","_number":1,"created":"2026-05-18 14:06:04.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/1"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-18 14:05:30.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This could allow\na MITM attacker with a certificate trusted by the deployment CA to\nimpersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled together\nwith ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed explicitly.\nFor multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required when hostname\nverification is enforced. Older Kombu versions will result in an explicit\nfailure instead of silently disabling verification.\n\nA security release note is included for CVE-2026-44393.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 8cd2e33d3d21812244aa10bec4d5838f887cbc4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/559b20e59cbe474c9307880028d7101734479b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/559b20e59cbe474c9307880028d7101734479b99"}]},"branch":"refs/heads/stable/2025.2"},"99d2e43ab14ac70e49f1e8e26a7c1ed68eda2be6":{"kind":"REWORK","_number":2,"created":"2026-05-19 12:30:10.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/2"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-19 12:25:01.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This could allow\na MITM attacker with a certificate trusted by the deployment CA to\nimpersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled together\nwith ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed explicitly.\nFor multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required when hostname\nverification is enforced. Older Kombu versions will result in an explicit\nfailure instead of silently disabling verification.\n\nA security release note is included for CVE-2026-44393.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 8cd2e33d3d21812244aa10bec4d5838f887cbc4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99d2e43ab14ac70e49f1e8e26a7c1ed68eda2be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/99d2e43ab14ac70e49f1e8e26a7c1ed68eda2be6"}]},"branch":"refs/heads/stable/2025.2"},"6180631b11a522d82d40d3cac1e2fa4ee9056cd1":{"kind":"REWORK","_number":3,"created":"2026-05-20 13:18:16.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/3"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-20 13:13:39.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6180631b11a522d82d40d3cac1e2fa4ee9056cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6180631b11a522d82d40d3cac1e2fa4ee9056cd1"}]},"branch":"refs/heads/stable/2025.2"},"6fbccd32b31befdf2b8c803bd1001ddd5430a7cb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-21 14:54:30.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/4"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-21 14:53:56.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6fbccd32b31befdf2b8c803bd1001ddd5430a7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6fbccd32b31befdf2b8c803bd1001ddd5430a7cb"}]},"branch":"refs/heads/stable/2025.2"},"a0645cfa8b0b7b6f26d1687cb1767800ef93f8b8":{"kind":"REWORK","_number":5,"created":"2026-05-25 15:00:26.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/5"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-25 12:41:55.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0645cfa8b0b7b6f26d1687cb1767800ef93f8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0645cfa8b0b7b6f26d1687cb1767800ef93f8b8"}]},"branch":"refs/heads/stable/2025.2"},"18f050f89ec5b578993b711e48ff1fb3f3e1758f":{"kind":"REWORK","_number":6,"created":"2026-05-26 13:52:57.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/6"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-26 13:51:11.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/18f050f89ec5b578993b711e48ff1fb3f3e1758f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/18f050f89ec5b578993b711e48ff1fb3f3e1758f"}]},"branch":"refs/heads/stable/2025.2"},"06e1113de5bbd17fa64a14a963a407d0702b907d":{"kind":"REWORK","_number":7,"created":"2026-05-27 12:49:49.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/7"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-27 12:47:14.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/06e1113de5bbd17fa64a14a963a407d0702b907d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/06e1113de5bbd17fa64a14a963a407d0702b907d"}]},"branch":"refs/heads/stable/2025.2"},"d49f2ae43623668d9b12e6a4a8b722e55cf91026":{"kind":"REWORK","_number":8,"created":"2026-05-28 08:32:02.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/8"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-28 08:28:07.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d49f2ae43623668d9b12e6a4a8b722e55cf91026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d49f2ae43623668d9b12e6a4a8b722e55cf91026"}]},"branch":"refs/heads/stable/2025.2"},"8e1d93d8417883701896404ab07f25db509beaa1":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2026-05-28 16:33:07.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/80/988980/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/9"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-28 16:28:42.000000000","tz":120},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 399f96e8044419ea16929a39174617ba59644052)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e1d93d8417883701896404ab07f25db509beaa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e1d93d8417883701896404ab07f25db509beaa1"}]},"branch":"refs/heads/stable/2025.2"},"7092b7c044ecb0898481a8b7932f03075419dc88":{"kind":"REWORK","_number":10,"created":"2026-05-29 16:17:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/988980/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/10"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-29 16:17:31.000000000","tz":540},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nConflicts:\n\toslo_messaging/tests/drivers/test_impl_rabbit.py\n\nResolved conflict caused by remaining unit tests for FIPS mode support.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n(cherry picked from commit 399f96e8044419ea16929a39174617ba59644052)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7092b7c044ecb0898481a8b7932f03075419dc88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7092b7c044ecb0898481a8b7932f03075419dc88"}]},"branch":"refs/heads/stable/2025.2"},"3a9ea484debcf208cc65eea557c78579418e196c":{"kind":"REWORK","_number":11,"created":"2026-05-29 16:21:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/988980/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/988980/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/988980/11"}}},"commit":{"parents":[{"commit":"d610ce64e3128489f1af547e5f53c07da159ff8a","subject":"rabbit: Fix ignored PreconditionFailed error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d610ce64e3128489f1af547e5f53c07da159ff8a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2026-05-11 11:23:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-29 16:21:11.000000000","tz":540},"subject":"Fix RabbitMQ TLS hostname verification","message":"Fix RabbitMQ TLS hostname verification\n\nWhen TLS is used with ssl_ca_file, the Rabbit driver validates the\ncertificate chain but does not verify the broker hostname. This\ncould allow a MITM attacker with a certificate trusted by the\ndeployment CA to impersonate the RabbitMQ broker.\n\nThis change adds ssl_enforce_hostname_verification. When enabled\ntogether with ssl_ca_file, the driver enables hostname verification.\n\nFor single-broker configurations, the broker hostname is passed\nexplicitly. For multi-broker configurations, Kombu \u003e\u003d 5.2.0 is required\nwhen hostname verification is enforced. Older Kombu versions will result\nin an explicit failure instead of silently disabling verification.\n\nNOTE: Unlike master, the backport keeps\nssl_enforce_hostname_verification disabled by default to preserve\nstable branch behavior compatibility.\n\nConflicts:\n\toslo_messaging/tests/drivers/test_impl_rabbit.py\n\nResolved conflict caused by remaining unit tests for FIPS mode support.\n\nRelated-Bug: #2150316\nChange-Id: I317ebe5c3728041d15152b0c372cf471e527d1fb\nSigned-off-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n(cherry picked from commit 73dc887a9caf7540685bdcb148f63d1a91f34bc0)\n(cherry picked from commit 399f96e8044419ea16929a39174617ba59644052)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3a9ea484debcf208cc65eea557c78579418e196c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3a9ea484debcf208cc65eea557c78579418e196c"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
