)]}'
{"id":"openstack%2Foslo.messaging~994668","triplet_id":"openstack%2Foslo.messaging~master~Ic1d6b8e11b6d26b62b789562437f9819506c30da","project":"openstack/oslo.messaging","branch":"master","topic":"python-next-jobs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-24 16:39:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1d6b8e11b6d26b62b789562437f9819506c30da","subject":"pre-commit: Bump versions","status":"MERGED","created":"2026-06-24 10:21:39.000000000","updated":"2026-06-24 16:40:28.000000000","submitted":"2026-06-24 16:39:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994668-python-next-jobs","meta_rev_id":"d2dcd9337e2984b1b1f8eb1879813acc4ab89084","_number":994668,"virtual_id_number":994668,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 16:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-06-24 13:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-06-24 13:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 10:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 13:09:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"bfbda954752862353972ba40ea2bae26e0efeb68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 10:21:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e649305eff30e037f8f8bec192612cfa562302f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:56:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d62ef5fb78b84f25b646f4118f169a0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba2a167ff1e4100a30ea2363d1296b3 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ba0391bc4a4f7eadcf5b3350c97f86 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35691411957d42bc9389eae999c50469 : SUCCESS in 5m 16s\n\nWarning:\n  Comments left for invalid file tools/simulator.py","accounts_in_message":[],"_revision_number":1},{"id":"05ce5d9e52173685ffc1736e91e7e00829021c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 12:46:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2703c5e3b5a01931fef470abaec5bd4625d33441","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-24 13:09:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"581811f9f7a04674b520723f894174aff8c0d18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:35:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd52d2118144d5689d341dadf389b1d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/529d733951dd4d568325280e6d0128a2 : SUCCESS in 1h 41m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92207199e17a4ab581f95dcc79a9ffd8 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11b451224f341d188e648590632b9ec : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73bbb6ddf3754228a40118f9ad58764c : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a86f04c122054809aeadd0ce5682d44a : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/620c0b8922b84b1cb196b8a450d5a951 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b061e413cbd845abaf93854614646001 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7690ccbc25a345a189211524caed8ace : SUCCESS in 5m 26s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/0fc82967c5e7402bad85a53836c83ef3 : SUCCESS in 5m 14s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/06e1f7b1db794547bf886b72e91d4227 : FAILURE in 9m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a56b584e44ac4ba8b89f49485e2d3d67 : SUCCESS in 1h 37m 29s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c6b2b0cf27c34c83bfeabf21586171ef : SUCCESS in 1h 40m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/776d7d001786441f9912330e809b699c : SUCCESS in 58m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6248e74eddf74a65b1b0dce133daf73e : SUCCESS in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59d18212c19807e1e04d753b0afaf17782fde333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:36:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ec94e89357282dc3b9afc461e1941ba52b4963d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:39:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6573da130e184822935a47dd4e203a9f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50b43463e6c04556bbcdd2cd095d6691 : SUCCESS in 1h 56m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa696ee92f647ebb0fe4e9f18a93e2f : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c8b8a81b45845b6a63227d350e409ae : SUCCESS in 6m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50faf0ddb5344bb29c4f26fa6c583dfc : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1d5edfee744d969dde9ccc53c3df47 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99b73a7a78334966bf4a3737d85b5633 : SUCCESS in 4m 03s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/7361ac13af3842bfbd37ee287d28bb36 : SUCCESS in 4m 41s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d2223f2c90e9463eaa0895f9d03837c5 : SUCCESS in 1h 25m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3d39f10d3d654efccf8f71990d9c0af69f90188c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:39:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d2dcd9337e2984b1b1f8eb1879813acc4ab89084","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 16:40:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d731cb39273640db8fe26a267f77d907\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ffff2b720343f2961361345e843deb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c78fcf78081267db3135701898c380c2d401f4e7","revisions":{"13b68c019255e1e70438f5ef5f12820b7284bcf2":{"kind":"REWORK","_number":1,"created":"2026-06-24 10:21:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/994668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/68/994668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/1"}}},"commit":{"parents":[{"commit":"5d82c240b6f73784e73eeb9006fc1c529a0c99af","subject":"Do not install code to build release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d82c240b6f73784e73eeb9006fc1c529a0c99af"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:21:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:21:04.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: Ic1d6b8e11b6d26b62b789562437f9819506c30da\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/13b68c019255e1e70438f5ef5f12820b7284bcf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/13b68c019255e1e70438f5ef5f12820b7284bcf2"}]},"branch":"refs/heads/master"},"c78fcf78081267db3135701898c380c2d401f4e7":{"kind":"REWORK","_number":2,"created":"2026-06-24 12:46:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/994668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/68/994668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/68/994668/2"}}},"commit":{"parents":[{"commit":"5d82c240b6f73784e73eeb9006fc1c529a0c99af","subject":"Do not install code to build release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d82c240b6f73784e73eeb9006fc1c529a0c99af"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:21:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:46:40.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nChange-Id: Ic1d6b8e11b6d26b62b789562437f9819506c30da\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c78fcf78081267db3135701898c380c2d401f4e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c78fcf78081267db3135701898c380c2d401f4e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
