)]}'
{"id":"openstack%2Foslo.messaging~996924","triplet_id":"openstack%2Foslo.messaging~master~If4730944deee45d734b7968fe34e637061437155","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-22 20:49:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-22 22:47:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4730944deee45d734b7968fe34e637061437155","subject":"Add ruff","status":"MERGED","created":"2026-07-12 07:25:46.000000000","updated":"2026-09-22 22:48:55.000000000","submitted":"2026-09-22 22:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996924","meta_rev_id":"4e9458d873765a43a6bc498ce9074eb7d264fb04","_number":996924,"virtual_id_number":996924,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-22 22:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-22 20:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-22 20:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 10:32:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 09:10:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-09-22 20:49:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"80d79c0fbe09c57bd445ee5c2305729ec4879420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 07:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20ef2198a7d735664278164d4dc918beb9ec5750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 07:27:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ab92bd5f5926c2f4d06134be464346a02b1d8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 10:32:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/150c549eddd24b398a6845c50f2ded44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f84f83983d8549da8b8ac09fd9b20a55 : SUCCESS in 2m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/db9c96de0a034681955b5eb2733c21b7 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35d0edc0cc344c1eb851f78f2eafbccc : FAILURE in 2h 29m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d535e4b40564433b1276bb64d966ddf : TIMED_OUT in 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2801f533c75e432ca9ebfefddf8afc48 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76f83ee5cbc343559e3d7760dcf6ddc9 : TIMED_OUT in 40m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6309ce609d26475895555507467d25db : TIMED_OUT in 41m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9dcb06ae74e64f1aa304e3ff5c26f19f : TIMED_OUT in 40m 54s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/81cc6c4cdfd84d07a4bdea7f2c14ec75 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bec6b1f68d42a1b0d9a0809d63049c : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf3115a2171429d90825c7f908bda86 : SUCCESS in 3m 41s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/03458232da19404baac191531c33f44d : TIMED_OUT in 40m 56s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/ed34a336de0a4565aa9f31e43ddee8f1 : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a9bd9ea248be4bb6be7de89c636096bb : FAILURE in 2h 39m 42s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/938e6eeec1d14eb5ba46a34754ec6951 : FAILURE in 3h 01m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7600778ad2734e47865f125a15a060e6 : FAILURE in 57m 44s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/848d5f88b848444ba21729dcee4400a0 : FAILURE in 1h 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11799e9b2386bbb40ca1de18993f863da9a168e6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-05 09:10:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc3f992772212430c53abb002261913e43614de2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-06 01:23:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"51d1d2f7b06d5ada5eac4d73a285617089025ecd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-22 13:05:51.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"306e606aa767f65d9557056487ca0b03af57766f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-22 14:03:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"de69f4a38781b3b90128f480e7ea6193929181f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-22 14:04:46.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bf824b0d4deda47cd14816b970a50e4ccff7aaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:59:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bacff6092e64427896ed953d0addfd9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9346699b37834b24aec63cf700359185 : SUCCESS in 1m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b25f5ce8a29c4655ad99d8cd91bcc9b9 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24db3f6ffb2e42eb9a367d805776aea2 : SUCCESS in 1h 43m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc23e32dd07b4f8281ddd9fc05fee47f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17cf3c5608f4e7eaacf82679501f525 : SUCCESS in 2m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69b12458fed448058d74912bd64911a5 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9a9ad89dbf54234baf2523852a20577 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16c47716a3b743c586c9feb79d7d68f8 : SUCCESS in 4m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/10cd9b160d0849fea987ae2cd8c71efb : FAILURE in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71204c44972142b9b3a9606dffe05c28 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fa18a7ee6214fdd93871150e2305e26 : SUCCESS in 5m 16s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/727f729934be4171bd4746f99610b8f5 : SUCCESS in 4m 09s\n- oslo.messaging-tox-py311-func-scenario02 https://zuul.opendev.org/t/openstack/build/3f1ab31747ee4811af20fc9a016dcd47 : FAILURE in 14m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/af11862e58cb44e9aea38b16a95fa72d : SUCCESS in 1h 36m 49s\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/295d728614a54b159bd0a1639c99783a : SUCCESS in 1h 40m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/65a064e5c32c4507befd91551caae02d : SUCCESS in 33m 23s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5da29dd474704a2c848f9f03068e11b0 : SUCCESS in 1h 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7485a5badc07b05c36720bbd9412f19afcbec328","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-22 20:49:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4fcbba1e3bc2e09920f34430b83891e552c1c866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 20:50:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e1d97d91cb5f1defa21064ef674e3eb2fb5c321","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:47:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4bae603199249178a8917c96e8bfe1c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e693a4bc09e4f0c9c9dbe05e300a28f : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90558f43b6174123a38da71abb6dbe58 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bd96dff32c546fbbcc5ec95949e4ed0 : SUCCESS in 1h 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef062e38c6264fc8bf4cbb2131c256f5 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d232146b61864b94a4fcb948654865f0 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41e2fad5db1a47509cafdf9bfda86f91 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ace08d624745e78f9dfd4f13db62a1 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cab00cc23d44fcfa4fb08d711e2ab4f : SUCCESS in 4m 37s\n- oslo.messaging-tox-py311-func-scenario01 https://zuul.opendev.org/t/openstack/build/0be147b96bfa406f9eb2e3205f7cb2e1 : SUCCESS in 5m 37s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/55628c4421d142f7be08921eb5f692a3 : SUCCESS in 1h 51m 47s","accounts_in_message":[],"_revision_number":3},{"id":"42879cca0ecc6fc6515c44169005d023b17b53f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:47:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4e9458d873765a43a6bc498ce9074eb7d264fb04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:48:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d81b872eda4113874b852e8d33ce53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b54152b0184413a38130a6e926c07b : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/111f9cd4b75640bd8a1d8ffc21e67e61 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a6b51136bc68fa977745c6b01ddf6f6390b7ac6e","revisions":{"01417619ccdb5e235e09227025e85812b7e40432":{"kind":"REWORK","_number":1,"created":"2026-07-12 07:25:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/996924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/996924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/1"}}},"commit":{"parents":[{"commit":"8ca4a4bfed0af710c90af95a7821471f898b3d93","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ca4a4bfed0af710c90af95a7821471f898b3d93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 07:25:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 07:25:14.000000000","tz":540},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: If4730944deee45d734b7968fe34e637061437155\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01417619ccdb5e235e09227025e85812b7e40432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01417619ccdb5e235e09227025e85812b7e40432"}]},"branch":"refs/heads/master"},"7ea76ad5150b6955106b742b43bdcdad5db200d5":{"kind":"REWORK","_number":2,"created":"2026-07-12 07:27:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/996924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/996924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/2"}}},"commit":{"parents":[{"commit":"8ca4a4bfed0af710c90af95a7821471f898b3d93","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8ca4a4bfed0af710c90af95a7821471f898b3d93"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 07:25:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 07:27:41.000000000","tz":540},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: If4730944deee45d734b7968fe34e637061437155\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7ea76ad5150b6955106b742b43bdcdad5db200d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7ea76ad5150b6955106b742b43bdcdad5db200d5"}]},"branch":"refs/heads/master"},"a6b51136bc68fa977745c6b01ddf6f6390b7ac6e":{"kind":"REWORK","_number":3,"created":"2026-09-22 14:03:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/996924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/24/996924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/24/996924/3"}}},"commit":{"parents":[{"commit":"87177009fb1074be0440c90364c5d4218a6f2461","subject":"Merge \"Fix rabbit consumer reconnection loop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/87177009fb1074be0440c90364c5d4218a6f2461"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-22 13:41:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-22 13:47:12.000000000","tz":540},"subject":"Add ruff","message":"Add ruff\n\nNote that format is enabled in the next change separately.\n\nChange-Id: If4730944deee45d734b7968fe34e637061437155\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a6b51136bc68fa977745c6b01ddf6f6390b7ac6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a6b51136bc68fa977745c6b01ddf6f6390b7ac6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
