)]}'
{"id":"openstack%2Foslo.metrics~855752","triplet_id":"openstack%2Foslo.metrics~master~I3d56984f6cfa569e20980aed147ea864f983720d","project":"openstack/oslo.metrics","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:24:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3d56984f6cfa569e20980aed147ea864f983720d","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:36:30.000000000","updated":"2022-09-07 09:24:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c87414d4ed2915b9a649f77bbb8649e6f25fec52","_number":855752,"virtual_id_number":855752,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-03 22:30:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 22:30:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"053d0f5cd6f59b56532577be0bafacefe311c52a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b8a5bfa8261dd47b9a684876a002b545c2f48e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 22:30:23.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/176108561a88447dbc63b266061c96c7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65957bc0431947d187e3c80368278bf7 : SUCCESS in 1h 50m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39cea0f0850f485b8bf6b4871afd3e1a : FAILURE in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/654431df762d440482012b352c6b350b : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b18dc5e44b1842e984196c2ca200e337 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c466b51de5f4a1489323909362e4293 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb94b62b8556427f95a9529e18de2962 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c87414d4ed2915b9a649f77bbb8649e6f25fec52","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:24:46.000000000","message":"Abandoned\n\nGate check passed (pep8 job failure is related to the health check patch), no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d743a340d8814d05d124969567fbe2640f64e14","revisions":{"0d743a340d8814d05d124969567fbe2640f64e14":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:36:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/52/855752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.metrics","ref":"refs/changes/52/855752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.metrics refs/changes/52/855752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.metrics refs/changes/52/855752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.metrics refs/changes/52/855752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.metrics refs/changes/52/855752/1"}}},"commit":{"parents":[{"commit":"fc22d0da1f8b58c0dd8b4dcdc86b8e87223b1bf7","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.metrics/commit/fc22d0da1f8b58c0dd8b4dcdc86b8e87223b1bf7"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:36:17.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:36:28.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I3d56984f6cfa569e20980aed147ea864f983720d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.metrics/commit/0d743a340d8814d05d124969567fbe2640f64e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.metrics/commit/0d743a340d8814d05d124969567fbe2640f64e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
