)]}'
{"id":"openstack%2Foslo.middleware~321179","triplet_id":"openstack%2Foslo.middleware~master~I2f187214f2e1a8f81a12a7705fe396f8cf07fd73","project":"openstack/oslo.middleware","branch":"master","topic":"311482","hashtags":[],"change_id":"I2f187214f2e1a8f81a12a7705fe396f8cf07fd73","subject":"Add ability to show full object table and referrers + referents","status":"ABANDONED","created":"2016-05-25 18:54:18.000000000","updated":"2021-04-21 15:36:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3cbfbcd9c93f6f8e4127edca155c6da10f2af211","_number":321179,"virtual_id_number":321179,"owner":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-18 12:03:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"date":"2020-03-07 19:58:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},{"value":2,"date":"2019-12-18 11:17:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-21 06:44:06.000000000","updated_by":{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},"reviewer":{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},"state":"REVIEWER"},{"updated":"2016-07-25 03:05:49.000000000","updated_by":{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},"reviewer":{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},"state":"REVIEWER"},{"updated":"2018-04-17 16:19:52.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-18 11:17:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-18 12:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-07 19:58:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"a796278dd82c3f5c4e7204ea4e611c19cd567475","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-05-25 18:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce293e42f11f14f6dafc39e3a889042f44d4eae6","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-05-25 19:10:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b436a2cae5bd43b1b13ea4f6ef0654fd8ba08b97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-25 23:00:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.middleware-docs http://docs-draft.openstack.org/79/321179/2/check/gate-oslo.middleware-docs/e8aba7e//doc/build/html/ : SUCCESS in 1m 21s\n- gate-oslo.middleware-pep8 http://logs.openstack.org/79/321179/2/check/gate-oslo.middleware-pep8/2718f78/ : SUCCESS in 1m 50s\n- gate-oslo.middleware-python27 http://logs.openstack.org/79/321179/2/check/gate-oslo.middleware-python27/6481224/ : FAILURE in 1m 56s\n- gate-oslo.middleware-python34 http://logs.openstack.org/79/321179/2/check/gate-oslo.middleware-python34/24637bc/ : FAILURE in 3m 35s\n- gate-oslo.middleware-pypy http://logs.openstack.org/79/321179/2/check/gate-oslo.middleware-pypy/18a5685/ : FAILURE in 3m 19s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware http://logs.openstack.org/79/321179/2/check/gate-tempest-dsvm-neutron-src-oslo.middleware/6eaabee/ : SUCCESS in 1h 02m 28s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka http://logs.openstack.org/79/321179/2/check/gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka/0af2842/ : SUCCESS in 54m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02bd961439476f4dc0bc0f70ef7271c7606dda75","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-05-27 07:00:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"74c5f02d03e3e35464e9a76a013f92bb402a2994","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-27 07:56:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-oslo.middleware-docs http://docs-draft.openstack.org/79/321179/3/check/gate-oslo.middleware-docs/27d6d8a//doc/build/html/ : SUCCESS in 1m 24s\n- gate-oslo.middleware-pep8 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-pep8/157c26a/ : SUCCESS in 2m 51s\n- gate-oslo.middleware-python27 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-python27/380238c/ : SUCCESS in 1m 24s\n- gate-oslo.middleware-python34 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-python34/aa1c576/ : SUCCESS in 1m 33s\n- gate-oslo.middleware-pypy http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-pypy/59ef833/ : FAILURE in 4m 13s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware http://logs.openstack.org/79/321179/3/check/gate-tempest-dsvm-neutron-src-oslo.middleware/957c5a1/ : FAILURE in 46s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka http://logs.openstack.org/79/321179/3/check/gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka/1fd0fdb/ : FAILURE in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f7855cfbffb4139318b6ca51156150daf835ba5","author":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"date":"2016-05-27 21:26:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"63256decfecd06296cc3692e96bcade010880b55","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-05-27 22:30:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.middleware-docs http://docs-draft.openstack.org/79/321179/3/check/gate-oslo.middleware-docs/f922921//doc/build/html/ : SUCCESS in 3m 43s\n- gate-oslo.middleware-pep8 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-pep8/1225ec1/ : SUCCESS in 3m 03s\n- gate-oslo.middleware-python27 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-python27/bb0d761/ : SUCCESS in 3m 32s\n- gate-oslo.middleware-python34 http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-python34/2a4a715/ : SUCCESS in 3m 30s\n- gate-oslo.middleware-pypy http://logs.openstack.org/79/321179/3/check/gate-oslo.middleware-pypy/9c571f2/ : FAILURE in 4m 57s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware http://logs.openstack.org/79/321179/3/check/gate-tempest-dsvm-neutron-src-oslo.middleware/1d1918b/ : SUCCESS in 1h 03m 21s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka http://logs.openstack.org/79/321179/3/check/gate-tempest-dsvm-neutron-src-oslo.middleware-mitaka/a708851/ : SUCCESS in 1h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"064e5a2298271880eb6cedf2229b1f7d9ae644c6","author":{"_account_id":5538,"name":"Rushi Agrawal","email":"rushi.openstack@gmail.com","username":"rushiagr"},"date":"2016-07-21 06:44:06.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM. Note: I\u0027m new to oslo middleware, just getting started..","accounts_in_message":[],"_revision_number":3},{"id":"0a29d0822417b18caf6e35f51674081f7fa53df0","author":{"_account_id":9459,"name":".Haifeng Yan","email":"yanheven@qq.com","username":"yan"},"date":"2016-07-25 03:05:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7ee7645375e65b2930fdf46075b2f3c1e174e09","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-11-02 02:55:13.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b110a8780c1612436d4e7be5f9d46c98d3a8e26e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 04:43:47.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":3},{"id":"3662fb253bfad7555534560f2062687eef77eb95","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2018-04-17 16:19:52.000000000","message":"Patch Set 3: Workflow-1\n\nThis still seems useful, but it\u0027s been in merge conflict for a long time.","accounts_in_message":[],"_revision_number":3},{"id":"c7880f068fa853f2674ca7993d349f1b5d2a9d54","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 11:16:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2a095f5774786956926e6b41945f6d09caa86c98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 11:17:15.000000000","message":"Patch Set 4: Code-Review+2\n\nAddressed the (trivial) merge conflict. Idea makes sense to me.","accounts_in_message":[],"_revision_number":4},{"id":"4751586ce9609345bf5ec23ec351a5122128abd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 12:03:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/617360d63b964f0187acbf58336b8b4c : FAILURE in 26m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1ef2e40645e4748b2a24b4bf62541f8 : FAILURE in 24m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7590b7e4fa364345bf684db79a2d957c : TIMED_OUT in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a51ed89337f43e6b9435fd7cbd8e22a : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57187cdd9cf04b7f9e85e23a62c6e1e1 : TIMED_OUT in 41m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ceb0953d26214f3c86d4ed8b88a5b510 : TIMED_OUT in 41m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be1c09a5396d4cd18d280c5f86247434 : TIMED_OUT in 41m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bacb46570874ddc8b88ebdebba1503d : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"7bd443eb4aa7f1d777b44131120caae4a2d129b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 13:06:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d09c8fe8f313084dd3fa65e157db7974035bfe92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:55:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4bbfa9cc45cb481da77bb298e866180b : FAILURE in 19m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a88633c1c5049638aa015b4deb154c3 : FAILURE in 25m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84a0e9c28ff24214b5a6a5381e28c8c3 : TIMED_OUT in 41m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1fedec0cc134a27ab4af4fd0f0ec29d : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fe3589a8ea54af5949622ef4d1c383d : TIMED_OUT in 41m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b1ede7ddd0d490b892755754ab57935 : TIMED_OUT in 41m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c7daa6dc2ca46ceb4f771bd02aea593 : TIMED_OUT in 42m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0685ba666bae486986a48d3593223132 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2dbf4a8a8f040799e6ac8ed2ebbc9f34d858eab9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-07 19:24:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck\n\nLooks good to me. Some py2 compat code that isn\u0027t needed anymore, but that can be cleaned up later.","accounts_in_message":[],"_revision_number":4},{"id":"ae531e074edf82022ccd8b632e2ca88f601bbbfb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-07 19:58:14.000000000","message":"Patch Set 4: -Code-Review -Workflow\n\nKeystone issue to address?\n\nERROR keystone.server.flask.application NameError: name \u0027conf\u0027 is not defined","accounts_in_message":[],"_revision_number":4},{"id":"0c29a2bc2ff7abd81b72eed363aeface404ca84e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 20:10:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/599a1717876d41f6903324de01b9579e : FAILURE in 18m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a59bcf7b3241441893c16ebc1902698d : TIMED_OUT in 41m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d55c3d4649442b8fee18db84e09a0d : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30157743f51840ffb865b30526d94ae5 : TIMED_OUT in 40m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d390056fee247e89daf3d5779501e0f : TIMED_OUT in 41m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6346ef6724d46c1b041105b5603bf9f : TIMED_OUT in 41m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60729de2c9fc40d285cb6fc37ed0bcee : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"3cbfbcd9c93f6f8e4127edca155c6da10f2af211","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-21 15:36:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2d942a4da2a70e16ac82cbb74f68f59f7fbbadf5","revisions":{"6d20f4985e21e690ce13412c4c56d3156b0fbc38":{"kind":"REWORK","_number":1,"created":"2016-05-25 18:54:18.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/79/321179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/321179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/1"}}},"commit":{"parents":[{"commit":"6f1db02ef07137907977e7437d1e2df7619a429d","subject":"cors: remove unused import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6f1db02ef07137907977e7437d1e2df7619a429d"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 18:52:32.000000000","tz":-420},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 18:53:45.000000000","tz":-420},"subject":"Add ability to show full object table and referrers + referents","message":"Add ability to show full object table and referrers + referents\n\nIt can be quite useful to see the full object table and how many\nobjects are being referred to diagnois GC issues so include that\ninformation (which takes a long time to gather) via a new option\n\u0027gc_detailed\u0027 (which is by default off).\n\nChange-Id: I2f187214f2e1a8f81a12a7705fe396f8cf07fd73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6d20f4985e21e690ce13412c4c56d3156b0fbc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6d20f4985e21e690ce13412c4c56d3156b0fbc38"}]},"branch":"refs/heads/master"},"d1a2524985ec174950e878332e6985cc6083ad47":{"kind":"REWORK","_number":2,"created":"2016-05-25 19:10:12.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/79/321179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/321179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/2"}}},"commit":{"parents":[{"commit":"6f1db02ef07137907977e7437d1e2df7619a429d","subject":"cors: remove unused import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6f1db02ef07137907977e7437d1e2df7619a429d"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 18:52:32.000000000","tz":-420},"committer":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 19:08:52.000000000","tz":-420},"subject":"Add ability to show full object table and referrers + referents","message":"Add ability to show full object table and referrers + referents\n\nIt can be quite useful to see the full object table and how many\nobjects are being referred to diagnois GC issues so include that\ninformation (which takes a long time to gather) via a new option\n\u0027gc_detailed\u0027 (which is by default off).\n\nChange-Id: I2f187214f2e1a8f81a12a7705fe396f8cf07fd73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d1a2524985ec174950e878332e6985cc6083ad47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d1a2524985ec174950e878332e6985cc6083ad47"}]},"branch":"refs/heads/master"},"b496fb10cfccb49ea0d5032ff7d240cfcd081b3c":{"kind":"REWORK","_number":3,"created":"2016-05-27 07:00:43.000000000","uploader":{"_account_id":1297,"name":"Joshua Harlow","email":"jxharlow@godaddy.com","username":"harlowja"},"ref":"refs/changes/79/321179/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/321179/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/3"}}},"commit":{"parents":[{"commit":"6f1db02ef07137907977e7437d1e2df7619a429d","subject":"cors: remove unused import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/6f1db02ef07137907977e7437d1e2df7619a429d"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 18:52:32.000000000","tz":-420},"committer":{"name":"Joshua Harlow","email":"harlowja@gmail.com","date":"2016-05-27 07:00:46.000000000","tz":-420},"subject":"Add ability to show full object table and referrers + referents","message":"Add ability to show full object table and referrers + referents\n\nIt can be quite useful to see the full object table and how many\nobjects are being referred to diagnois GC issues so include that\ninformation (which takes a long time to gather) via a new option\n\u0027gc_detailed\u0027 (which is by default off).\n\nChange-Id: I2f187214f2e1a8f81a12a7705fe396f8cf07fd73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/b496fb10cfccb49ea0d5032ff7d240cfcd081b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/b496fb10cfccb49ea0d5032ff7d240cfcd081b3c"}]},"branch":"refs/heads/master"},"2d942a4da2a70e16ac82cbb74f68f59f7fbbadf5":{"kind":"REWORK","_number":4,"created":"2019-12-18 11:16:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/321179/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/79/321179/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/79/321179/4"}}},"commit":{"parents":[{"commit":"7c47de583a1928dafd3a9e7796360a8b7e3fb6ce","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7c47de583a1928dafd3a9e7796360a8b7e3fb6ce"}]}],"author":{"name":"Joshua Harlow","email":"jxharlow@godaddy.com","date":"2016-05-25 18:52:32.000000000","tz":-420},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-12-18 11:16:46.000000000","tz":0},"subject":"Add ability to show full object table and referrers + referents","message":"Add ability to show full object table and referrers + referents\n\nIt can be quite useful to see the full object table and how many\nobjects are being referred to diagnois GC issues so include that\ninformation (which takes a long time to gather) via a new option\n\u0027gc_detailed\u0027 (which is by default off).\n\nChange-Id: I2f187214f2e1a8f81a12a7705fe396f8cf07fd73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2d942a4da2a70e16ac82cbb74f68f59f7fbbadf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2d942a4da2a70e16ac82cbb74f68f59f7fbbadf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
