)]}'
{"id":"openstack%2Foslo.middleware~453712","triplet_id":"openstack%2Foslo.middleware~master~I4e0932dd8664d96b2733c5d3b1fa846af0c7d132","project":"openstack/oslo.middleware","branch":"master","topic":"bug/1679658","hashtags":[],"change_id":"I4e0932dd8664d96b2733c5d3b1fa846af0c7d132","subject":"Fix test_request_too_large_no_content_length failure","status":"MERGED","created":"2017-04-05 15:30:07.000000000","updated":"2017-04-14 15:20:30.000000000","submitted":"2017-04-14 15:20:30.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7450f50371e2fb54e19d1e040515467bd08eb8f","_number":453712,"virtual_id_number":453712,"owner":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-04-14 15:20:30.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2017-04-14 14:03:58.000000000","_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2017-04-13 13:13:00.000000000","_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":1,"date":"2017-04-14 14:03:58.000000000","_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-07 10:54:07.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2017-04-07 16:01:38.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2017-04-13 13:13:00.000000000","updated_by":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"reviewer":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"state":"REVIEWER"},{"updated":"2017-04-14 14:03:58.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2017-04-14 15:20:30.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"fd40cb3bae28adb80991e0ee786e099271c61197","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-05 15:30:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"445de219244a065c38e767087c14451749a80a8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 16:48:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.middleware-docs-ubuntu-xenial http://docs-draft.openstack.org/12/453712/1/check/gate-oslo.middleware-docs-ubuntu-xenial/8e1fb0c//doc/build/html/ : SUCCESS in 3m 34s\n- gate-oslo.middleware-pep8-ubuntu-xenial http://logs.openstack.org/12/453712/1/check/gate-oslo.middleware-pep8-ubuntu-xenial/7b0e553/ : SUCCESS in 3m 05s\n- gate-oslo.middleware-python27-ubuntu-xenial http://logs.openstack.org/12/453712/1/check/gate-oslo.middleware-python27-ubuntu-xenial/067629e/ : SUCCESS in 3m 36s\n- gate-oslo.middleware-python35 http://logs.openstack.org/12/453712/1/check/gate-oslo.middleware-python35/f1225e0/ : SUCCESS in 3m 08s\n- gate-oslo.middleware-pypy http://logs.openstack.org/12/453712/1/check/gate-oslo.middleware-pypy/7c958e9/ : FAILURE in 4m 30s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial http://logs.openstack.org/12/453712/1/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial/b2d6616/ : SUCCESS in 1h 15m 12s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-newton http://logs.openstack.org/12/453712/1/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-newton/6a473f8/ : FAILURE in 15m 41s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-ocata http://logs.openstack.org/12/453712/1/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-ocata/cc74947/ : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2af9794a37df55a34cb1a44ff0744c535d5f8dd","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2017-04-07 10:54:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f35da3891e0f7b0b714e010f45c55e5a953b298","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-04-07 16:01:38.000000000","message":"Patch Set 1:\n\nShould this patch also raise the minimum version of WebOb so that the change in behavior reflects what we say works with the library? Or do the changes also actually work with 1.6.0 (the minimum version listed in master for oslo.middleware\u0027s requirements right now)? Maybe merge https://review.openstack.org/#/c/452481/1 into this patch?","accounts_in_message":[],"_revision_number":1},{"id":"86e6c42102f13a7d346e81411e7c66fc28a5809c","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2017-04-07 18:48:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f36cba52cea9732bef6e4fb5d43dfbd08bae106","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-10 03:21:43.000000000","message":"Patch Set 1:\n\n\u003e Should this patch also raise the minimum version of WebOb so that\n \u003e the change in behavior reflects what we say works with the library?\n \u003e Or do the changes also actually work with 1.6.0 (the minimum\n \u003e version listed in master for oslo.middleware\u0027s requirements right\n \u003e now)? Maybe merge https://review.openstack.org/#/c/452481/1 into\n \u003e this patch?\n\nThis works with 1.6.0 , we can rebase https://review.openstack.org/#/c/452481/1  when we merge this one","accounts_in_message":[],"_revision_number":1},{"id":"e4f3b54fe662a35833ea217277b28987538b60b7","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-04-11 09:02:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e1687013c21b9d6df9da1ab3d9d8f7a82be487a","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-11 10:44:03.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nThat\u0027s the behavior WebOb\u003e \u003d1.7.0 and set body with bytes directly, our test just covers this case. We still can read body from file,  context_length may be None.","accounts_in_message":[],"_revision_number":1},{"id":"cad05219b43b71e602b0fc06b2e80683046adaf2","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-04-13 10:31:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"664c97426bc9146815b94d4c54e8d625819ef0df","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-13 12:56:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1377124d2923ec2a4b22ef59f59f84bab3dc3c96","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2017-04-13 12:58:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b7141432aff2fe665a45be612b8653771a29c4e0","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2017-04-13 13:13:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"018db841ce0c4651799c569caae2adfc05ded171","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-13 14:13:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-oslo.middleware-docs-ubuntu-xenial http://docs-draft.openstack.org/12/453712/2/check/gate-oslo.middleware-docs-ubuntu-xenial/b4fb6c6//doc/build/html/ : SUCCESS in 2m 25s\n- gate-oslo.middleware-pep8-ubuntu-xenial http://logs.openstack.org/12/453712/2/check/gate-oslo.middleware-pep8-ubuntu-xenial/469b004/ : SUCCESS in 2m 40s\n- gate-oslo.middleware-python27-ubuntu-xenial http://logs.openstack.org/12/453712/2/check/gate-oslo.middleware-python27-ubuntu-xenial/88419ac/ : SUCCESS in 2m 06s\n- gate-oslo.middleware-python35 http://logs.openstack.org/12/453712/2/check/gate-oslo.middleware-python35/3ab7cae/ : SUCCESS in 2m 47s\n- gate-oslo.middleware-pypy http://logs.openstack.org/12/453712/2/check/gate-oslo.middleware-pypy/c88e688/ : FAILURE in 3m 33s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial http://logs.openstack.org/12/453712/2/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial/5427d72/ : SUCCESS in 1h 12m 23s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-newton http://logs.openstack.org/12/453712/2/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-newton/44d2a04/ : FAILURE in 7m 57s (non-voting)\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-ocata http://logs.openstack.org/12/453712/2/check/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial-ocata/5c5eb86/ : FAILURE in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41bd131b7dfb71da0f5e8ea7ba92f1b142ff3bd3","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-04-14 14:03:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI don\u0027t like having a test removed but I will trust WebOb then.","accounts_in_message":[],"_revision_number":2},{"id":"3264a76049da7fdf57219c4bda790e6fb53992bf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-14 14:04:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"737f4a8a0923cabffec34b2b2dc850b33f3c9b64","date":"2017-04-14 15:20:30.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"a6822995ba12a858462a7bf18af03976d418c8e8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-14 15:20:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-oslo.middleware-docs-ubuntu-xenial http://docs-draft.openstack.org/12/453712/2/gate/gate-oslo.middleware-docs-ubuntu-xenial/78d4a62//doc/build/html/ : SUCCESS in 2m 47s\n- gate-oslo.middleware-pep8-ubuntu-xenial http://logs.openstack.org/12/453712/2/gate/gate-oslo.middleware-pep8-ubuntu-xenial/67d77f5/ : SUCCESS in 2m 38s\n- gate-oslo.middleware-python27-ubuntu-xenial http://logs.openstack.org/12/453712/2/gate/gate-oslo.middleware-python27-ubuntu-xenial/349ad40/ : SUCCESS in 2m 51s\n- gate-oslo.middleware-python35 http://logs.openstack.org/12/453712/2/gate/gate-oslo.middleware-python35/e351b9a/ : SUCCESS in 1m 57s\n- gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial http://logs.openstack.org/12/453712/2/gate/gate-tempest-dsvm-neutron-src-oslo.middleware-ubuntu-xenial/fd1895f/ : SUCCESS in 1h 16m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fed29257358f990dd91eccd9be08e4cbcea53dca","revisions":{"d19143da0f7fb42621512dddcd2d189287fe8008":{"kind":"REWORK","_number":1,"created":"2017-04-05 15:30:07.000000000","uploader":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"ref":"refs/changes/12/453712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/12/453712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/1"}}},"commit":{"parents":[{"commit":"2dc2cf3c31ca14ac2d4894b1cd98a4d8e4a76b9d","subject":"Remove log translations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2dc2cf3c31ca14ac2d4894b1cd98a4d8e4a76b9d"}]}],"author":{"name":"ChangBo Guo(gcb)","email":"eric.guo@easystack.cn","date":"2017-04-05 15:12:13.000000000","tz":480},"committer":{"name":"ChangBo Guo(gcb)","email":"eric.guo@easystack.cn","date":"2017-04-05 15:25:05.000000000","tz":480},"subject":"Fix test_request_too_large_no_content_length failure","message":"Fix test_request_too_large_no_content_length failure\n\nWebob \u003e\u003d1.7.0 change its way to check if body is readable [1],\nThe is_body_readable() method returns False now. request.content_length\nwill be set when set request.body automatically, and if\nrequest.context_length is None request.body will return returns b\u0027\u0027.[3]\nSo just set request.body in test.\n\n[1] https://github.com/Pylons/webob/commit/0e6fcbfbc58af6fdd4e125543bf8f5fb91b984c6\n[2] https://github.com/Pylons/webob/blame/master/webob/request.py#L699\n[3] https://github.com/Pylons/webob/blame/master/webob/request.py#L692\n\nCloses-Bug: #1679658\n\nChange-Id: I4e0932dd8664d96b2733c5d3b1fa846af0c7d132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d19143da0f7fb42621512dddcd2d189287fe8008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d19143da0f7fb42621512dddcd2d189287fe8008"}]},"branch":"refs/heads/master"},"fed29257358f990dd91eccd9be08e4cbcea53dca":{"kind":"REWORK","_number":2,"created":"2017-04-13 12:58:26.000000000","uploader":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"ref":"refs/changes/12/453712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/12/453712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/12/453712/2"}}},"commit":{"parents":[{"commit":"2dc2cf3c31ca14ac2d4894b1cd98a4d8e4a76b9d","subject":"Remove log translations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/2dc2cf3c31ca14ac2d4894b1cd98a4d8e4a76b9d"}]}],"author":{"name":"ChangBo Guo(gcb)","email":"eric.guo@easystack.cn","date":"2017-04-05 15:12:13.000000000","tz":480},"committer":{"name":"ChangBo Guo(gcb)","email":"eric.guo@easystack.cn","date":"2017-04-13 12:57:50.000000000","tz":480},"subject":"Fix test_request_too_large_no_content_length failure","message":"Fix test_request_too_large_no_content_length failure\n\nWebob \u003e\u003d1.7.0 change its way to check if body is readable [1],\nThe is_body_readable() method returns False now. request.content_length\nwill be set when set request.body automatically, and if\nrequest.context_length is None request.body will return returns b\u0027\u0027.[3]\nSo just set request.body in test.\n\n[1] https://github.com/Pylons/webob/commit/0e6fcbfbc58af6fdd4e125543bf8f5fb91b984c6\n[2] https://github.com/Pylons/webob/blame/master/webob/request.py#L699\n[3] https://github.com/Pylons/webob/blame/master/webob/request.py#L692\n\nCloses-Bug: #1679658\n\nChange-Id: I4e0932dd8664d96b2733c5d3b1fa846af0c7d132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/fed29257358f990dd91eccd9be08e4cbcea53dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/fed29257358f990dd91eccd9be08e4cbcea53dca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
