)]}'
{"id":"openstack%2Foslo.middleware~610642","triplet_id":"openstack%2Foslo.middleware~master~Ib541de637ecbee664fa281b5920ece2c66055b13","project":"openstack/oslo.middleware","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ib541de637ecbee664fa281b5920ece2c66055b13","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2018-10-15 16:26:51.000000000","updated":"2019-07-10 21:34:58.000000000","submitted":"2019-07-10 21:34:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"610642-1562794498314-31ef7618","meta_rev_id":"6ac35ee6181af3de97d29652fb6ca7681df4cc50","_number":610642,"virtual_id_number":610642,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-10 21:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-07-10 19:45:13.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-08 14:37:58.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-07-10 19:45:13.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 14:37:58.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-10 19:45:13.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-10 21:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9037e2a5b4f57c561e22f969529c005e44ac86b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2018-10-15 16:26:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b137ccff9174c95e77f037b65cac4480f69ea236","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2018-10-15 19:43:36.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s wait for this to settle: https://review.openstack.org/#/c/610708/","accounts_in_message":[],"_revision_number":1},{"id":"33fc446fc47abc4e861de6e3ca9b380289c0285c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-15 20:27:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/42/610642/1/check/openstack-tox-pep8/7fcb683/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/42/610642/1/check/openstack-tox-py27/9eec91d/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/42/610642/1/check/openstack-tox-docs/96ec113/html/ : SUCCESS in 4m 06s\n- tempest-full http://logs.openstack.org/42/610642/1/check/tempest-full/a351434/ : SUCCESS in 1h 42m 02s\n- tempest-full-py3 http://logs.openstack.org/42/610642/1/check/tempest-full-py3/e177c50/ : SUCCESS in 1h 25m 11s\n- openstack-tox-py36 http://logs.openstack.org/42/610642/1/check/openstack-tox-py36/4d707d7/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/42/610642/1/check/openstack-tox-py37/1af4508/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/610642/1/check/openstack-tox-lower-constraints/bd7edb3/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f4ed3c40a46a359a624a684ae5db3738265b645d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:48:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c5e6bbe1de4b745ad7084d1c524728cfb37d4a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:43:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/42/610642/2/check/test-release-openstack/cfc6b5b/ : SUCCESS in 2m 56s\n- tempest-full http://logs.openstack.org/42/610642/2/check/tempest-full/251aadd/ : SUCCESS in 1h 33m 06s\n- tempest-full-py3 http://logs.openstack.org/42/610642/2/check/tempest-full-py3/7a56677/ : SUCCESS in 1h 51m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/610642/2/check/openstack-tox-lower-constraints/85f335b/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/42/610642/2/check/openstack-tox-pep8/db6ebeb/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/42/610642/2/check/openstack-tox-py27/12e2ae5/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/42/610642/2/check/openstack-tox-py36/3d2719e/ : SUCCESS in 2m 57s\n- openstack-tox-py37 http://logs.openstack.org/42/610642/2/check/openstack-tox-py37/ea42eb5/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/42/610642/2/check/openstack-tox-docs/db3ac1d/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/42/610642/2/check/build-openstack-releasenotes/62f633e/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a5a9ff2bb46b4bc3c709e6370b7c73020f9bdf6b","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-08 14:37:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c350b368f9884a0a867c541c7fce9ff330b9a806","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-10 19:45:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f3404a5226f2a86b9cf1aaa976d8bbf8fb090582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:45:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad858713bada732247b884cf7d29671cce90c10f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:34:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"20dcb86071c4771da8099cd67ce6c0639c04f560","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 21:34:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/42/610642/2/gate/test-release-openstack/0f1ec47/ : SUCCESS in 1m 47s\n- tempest-full http://logs.openstack.org/42/610642/2/gate/tempest-full/12eda16/ : SUCCESS in 1h 42m 43s\n- tempest-full-py3 http://logs.openstack.org/42/610642/2/gate/tempest-full-py3/34f175a/ : SUCCESS in 1h 28m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/42/610642/2/gate/openstack-tox-lower-constraints/bb69fbc/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/42/610642/2/gate/openstack-tox-pep8/6306730/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/42/610642/2/gate/openstack-tox-py27/14cfd4a/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/42/610642/2/gate/openstack-tox-py36/34d0697/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/42/610642/2/gate/openstack-tox-py37/f312210/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/42/610642/2/gate/openstack-tox-docs/93b9f0e/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/42/610642/2/gate/build-openstack-releasenotes/a350589/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1a47b9e6abb7738e751bdd4edaa588a4151bf60","revisions":{"d8d95dfbff432093daffccbf415951205ac1b2c8":{"kind":"REWORK","_number":1,"created":"2018-10-15 16:26:51.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/42/610642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/42/610642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/1"}}},"commit":{"parents":[{"commit":"dc2ba62297b5b8a8c9eeecbe2a4678fe4844a1f7","subject":"Remove moxstubout usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/dc2ba62297b5b8a8c9eeecbe2a4678fe4844a1f7"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 13:34:58.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2018-10-15 13:34:58.000000000","tz":-240},"subject":"Change python3.5 job to python3.7 job on Stein+","message":"Change python3.5 job to python3.7 job on Stein+\n\nThis is a mechanically generated patch to add a unit test job running\nunder Python 3.7 and drop the Python 3.5 unit test job.\n\npython3.5 was the only supported python3 version on Xenial. Now that\nwe have Bionic nodes supporting python3.6 and python3.7, let\u0027s switch\nto testing with python3.7 in addition to python3.6 in Stein and\nbeyond.\n\nSee ML discussion here [1] for context.\n\n[1] http://lists.openstack.org/pipermail/openstack-dev/2018-October/135626.html\n\nChange-Id: Ib541de637ecbee664fa281b5920ece2c66055b13\nStory: #2004073\nTask: #27440\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d8d95dfbff432093daffccbf415951205ac1b2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d8d95dfbff432093daffccbf415951205ac1b2c8"}]},"branch":"refs/heads/master"},"f1a47b9e6abb7738e751bdd4edaa588a4151bf60":{"kind":"REWORK","_number":2,"created":"2019-07-05 18:48:44.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/42/610642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/42/610642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/42/610642/2"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:51.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:48:43.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ib541de637ecbee664fa281b5920ece2c66055b13\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f1a47b9e6abb7738e751bdd4edaa588a4151bf60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f1a47b9e6abb7738e751bdd4edaa588a4151bf60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
