)]}'
{"id":"openstack%2Foslo.middleware~659314","triplet_id":"openstack%2Foslo.middleware~master~Ifdad77dfc957ed90da88efe729cc6b35e96f0392","project":"openstack/oslo.middleware","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"Ifdad77dfc957ed90da88efe729cc6b35e96f0392","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","status":"MERGED","created":"2019-05-15 15:23:50.000000000","updated":"2019-05-16 11:02:50.000000000","submitted":"2019-05-16 11:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659314-1558004570422-4a183efd","meta_rev_id":"0c8c1dee0e0c3e3d5a8b3d549448b669c2a3b1c5","_number":659314,"virtual_id_number":659314,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-05-16 11:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-05-15 16:14:22.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-16 09:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 16:14:22.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-16 09:06:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-05-16 11:02:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f1aced0b1519c74875e80644d297a4f9ede69a4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-15 15:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb4642ae5383f022b9c80d93f8598b31756f575b","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-15 16:14:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"836b93583e8a3e46c99f8396ed0d3106a0add6ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/14/659314/1/check/requirements-check/b886b35/ : SUCCESS in 9m 00s\n- tempest-full http://logs.openstack.org/14/659314/1/check/tempest-full/6979e6c/ : SUCCESS in 2h 01m 02s\n- tempest-full-py3 http://logs.openstack.org/14/659314/1/check/tempest-full-py3/95ae38e/ : SUCCESS in 1h 46m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/659314/1/check/openstack-tox-lower-constraints/c8c5782/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/14/659314/1/check/openstack-tox-pep8/8759f85/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/14/659314/1/check/openstack-tox-py27/f455322/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/14/659314/1/check/openstack-tox-py35/25ca563/ : SUCCESS in 6m 49s\n- openstack-tox-py36 http://logs.openstack.org/14/659314/1/check/openstack-tox-py36/b50e7c0/ : SUCCESS in 4m 43s\n- openstack-tox-py37 http://logs.openstack.org/14/659314/1/check/openstack-tox-py37/7555181/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/14/659314/1/check/openstack-tox-docs/4bad767/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes http://logs.openstack.org/14/659314/1/check/build-openstack-releasenotes/3888644/html/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0cc86ce88b648b957a9103544c6de26bdd4af1dc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-05-16 09:06:44.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3cd1690159d82a728ca9af882ec72e123a58ba7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:06:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"130af6df909a0b62aaa428655d734e225b983fc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 11:02:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d647301119853449df8d4694715aafbce649b386","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 11:02:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/14/659314/1/gate/requirements-check/9374816/ : SUCCESS in 7m 15s\n- tempest-full http://logs.openstack.org/14/659314/1/gate/tempest-full/07b6368/ : SUCCESS in 1h 53m 30s\n- tempest-full-py3 http://logs.openstack.org/14/659314/1/gate/tempest-full-py3/fb6be70/ : SUCCESS in 1h 29m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/14/659314/1/gate/openstack-tox-lower-constraints/73b1ae9/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/14/659314/1/gate/openstack-tox-pep8/c04b239/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/14/659314/1/gate/openstack-tox-py27/8acc902/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/14/659314/1/gate/openstack-tox-py35/d4a7020/ : SUCCESS in 8m 06s\n- openstack-tox-py36 http://logs.openstack.org/14/659314/1/gate/openstack-tox-py36/4d156d7/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/14/659314/1/gate/openstack-tox-py37/2091514/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/14/659314/1/gate/openstack-tox-docs/6e3091b/html/ : SUCCESS in 4m 21s\n- build-openstack-releasenotes http://logs.openstack.org/14/659314/1/gate/build-openstack-releasenotes/2b1e85c/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2e62c152ae5269411862d6bfc8d98b231637b14","revisions":{"d2e62c152ae5269411862d6bfc8d98b231637b14":{"kind":"REWORK","_number":1,"created":"2019-05-15 15:23:50.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/14/659314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/14/659314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/14/659314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/14/659314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/14/659314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/14/659314/1"}}},"commit":{"parents":[{"commit":"f0a9926b11a669fb55131fb6ec6e52db2a82c20a","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/f0a9926b11a669fb55131fb6ec6e52db2a82c20a"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-15 15:23:22.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-15 15:23:22.000000000","tz":0},"subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","message":"Cap Bandit below 1.6.0 and update Sphinx requirement\n\nBandit 1.6.0 accidentally changed how the exclusion list option is\nhandled and breaks our use of it. Cap to the previous version until\nBandit has fixed the problem.\n\nSphinx 2.0 no longer works on python 2.7, so we need to start capping\nit there as well.\n\nChange-Id: Ifdad77dfc957ed90da88efe729cc6b35e96f0392\nReference: https://github.com/PyCQA/bandit/pull/489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d2e62c152ae5269411862d6bfc8d98b231637b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d2e62c152ae5269411862d6bfc8d98b231637b14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
