)]}'
{"id":"openstack%2Foslo.middleware~659315","triplet_id":"openstack%2Foslo.middleware~master~I94b62aabe2a02d066b9afee369addc014383aeea","project":"openstack/oslo.middleware","branch":"master","topic":"cap-bandit","hashtags":[],"change_id":"I94b62aabe2a02d066b9afee369addc014383aeea","subject":"Uncap Bandit","status":"ABANDONED","created":"2019-05-15 15:23:50.000000000","updated":"2019-05-22 14:18:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5d7ef437ffb3314d225118cd32c916d99e60348","_number":659315,"virtual_id_number":659315,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-05-15 17:31:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 17:31:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6cb5ceda58e2305107b14a28979fa46dba7afe2","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-15 15:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae7bd80a8a8a478317900d959bc52104dee3ad8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 17:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/15/659315/1/check/requirements-check/6201eb4/ : SUCCESS in 3m 39s\n- tempest-full http://logs.openstack.org/15/659315/1/check/tempest-full/5e947df/ : SUCCESS in 1h 58m 08s\n- tempest-full-py3 http://logs.openstack.org/15/659315/1/check/tempest-full-py3/46e5cb4/ : SUCCESS in 1h 39m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/659315/1/check/openstack-tox-lower-constraints/e89432e/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/15/659315/1/check/openstack-tox-pep8/7a112be/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/15/659315/1/check/openstack-tox-py27/cf1b426/ : SUCCESS in 5m 53s\n- openstack-tox-py35 http://logs.openstack.org/15/659315/1/check/openstack-tox-py35/9c0326f/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/15/659315/1/check/openstack-tox-py36/371c9bc/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/15/659315/1/check/openstack-tox-py37/23873da/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/15/659315/1/check/openstack-tox-docs/a46c222/html/ : SUCCESS in 3m 58s\n- build-openstack-releasenotes http://logs.openstack.org/15/659315/1/check/build-openstack-releasenotes/c4429a7/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b43bb89de965a475c6036382ca6eab5135a59914","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-22 14:18:46.000000000","message":"Abandoned\n\nAbandoning per http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006432.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f0652012835d5800e1328e4e0aa4b9700cafc46","revisions":{"7f0652012835d5800e1328e4e0aa4b9700cafc46":{"kind":"REWORK","_number":1,"created":"2019-05-15 15:23:50.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/15/659315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/15/659315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/15/659315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/15/659315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/15/659315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/15/659315/1"}}},"commit":{"parents":[{"commit":"d2e62c152ae5269411862d6bfc8d98b231637b14","subject":"Cap Bandit below 1.6.0 and update Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d2e62c152ae5269411862d6bfc8d98b231637b14"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-15 15:23:38.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-15 15:23:38.000000000","tz":0},"subject":"Uncap Bandit","message":"Uncap Bandit\n\nThis was capped due to a problem with how it processed exclusions.\nOnce that problem is fixed we want to uncap it and exlude only the\nspecific versions that are broken.\n\nChange-Id: I94b62aabe2a02d066b9afee369addc014383aeea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7f0652012835d5800e1328e4e0aa4b9700cafc46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7f0652012835d5800e1328e4e0aa4b9700cafc46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
