)]}'
{"id":"openstack%2Foslo.middleware~669061","triplet_id":"openstack%2Foslo.middleware~master~Ic66882afba25222ab8464ac9194c5002c8666db1","project":"openstack/oslo.middleware","branch":"master","topic":"bug/1835363","hashtags":[],"change_id":"Ic66882afba25222ab8464ac9194c5002c8666db1","subject":"More details when logging 413 Entity Too Large","status":"MERGED","created":"2019-07-04 06:20:04.000000000","updated":"2019-07-16 17:18:44.000000000","submitted":"2019-07-16 17:18:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669061-1563297524860-14776782","meta_rev_id":"63d87904ed212e3ab4d89bf149816c166f85e502","_number":669061,"virtual_id_number":669061,"owner":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-07-11 04:41:55.000000000","_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-07-16 17:18:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-16 15:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},{"value":2,"date":"2019-07-16 15:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-07-16 15:34:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 15:09:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-16 15:34:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-16 17:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f571ebf2eeeeefe0ab02c09ae9ca115828c1db42","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-04 06:20:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8079ce2fd30dda07a50d4fb02681c88bf3be5a7a","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-04 06:44:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01ce31525ab941e2af55ce66feb60d20f10c0d33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:34:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/669061/2/check/tempest-full/33d2ca1/ : SUCCESS in 1h 37m 36s\n- tempest-full-py3 http://logs.openstack.org/61/669061/2/check/tempest-full-py3/ddae61a/ : SUCCESS in 1h 34m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/2/check/openstack-tox-lower-constraints/ace847b/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/2/check/openstack-tox-pep8/fdf7386/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/2/check/openstack-tox-py27/28b0626/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/2/check/openstack-tox-py36/3c8c992/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/2/check/openstack-tox-py37/eca7bb2/ : SUCCESS in 3m 03s\n- openstack-tox-docs http://logs.openstack.org/61/669061/2/check/openstack-tox-docs/43758c4/html/ : SUCCESS in 2m 33s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/2/check/build-openstack-releasenotes/8163f9c/html/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9a03166306fb99807b24c8ff0d6c4222b7507f36","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 10:50:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aa6648b6a8569fdc6b32acdb743fa3495b55fb2d","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-04 13:23:53.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThanks for the review!\nIf you have some further suggestion for how I could make it clearer to the receiver of the 413 what is sending it I am very open for suggestions.","accounts_in_message":[],"_revision_number":2},{"id":"de9c9d996e7ac04989951111b3aad50717c161be","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 04:37:37.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ddd3c50645338567c0e0dbd2583f85969a420257","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 05:11:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"45ded6e4ca520e38cd702aee7bec562c819716c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 06:40:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/669061/3/check/tempest-full/227526a/ : SUCCESS in 2h 02m 42s\n- tempest-full-py3 http://logs.openstack.org/61/669061/3/check/tempest-full-py3/bedd3b7/ : SUCCESS in 1h 29m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/3/check/openstack-tox-lower-constraints/6298bec/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/3/check/openstack-tox-pep8/cca19e8/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/3/check/openstack-tox-py27/050a634/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/3/check/openstack-tox-py36/f4bda59/ : FAILURE in 2m 59s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/3/check/openstack-tox-py37/6500813/ : SUCCESS in 3m 10s\n- openstack-tox-docs http://logs.openstack.org/61/669061/3/check/openstack-tox-docs/92254a1/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/3/check/build-openstack-releasenotes/020a850/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fedf7e576d437086b74a0f8ce8dae4030fbdb981","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 07:06:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0aefb29d2180f6ae0751b36e7ab920264e90e2e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 08:47:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/669061/3/check/tempest-full/01312fe/ : SUCCESS in 1h 40m 24s\n- tempest-full-py3 http://logs.openstack.org/61/669061/3/check/tempest-full-py3/61162d6/ : SUCCESS in 1h 23m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/3/check/openstack-tox-lower-constraints/99ff9b4/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/3/check/openstack-tox-pep8/7e9bbaa/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/3/check/openstack-tox-py27/2508a89/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/3/check/openstack-tox-py36/d2332bf/ : SUCCESS in 2m 57s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/3/check/openstack-tox-py37/d51a88b/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/61/669061/3/check/openstack-tox-docs/602a968/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/3/check/build-openstack-releasenotes/f8c2fbe/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ba7b7bfc273529fed5735f15415128ef8be2149f","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 11:49:46.000000000","message":"Patch Set 3:\n\n\u003e (2 comments)\n\nI now read your comment carefully. Updated the commit message and it now has an example of how the log actually looks.","accounts_in_message":[],"_revision_number":3},{"id":"892db651378bb395625a37fadc117c331c5acaf3","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 16:40:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96a1440a506cd6754e21452f1feb5fd4af4b1140","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 17:13:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5ade3c0b0d9dd613b773e0be31d8a054f383e256","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 19:26:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/61/669061/5/check/tempest-full/293b752/ : SUCCESS in 1h 45m 36s\n- tempest-full-py3 http://logs.openstack.org/61/669061/5/check/tempest-full-py3/8cfead1/ : SUCCESS in 1h 47m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/5/check/openstack-tox-lower-constraints/8521611/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/5/check/openstack-tox-pep8/c61417a/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/5/check/openstack-tox-py27/cb0beec/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/5/check/openstack-tox-py36/df39c9c/ : FAILURE in 2m 24s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/5/check/openstack-tox-py37/0564302/ : SUCCESS in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/61/669061/5/check/openstack-tox-docs/08a8b1b/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/5/check/build-openstack-releasenotes/32d66bb/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":5},{"id":"aabb07bcfbea2b47e74877aaac125f9c4b141ea9","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-07 19:31:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"731c551c9a5d952466640f05740e59023cce977a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-07 21:11:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/669061/5/check/tempest-full/b4ce133/ : SUCCESS in 1h 38m 54s\n- tempest-full-py3 http://logs.openstack.org/61/669061/5/check/tempest-full-py3/38f2e5e/ : SUCCESS in 1h 30m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/5/check/openstack-tox-lower-constraints/d0b7923/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/5/check/openstack-tox-pep8/8a44385/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/5/check/openstack-tox-py27/afaa03f/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/5/check/openstack-tox-py36/773c52d/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/5/check/openstack-tox-py37/bf2a550/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/61/669061/5/check/openstack-tox-docs/67ec24c/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/5/check/build-openstack-releasenotes/a4d6a07/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":5},{"id":"fb0c26f39cec3a93b2bca2c43d4b872656642f2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-10 16:30:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6db097b8542f95a020cb6f89fdb8ea514aedf79f","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-10 17:35:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"40ed092a62f65acd448d3fee556a521313e44036","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:45:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/669061/6/check/tempest-full/8d29422/ : SUCCESS in 2h 03m 09s\n- tempest-full-py3 http://logs.openstack.org/61/669061/6/check/tempest-full-py3/3f7653b/ : SUCCESS in 1h 43m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/6/check/openstack-tox-lower-constraints/d886439/ : SUCCESS in 3m 11s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/6/check/openstack-tox-pep8/9aa6637/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/6/check/openstack-tox-py27/eb1fe65/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/6/check/openstack-tox-py36/6d86ddc/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/6/check/openstack-tox-py37/bc19898/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/61/669061/6/check/openstack-tox-docs/7ac7064/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/6/check/build-openstack-releasenotes/6e3046e/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c6a60ff1d75aba6c5e4c7b9479a1ad4094feb20f","author":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"date":"2019-07-11 04:41:55.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n\nBecause I didn\u0027t think of it!\nUpdated the commit with your proposal.","accounts_in_message":[],"_revision_number":6},{"id":"bcd398d21e5c828bed92e1c61e8dbd5f53344865","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-16 15:09:21.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"f4573fa38579304a91a1cee5172f43f6a96398e9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-16 15:34:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThanks for the updates. This looks good to go now","accounts_in_message":[],"_revision_number":6},{"id":"d38461c798861e0f0e72ee5b9a0a056bac9d0285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:34:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e6c6b78b31d26bb78b2ed55a32d81cb0b270feca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:18:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"28bdfa3dd25fd82989e193b5a7a50b8c5972c613","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:18:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/61/669061/6/gate/tempest-full/978e41f/ : SUCCESS in 1h 38m 34s\n- tempest-full-py3 http://logs.openstack.org/61/669061/6/gate/tempest-full-py3/c3e77ba/ : SUCCESS in 1h 30m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669061/6/gate/openstack-tox-lower-constraints/44ef5ee/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/61/669061/6/gate/openstack-tox-pep8/2a909aa/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/61/669061/6/gate/openstack-tox-py27/9e44f86/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/61/669061/6/gate/openstack-tox-py36/1d2540a/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/61/669061/6/gate/openstack-tox-py37/789b3bc/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/61/669061/6/gate/openstack-tox-docs/abea000/html/ : SUCCESS in 2m 37s\n- build-openstack-releasenotes http://logs.openstack.org/61/669061/6/gate/build-openstack-releasenotes/b74c4b9/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d6b74f5ce9af172d0b3a33234b1489d9af59e5fe","revisions":{"133ee7f567e8643a4bfe95c3a29472cf1ecc5da7":{"kind":"REWORK","_number":1,"created":"2019-07-04 06:20:04.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/1"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:20:02.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/133ee7f567e8643a4bfe95c3a29472cf1ecc5da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/133ee7f567e8643a4bfe95c3a29472cf1ecc5da7"}]},"branch":"refs/heads/master"},"c713c3df0c18dbbe2aae554b0e73a6bff4bfe1ff":{"kind":"REWORK","_number":2,"created":"2019-07-04 06:44:36.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/2"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:42:43.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nResulting log error in keystone looks like:\n\n2019-07-04 09:14:10.467 1464322 WARNING oslo_middleware.sizelimit\n[req-9ab12d1b-1c3b-44c8-9e87-a2def9f4577b - - - - -] Request is too\nlarge. It is larger than max_request_body_size (114688).\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c713c3df0c18dbbe2aae554b0e73a6bff4bfe1ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/c713c3df0c18dbbe2aae554b0e73a6bff4bfe1ff"}]},"branch":"refs/heads/master"},"84e23cfd664b81808221b99cf60f5dca15642d68":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-07 04:37:37.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/3"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-07 04:36:08.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nResulting log error in keystone looks like:\n\n2019-07-04 06:52:13.217789 | ubuntu-bionic |  WARNING\n[oslo_middleware.sizelimit] Request is too large.  \u003e\nmax_request_body_size (114688).\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/84e23cfd664b81808221b99cf60f5dca15642d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/84e23cfd664b81808221b99cf60f5dca15642d68"}]},"branch":"refs/heads/master"},"bb4f769178bf5aaa38c637f43853a67eea47537f":{"kind":"REWORK","_number":4,"created":"2019-07-07 16:40:59.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/4"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-07 16:40:31.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nResulting log error in keystone looks like:\n\n2019-07-04 06:52:13.217789 | ubuntu-bionic |  INFO\n[oslo_middleware.sizelimit] Request is too large. Larger than\nmax_request_body_size (114688).\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/bb4f769178bf5aaa38c637f43853a67eea47537f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/bb4f769178bf5aaa38c637f43853a67eea47537f"}]},"branch":"refs/heads/master"},"d6dafc7f285209719c30064726f18c435c2ce67f":{"kind":"REWORK","_number":5,"created":"2019-07-07 17:13:30.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/5"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-07 17:13:16.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nResulting log error in keystone looks like:\n\n2019-07-04 06:52:13.217789 | ubuntu-bionic |  INFO\n[oslo_middleware.sizelimit] Request is too large. Larger than\nmax_request_body_size (114688).\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d6dafc7f285209719c30064726f18c435c2ce67f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d6dafc7f285209719c30064726f18c435c2ce67f"}]},"branch":"refs/heads/master"},"d6b74f5ce9af172d0b3a33234b1489d9af59e5fe":{"kind":"REWORK","_number":6,"created":"2019-07-10 17:35:34.000000000","uploader":{"_account_id":28248,"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","username":"martbhell"},"ref":"refs/changes/61/669061/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/61/669061/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/61/669061/6"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-04 06:11:44.000000000","tz":180},"committer":{"name":"Johan Guldmyr","email":"johan.guldmyr@csc.fi","date":"2019-07-10 17:34:59.000000000","tz":180},"subject":"More details when logging 413 Entity Too Large","message":"More details when logging 413 Entity Too Large\n\nI spent a few days figuring out what was making the OS-FEDERATION\nmapping rules create/update API calls return \"413 Entity Too Large\"\nerrors for our CentOS7/haproxy/apache/mod_wsgi/keystone/python2 setup.\n\nTo make this a bit easier for future users then I would like to include\nmax_request_body_size and its value in the response and also log the\nerror.\n\nRetaining the \"Request is too large.\" bit in the message is intended to\nalso make it easier to find where the limit is set.\n\nResulting log error in keystone looks like:\n\n2019-07-04 06:52:13.217789 | ubuntu-bionic |  INFO\n[oslo_middleware.sizelimit] Request is too large. Larger than\nmax_request_body_size (114688).\n\nChange-Id: Ic66882afba25222ab8464ac9194c5002c8666db1\nCloses-Bug: #1835363\nSigned-off-by: Johan Guldmyr \u003cjohan.guldmyr@csc.fi\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d6b74f5ce9af172d0b3a33234b1489d9af59e5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/d6b74f5ce9af172d0b3a33234b1489d9af59e5fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
