)]}'
{"id":"openstack%2Foslo.middleware~669110","triplet_id":"openstack%2Foslo.middleware~master~I43d5cc2b101125296c60c4317a9a8f62e3b0639b","project":"openstack/oslo.middleware","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I43d5cc2b101125296c60c4317a9a8f62e3b0639b","subject":"Add Python 3 Train unit tests","status":"ABANDONED","created":"2019-07-04 08:49:29.000000000","updated":"2019-07-12 03:43:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f76e995601f7f6c892503712abda912bad7b6f48","_number":669110,"virtual_id_number":669110,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-2,"date":"2019-07-11 11:18:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-11 11:18:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-07-04 12:42:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"value":1,"date":"2019-07-11 11:18:38.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 12:42:21.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-11 11:18:38.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-07-11 11:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8892cff98098b6f37659d5e49b639e52a81416c5","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 08:49:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a54ef74a65e9f80742ba19346ee671533c3ac28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/10/669110/1/check/test-release-openstack/429fe53/ : SUCCESS in 1m 52s\n- tempest-full http://logs.openstack.org/10/669110/1/check/tempest-full/a04f3ce/ : SUCCESS in 1h 44m 58s\n- tempest-full-py3 http://logs.openstack.org/10/669110/1/check/tempest-full-py3/3614780/ : SUCCESS in 1h 44m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/669110/1/check/openstack-tox-lower-constraints/805482a/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/10/669110/1/check/openstack-tox-pep8/933f917/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/10/669110/1/check/openstack-tox-py27/04e6ee2/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/10/669110/1/check/openstack-tox-py36/4868b4e/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/10/669110/1/check/openstack-tox-py37/fcb5d59/ : SUCCESS in 3m 06s\n- openstack-tox-docs http://logs.openstack.org/10/669110/1/check/openstack-tox-docs/c27eaa8/html/ : SUCCESS in 4m 14s\n- build-openstack-releasenotes http://logs.openstack.org/10/669110/1/check/build-openstack-releasenotes/9bb436d/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5ca6490560ec0fd73ad6f31285fab91b4f2e7467","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 12:42:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1aaed05a03c53230df67998b8e6b8488282dcbe3","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-07-11 11:18:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4ba29b9a637747fa51a25c341bc3393908a7d21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:18:49.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"dfa13218d4a4ef88392f6faf28e30397be78f1e5","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-12 03:43:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"608146e051919f84bec9d848e099552e1815d267","revisions":{"608146e051919f84bec9d848e099552e1815d267":{"kind":"REWORK","_number":1,"created":"2019-07-04 08:49:29.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/10/669110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.middleware","ref":"refs/changes/10/669110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/669110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/669110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.middleware refs/changes/10/669110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.middleware refs/changes/10/669110/1"}}},"commit":{"parents":[{"commit":"7679e2057a07998fdf1b2cb4a8fd825c20ee469d","subject":"Merge \"Dropping the py35 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/7679e2057a07998fdf1b2cb4a8fd825c20ee469d"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:48:40.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 08:49:31.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I43d5cc2b101125296c60c4317a9a8f62e3b0639b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/608146e051919f84bec9d848e099552e1815d267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.middleware/commit/608146e051919f84bec9d848e099552e1815d267"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
